)]}'
{"id":"openstack%2Fglance~819399","triplet_id":"openstack%2Fglance~master~Id77010aac04aa9a4961c5bcafbf12694a0fe17c6","project":"openstack/glance","branch":"master","topic":"bug/1905672","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2022-04-20 05:25:27.000000000","reason":"Abhishek Kekane replied on the change"},"33786":{"account":{"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"},"last_update":"2022-08-02 14:43:28.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2022-08-02 13:00:14.000000000","reason":"\u003cGERRIT_ACCOUNT_8122\u003e replied on the change","reason_account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}}},"hashtags":[],"change_id":"Id77010aac04aa9a4961c5bcafbf12694a0fe17c6","subject":"Update exception for property protection file","status":"MERGED","created":"2021-11-26 06:28:18.000000000","updated":"2022-08-02 14:44:59.000000000","submitted":"2022-08-02 14:43:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"819399-bug/1905672","meta_rev_id":"a3119769e45a49a531fc22a99db54cc9d1cc1df5","_number":819399,"virtual_id_number":819399,"owner":{"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-02 14:43:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2022-08-02 05:44:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2022-08-02 13:00:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2022-08-02 13:00:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":33801,"name":"wang","email":"wangleic@uniontech.com"}],"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-26 06:30:12.000000000","updated_by":{"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"},"reviewer":{"_account_id":33801,"name":"wang","email":"wangleic@uniontech.com"},"state":"CC"},{"updated":"2021-11-26 06:30:12.000000000","updated_by":{"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2021-11-26 08:09:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-01 16:30:38.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-07 20:58:30.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"CC"},{"updated":"2022-08-02 13:00:14.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"4983bd56cec145e992573290b973d3d74d76eeea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"},"date":"2021-11-26 06:28:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c3c09da4bf8fbdbf08cb5dd2d2a18ad4309e5faf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"},"date":"2021-11-26 06:29:43.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"212e12d2261e7361b7db93cf7cf77e204357eaf7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-26 08:09:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/d5f2550f67054aa88e2d32b654d15baf : SUCCESS in 1h 05m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0b36fab657c5460ca06960446e33b7ec : SUCCESS in 1h 14m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94e07cca3013415195cb8e2860a55b1e : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/67ee9db4f283477bbc2baf26ad620866 : SUCCESS in 5m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5f7707a2529a428a950efc661196939a : SUCCESS in 6m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b960bb5c5fb4925aeff9f5b06824538 : SUCCESS in 7m 31s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/89d809f47c954b14b7d3cd5ccb54987a : SUCCESS in 13m 20s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4b56279a345444c1b75f85252756d602 : SUCCESS in 8m 11s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/0f829999fdcc4ac1844e7d364035d0ff : SUCCESS in 8m 21s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/b640d19863b9410fb79d6ac1a740facf : SUCCESS in 4m 32s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/aa3759d39aff47dfa5d9ec1d44ba59d7 : SUCCESS in 1h 23m 48s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/537cbc5586ff491eabaad00991f29488 : SUCCESS in 1h 14m 23s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/fb079f4c3c994839b48d0fbad3eb164c : SUCCESS in 1h 24m 02s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/0ce61511a7cd40d18b406f48fe142b9c : SUCCESS in 1h 32m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9e8cddf1b9404b888e0a8c1d97c449bb : SUCCESS in 57m 25s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8f34bb1b960042a89518683f3304d0de : SUCCESS in 1h 24m 35s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b69f7cf576074d71855c9ca5b76f9939 : SUCCESS in 35m 23s","accounts_in_message":[],"_revision_number":2},{"id":"fb148855848431775e5e6a817ab0bff3dbb09492","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-11-30 05:23:39.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5b22f3d0a69287c0f5eee5fce466cd315c0fd17c","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-11-30 05:23:45.000000000","message":"Patch Set 2: Code-Review-1","accounts_in_message":[],"_revision_number":2},{"id":"8cb1a762cd1dd4322e160c087c0ca08e0327949c","author":{"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"},"date":"2021-11-30 05:34:46.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d6b2f235f0ca04818459f685f023c3237efa3b0e","author":{"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"},"date":"2021-11-30 07:34:46.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"577d244e010c4170e035aa4af76bfb36a4942d5a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-11-30 07:39:03.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8484ef6be90cc77cdfa81d11a7f85312c6fa7936","author":{"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"},"date":"2021-11-30 07:41:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"43b17fcefddc859d57a2f0821817ac7564969f10","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-12-01 16:30:38.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/819399/819399/2/check-not-vote/ext-glance-zuul/2b9402e : FAILURE in 1m 39s","accounts_in_message":[],"_revision_number":2},{"id":"40d3dfa46ade872d503e8ad0ddfa70a7a853e151","author":{"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"},"date":"2022-04-14 13:56:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1d6969d8f1c10be855e58a9a27347467482049f2","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-04-14 14:10:33.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9928a0e4bfb4247440450474788883faeda8a960","author":{"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"},"date":"2022-04-20 01:33:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a898e7d2ca57a6adf47b52c5af166351e8f7a469","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-04-20 05:25:27.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7e7b1d6e631dfc21487688bcf7a2095b99669488","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-07-07 20:58:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"71664bbfc4709978244f036250b47e13a999ee48","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"},"date":"2022-07-25 09:45:06.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b49df66b13b86ce9b682ec66d6c123ca167f8eec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"},"date":"2022-07-25 09:51:41.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"7a5a1836105b593efec078af2841828271f9cc51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 11:21:46.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/817ce6b4738e426a814db7b62b267a33\n\n- grenade https://zuul.opendev.org/t/openstack/build/33b4067d56564e4ebc731e15aba514a2 : SUCCESS in 1h 09m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/92ea81d645cc48c89b6f0ba36b236e3e : SUCCESS in 1h 09m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0bfe194edbae469a955837e87ad3eb05 : SUCCESS in 52m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02691e6f752747d8aade01c72a6a3679 : FAILURE in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4bfa1e15a5464900933d15947cde2c60 : SUCCESS in 4m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f1735380616e4cb9aac1fcda4d54f8ec : SUCCESS in 7m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f16995590b364c6582143ce2a3047136 : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e8c8a1a86e241998223efc7b9c7b9e6 : SUCCESS in 5m 29s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/65ed801b3c8a46cea74a6990ca9379c7 : SUCCESS in 13m 32s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5b4daf39dacd43a6834ff0a6ed754b79 : SUCCESS in 9m 43s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/786546e57f7847ce9aeef86277c452c9 : SUCCESS in 8m 06s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/ebabaa4c3bf04113a7efca4a03218aed : SUCCESS in 4m 50s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/9bf5a534d6024d97bae4b107083d0eec : SUCCESS in 1h 14m 58s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6bad0c647cc04b5d84992c44e6f8a6c3 : SUCCESS in 1h 12m 17s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/639b50cf799542298c8f6dab37ec92b1 : SUCCESS in 1h 21m 38s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a70027739d8344f6950b0759c3168ff2 : SUCCESS in 1h 28m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e17ee475f93e425fb6b6314894b74fb3 : SUCCESS in 51m 32s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/42ae1c138f864e7b91e7b8e8c03de25d : SUCCESS in 1h 18m 11s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/fec2bc89d67a4ee5b5807eb0a10fea9c : SUCCESS in 39m 29s","accounts_in_message":[],"_revision_number":4},{"id":"3853d0dfe898172d75cb02eda7a59560408b8c60","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-08-01 17:38:52.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"743101702c320d7ddaec046ff443fe28a1fcd789","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-08-01 17:40:28.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"22fde7054cef8d5b6301e32ac18be4db674c642a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-01 19:31:19.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f0c64a42d9a41ada05ed7afe5b04eb3\n\n- grenade https://zuul.opendev.org/t/openstack/build/929a58df64ca43e88394d1e63db219bb : SUCCESS in 57m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f4a0602c493044448b3f29aded08a5b0 : SUCCESS in 1h 19m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5f5e5f23e174434d83fb91fa69e3fd0f : SUCCESS in 1h 02m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86f36d0517e3495fb57eaa4a92378842 : SUCCESS in 6m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/df3d42cc477649a7a18dc60694d4d4a2 : SUCCESS in 6m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/25d064cda4674ae697d62bbe2321e344 : SUCCESS in 6m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/47e0532ec9b14ea49a53b57c7855efab : SUCCESS in 6m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7544782c82a341ec9b06fbca797bbdd8 : SUCCESS in 6m 27s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/e28785791c42436b9f5bf9bbce34c383 : SUCCESS in 12m 22s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4673222439f04caf9c1b44fddd71a305 : SUCCESS in 10m 37s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/c3e4828e54894f19b322deeb8f40b02d : SUCCESS in 9m 41s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/e13ff12f4d524b7a950bac1cdae8d599 : SUCCESS in 4m 57s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/dc6a52d541ec4f019ce19489cd7f975b : SUCCESS in 1h 22m 25s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/55fd2a83351540c3a646f9e97e4ec08b : SUCCESS in 1h 46m 06s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/da047df09207448da148180b977481c0 : SUCCESS in 1h 21m 37s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/0b8e24ee7ed94030946a63ef08a9b57f : SUCCESS in 1h 30m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/56e662d9bb3c4f008233baab4f64842c : SUCCESS in 59m 51s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b22216051f8a42179255a91e8d141b2e : SUCCESS in 1h 13m 02s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5d13d753acc2400e8f0da92f9edfa467 : SUCCESS in 39m 39s","accounts_in_message":[],"_revision_number":5},{"id":"9f278219ee2487a70844930043ca87f27ac6eac8","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-08-02 05:44:36.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"bb27b5eb40b652a2d5541afa3541d2c4df0de9ce","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-08-02 13:00:14.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d88944f03bac129c9c4cad27f4b1e1b7b4cf4121","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-02 13:01:02.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"f7d992ecac9fe8f4738a1847cc4de356f952d0f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-02 14:43:28.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a1c10ef1935403ea7c5a10602bff033\n\n- grenade https://zuul.opendev.org/t/openstack/build/de98667f85174eaabe4ab6a69ac3f1cd : SUCCESS in 43m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5db9c40ae833415cb7eb5b53a32d8898 : SUCCESS in 1h 20m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3221b6a9d0da4242a95c9e6f31e30716 : SUCCESS in 1h 01m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9e6ae181f8441c8ab5b12c43f40e9a1 : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ad433afb9c3440d08f6b0a78693c9658 : SUCCESS in 6m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d38627c1a8a64ed48d8c8c2a6b21eed4 : SUCCESS in 8m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77ad2fa8becd4c19a13b0a420b69ae24 : SUCCESS in 7m 16s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ebc3d715b61a41199ed4f633e21e8ac3 : SUCCESS in 7m 44s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a53c2f1faae24126b6afafa315b980b6 : SUCCESS in 10m 35s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f3f555d9dcd04007a1d1d11e6a45214b : SUCCESS in 1h 40m 59s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/4c8ba96a4428433b847ce5e13ef35008 : SUCCESS in 1h 17m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3c8a7ed52a0443e687cb3af0003626fe : SUCCESS in 40m 25s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/64c69962bb60406bb921be00a58c2cb4 : SUCCESS in 1h 24m 32s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5ad13a35bb364954ae8da2ef5934ba02 : SUCCESS in 38m 33s","accounts_in_message":[],"_revision_number":5},{"id":"bad8c85aeeeb41ca005dc8e13b58295d62659ca5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-02 14:43:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"a3119769e45a49a531fc22a99db54cc9d1cc1df5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-02 14:44:59.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84a20a839dda44f8b2eb8c5d5f0c8705\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfbcb9fe589b403f8dbab55550224acd : SUCCESS in 53s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"4e59a04d27175533d7ae91b9acd23ffe32a5917a","revisions":{"c8ad30356a0d78e2d21226db97b41c967d0b3038":{"kind":"REWORK","_number":1,"created":"2021-11-26 06:28:18.000000000","uploader":{"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"},"ref":"refs/changes/99/819399/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/99/819399/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/819399/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/819399/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/819399/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/99/819399/1"}}},"commit":{"parents":[{"commit":"6fd949a966fea443bb4b0b9351b3fa04662a7ecb","subject":"Refactor glance cinder tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6fd949a966fea443bb4b0b9351b3fa04662a7ecb"}]}],"author":{"name":"Han Guangyu","email":"hanguangyu@uniontech.com","date":"2021-11-26 06:25:00.000000000","tz":480},"committer":{"name":"Han Guangyu","email":"hanguangyu@uniontech.com","date":"2021-11-26 06:28:10.000000000","tz":480},"subject":"Update exception for property protection file","message":"Update exception for property protection file\n\nNon existing property protection file raises 500 Internal server\nerror. If admin/operator specifies non existing property protection\nfile in glance-api.conf. For example:\n    [Default]\n    property_protection_file \u003d non_existing_file.yaml\n    property_protection_rule_format \u003d policies\nThen create or update image. The 500 Internal server error wil be\nraised. But the expected result is 400 Bad Request.\n\nAdd the except of InavlidPropertyProtectionConfiguration exception,\nand return 400 Bad Request.\n\nThe previous code lacks the specific process of\nPropertyProtectionConfiguration exception, so it returned the common\n500 error. Now, it returns 400 Bad Request.\n\nChange-Id: Id77010aac04aa9a4961c5bcafbf12694a0fe17c6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c8ad30356a0d78e2d21226db97b41c967d0b3038"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c8ad30356a0d78e2d21226db97b41c967d0b3038"}]},"branch":"refs/heads/master"},"e29a9884e3cc21a5ff45af43480337d7fa627f1c":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-11-26 06:29:43.000000000","uploader":{"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"},"ref":"refs/changes/99/819399/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/99/819399/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/819399/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/819399/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/819399/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/99/819399/2"}}},"commit":{"parents":[{"commit":"6fd949a966fea443bb4b0b9351b3fa04662a7ecb","subject":"Refactor glance cinder tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6fd949a966fea443bb4b0b9351b3fa04662a7ecb"}]}],"author":{"name":"Han Guangyu","email":"hanguangyu@uniontech.com","date":"2021-11-26 06:25:00.000000000","tz":480},"committer":{"name":"HanGuangyu","email":"hanguangyu@uniontech.com","date":"2021-11-26 06:29:43.000000000","tz":0},"subject":"Update exception for property protection file","message":"Update exception for property protection file\n\nNon existing property protection file raises 500 Internal server\nerror. If admin/operator specifies non existing property protection\nfile in glance-api.conf. For example:\n    [Default]\n    property_protection_file \u003d non_existing_file.yaml\n    property_protection_rule_format \u003d policies\nThen create or update image. The 500 Internal server error wil be\nraised. But the expected result is 400 Bad Request.\n\nAdd the except of InavlidPropertyProtectionConfiguration exception,\nand return 400 Bad Request.\n\nThe previous code lacks the specific process of\nPropertyProtectionConfiguration exception, so it returned the common\n500 error. Now, it returns 400 Bad Request.\n\nCloses-Bug: #1905672\nChange-Id: Id77010aac04aa9a4961c5bcafbf12694a0fe17c6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e29a9884e3cc21a5ff45af43480337d7fa627f1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e29a9884e3cc21a5ff45af43480337d7fa627f1c"}]},"branch":"refs/heads/master","description":"Edit commit message"},"a41328640a63c8364708425386629640586681d9":{"kind":"REWORK","_number":3,"created":"2022-07-25 09:45:06.000000000","uploader":{"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"},"ref":"refs/changes/99/819399/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/99/819399/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/819399/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/819399/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/819399/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/99/819399/3"}}},"commit":{"parents":[{"commit":"6fd949a966fea443bb4b0b9351b3fa04662a7ecb","subject":"Refactor glance cinder tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6fd949a966fea443bb4b0b9351b3fa04662a7ecb"}]}],"author":{"name":"Han Guangyu","email":"hanguangyu@uniontech.com","date":"2021-11-26 06:25:00.000000000","tz":480},"committer":{"name":"Han Guangyu","email":"hanguangyu@uniontech.com","date":"2022-07-25 09:45:03.000000000","tz":480},"subject":"Update exception for property protection file","message":"Update exception for property protection file\n\nNon existing property protection file raises 500 Internal server\nerror. If admin/operator specifies non existing property protection\nfile in glance-api.conf. For example:\n    [Default]\n    property_protection_file \u003d non_existing_file.yaml\n    property_protection_rule_format \u003d policies\nThen create or update image. The 500 Internal server error wil be\nraised. But the expected result is 400 Bad Request.\n\nAdd the except of InavlidPropertyProtectionConfiguration exception,\nand return 400 Bad Request.\n\nThe previous code lacks the specific process of\nPropertyProtectionConfiguration exception, so it returned the common\n500 error. Now, it returns 400 Bad Request.\n\nCloses-Bug: #1905672\nChange-Id: Id77010aac04aa9a4961c5bcafbf12694a0fe17c6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a41328640a63c8364708425386629640586681d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a41328640a63c8364708425386629640586681d9"}]},"branch":"refs/heads/master"},"0f19adbbc2d6eb16012f3934d4ecc199322cb7f5":{"kind":"REWORK","_number":4,"created":"2022-07-25 09:51:41.000000000","uploader":{"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"},"ref":"refs/changes/99/819399/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/99/819399/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/819399/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/819399/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/819399/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/99/819399/4"}}},"commit":{"parents":[{"commit":"6fd949a966fea443bb4b0b9351b3fa04662a7ecb","subject":"Refactor glance cinder tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6fd949a966fea443bb4b0b9351b3fa04662a7ecb"}]}],"author":{"name":"Han Guangyu","email":"hanguangyu@uniontech.com","date":"2021-11-26 06:25:00.000000000","tz":480},"committer":{"name":"Han Guangyu","email":"hanguangyu@uniontech.com","date":"2022-07-25 09:51:20.000000000","tz":480},"subject":"Update exception for property protection file","message":"Update exception for property protection file\n\nNon existing property protection file raises 500 Internal server\nerror. If admin/operator specifies non existing property protection\nfile in glance-api.conf. For example:\n    [Default]\n    property_protection_file \u003d non_existing_file.yaml\n    property_protection_rule_format \u003d policies\nThen create or update image. The 500 Internal server error wil be\nraised. But the expected result is 400 Bad Request.\n\nAdd the except of InavlidPropertyProtectionConfiguration exception,\nand return 400 Bad Request.\n\nThe previous code lacks the specific process of\nPropertyProtectionConfiguration exception, so it returned the common\n500 error. Now, it returns 400 Bad Request.\n\nCloses-Bug: #1905672\nChange-Id: Id77010aac04aa9a4961c5bcafbf12694a0fe17c6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0f19adbbc2d6eb16012f3934d4ecc199322cb7f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0f19adbbc2d6eb16012f3934d4ecc199322cb7f5"}]},"branch":"refs/heads/master"},"4e59a04d27175533d7ae91b9acd23ffe32a5917a":{"kind":"REWORK","_number":5,"created":"2022-08-01 17:38:52.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/99/819399/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/99/819399/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/819399/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/819399/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/819399/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/99/819399/5"}}},"commit":{"parents":[{"commit":"00f453372ccd4faf99e9a0bf527762a131968ad0","subject":"Merge \"Immediate caching of an image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/00f453372ccd4faf99e9a0bf527762a131968ad0"}]}],"author":{"name":"Han Guangyu","email":"hanguangyu@uniontech.com","date":"2021-11-26 06:25:00.000000000","tz":480},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2022-08-01 15:40:42.000000000","tz":120},"subject":"Update exception for property protection file","message":"Update exception for property protection file\n\nNon existing property protection file raises 500 Internal server\nerror. If admin/operator specifies non existing property protection\nfile in glance-api.conf. For example:\n    [Default]\n    property_protection_file \u003d non_existing_file.yaml\n    property_protection_rule_format \u003d policies\nThen create or update image. The 500 Internal server error will be\nraised. But the expected result is 400 Bad Request.\n\nAdd the except of InavlidPropertyProtectionConfiguration exception,\nand return 400 Bad Request.\n\nThe previous code lacks the specific process of\nPropertyProtectionConfiguration exception, so it returned the common\n500 error. Now, it returns 400 Bad Request.\n\nCloses-Bug: #1905672\nChange-Id: Id77010aac04aa9a4961c5bcafbf12694a0fe17c6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4e59a04d27175533d7ae91b9acd23ffe32a5917a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4e59a04d27175533d7ae91b9acd23ffe32a5917a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
