)]}'
{"id":"openstack%2Fglance~819599","triplet_id":"openstack%2Fglance~master~Ie0cac63dedf69728392f293f4551e495aebf4d40","project":"openstack/glance","branch":"master","topic":"log-warn","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-01-20 07:15:52.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2022-01-19 16:03:08.000000000","reason":"Cyril Roelandt replied on the change"}},"hashtags":[],"change_id":"Ie0cac63dedf69728392f293f4551e495aebf4d40","subject":"Use LOG.warning instead of deprecated LOG.warn","status":"MERGED","created":"2021-11-29 07:38:58.000000000","updated":"2022-01-20 07:17:22.000000000","submitted":"2022-01-20 07:15:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"819599-log-warn","meta_rev_id":"3978e0b822aa4c90789f226e658622181e881613","_number":819599,"virtual_id_number":819599,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-20 07:15:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":2,"date":"2022-01-20 05:41:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2022-01-19 16:03:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":1,"date":"2022-01-20 05:41:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-29 09:55:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-30 05:22:19.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2021-11-30 05:48:50.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"},{"updated":"2021-12-02 06:50:58.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-12-20 08:25:36.000000000","updated_by":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"reviewer":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"state":"REVIEWER"},{"updated":"2022-01-19 01:58:31.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"CC"},{"updated":"2022-01-19 16:03:08.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"acbfeaaece0487d21e3c5303cf9dfc7d871a3761","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-29 07:38:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b070ef8c369a8f8a88cb0e5692b25926eb7c0d49","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-29 08:08:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f4067af58859d746558b74daaef6cacd144ff7c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-29 08:08:48.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"991981a514e1c386f5d8d9bde9392b977e3c3f31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-29 09:55:15.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/b5e629ac318f412d891202fea42594c5 : FAILURE in 47m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fb06e3d5f9394556b0759eb8b2f09a45 : SUCCESS in 1h 25m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6236176689a24692b2fed88f664b3ece : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ddc209103ed647d48c5d864573d91377 : SUCCESS in 5m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dadbd1c396a846d29d8bd748e8d4602d : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38cd2503cb304721b7edc2f9c698e5f1 : SUCCESS in 6m 16s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f8610dee91e34f76bf73323b4091b663 : SUCCESS in 9m 08s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c3e3c1eaa0eb43b891fff39b54ffbf91 : SUCCESS in 8m 04s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/458bd7727553403e8041ad0bac45b22d : SUCCESS in 8m 36s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/595c81821c264a5094e242b9ac956ca3 : SUCCESS in 4m 34s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d7db5a201c554fe18f5dfa6060ee6457 : SUCCESS in 1h 44m 34s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0a6f117638df4785b5dc68e298d109b0 : SUCCESS in 1h 03m 19s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/a15a5cadb2974e32b8a1ddaba567cbe1 : SUCCESS in 1h 19m 36s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/218435c873564a59894d91f8087bd112 : SUCCESS in 1h 15m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/77b9d997bef646ca85e83cd3cd5e763f : SUCCESS in 45m 53s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6130f3a9b3124967a4bbe1761f303e20 : SUCCESS in 1h 22m 58s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/cf144601addc4fc4b737dad3889fc33d : SUCCESS in 39m 02s","accounts_in_message":[],"_revision_number":3},{"id":"1aa9585d2a44ea52d3fb24457c66e8b0eeb6dad2","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-29 11:27:49.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"083bd73bece24c23691e0b5f41d17cd735f4b03b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-29 13:05:51.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/89a57826c7804a73a1c840a3d229cdab : SUCCESS in 51m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8d89a3b989bb49bd972cf2c9cd45627e : SUCCESS in 59m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b434b2823a8f43aea72b3c68a8586f1e : SUCCESS in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/09faf82a0983404daf8a6e0f9ae561a5 : SUCCESS in 6m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2d425c2fe73a43198db2451b95c321d7 : SUCCESS in 6m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f67c1f5181dd478b9e26d9751a6c482b : SUCCESS in 6m 30s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/739030f2d5c346c381290724ca805114 : SUCCESS in 9m 30s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/970101cf40e84b43bdc3136da5dff3a0 : SUCCESS in 9m 49s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/05edbde2b5684d3aa0db8256465ab6eb : SUCCESS in 10m 21s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/afb1edf41cdf45bb8e973f611e11c56c : SUCCESS in 5m 31s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/6317989af6e24f1fa89ba2e1b4f3141e : SUCCESS in 1h 13m 20s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f95d0ff9c99a4c06ade91e904275d199 : SUCCESS in 1h 34m 45s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/e640fd0e694f48cd8306e9b5b017923c : SUCCESS in 1h 06m 35s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7adc3f28882a45cab4b842b5ebfd75c9 : FAILURE in 1h 30m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/43a55906a59e44bb90829b0302821afb : SUCCESS in 53m 00s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9b30531b3688485088fe4f0cb6173b22 : SUCCESS in 1h 15m 13s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/cbe9382fbcf643dd9e510e5ac528b49e : SUCCESS in 25m 02s","accounts_in_message":[],"_revision_number":3},{"id":"06c72ed127885e04f9e974d160b6da555c8a3053","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-29 13:18:11.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"47dc51960a381fab0844983a3201793456f3de2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-29 14:50:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/a4b12ae976be4a13b9379604cbcd1513 : SUCCESS in 1h 04m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/db9f1f0d9fa64ec691adadee3076d3a8 : SUCCESS in 1h 17m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e98271b27966463da8059ff76d0aa345 : SUCCESS in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4ccd2bcbcf1d4e37bdb7e3678b7598b4 : SUCCESS in 5m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/97a35ae3993945a692029a378e82f428 : SUCCESS in 6m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/839181910e3c46ca802f8436b39f4de7 : SUCCESS in 6m 30s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c63a648f3b7e4777b135707615a7291a : SUCCESS in 9m 15s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6a1f1bd6522943f8b81f8c89d62ff4fe : SUCCESS in 11m 32s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/53c112b3dc31402790f03e325ad63ac1 : SUCCESS in 10m 14s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/0eb3805c4d124f0a881d3c83fd8222ed : SUCCESS in 5m 07s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/db3d75de5e35400db069be691b8fdcc6 : SUCCESS in 1h 21m 23s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/59baa399fb9a4de682fe40824d2a1b2a : SUCCESS in 1h 19m 47s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/16793c592de440509100ad0f54efd8f9 : SUCCESS in 1h 05m 32s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d7971bc7a7f946ea9a12f88c2d4a24c3 : SUCCESS in 1h 24m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ae86fa2cd0104de38ac187bc9f0c86dd : SUCCESS in 48m 55s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/551d05df8a3a4bdebb88d1ddeea29dfa : SUCCESS in 1h 15m 11s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/794a50bf985b4e3caf408f28aeeae62c : SUCCESS in 40m 54s","accounts_in_message":[],"_revision_number":3},{"id":"045755e6d4d223a7a76b9fb901706b9050e5fa82","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-11-30 05:22:19.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"5ff7cc06df342d58d90e8c78b9726afa8eb6f291","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2021-11-30 05:48:50.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"5a289d3451db850de9b86d52ad5dc6f6026ba000","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-12-02 06:50:58.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/819599/819599/3/check-not-vote/ext-glance-zuul/63388e3 : FAILURE in 1m 38s","accounts_in_message":[],"_revision_number":3},{"id":"71f68de23d51e206fb8b24adc640e8d8b068c3c3","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2021-12-20 08:25:36.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"02c66d3381a5a22719fdbce3b8a0e77e307bff01","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-05 10:40:55.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b416fcf95a219f414a90aa3c58032d190251c49d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-01-05 11:55:50.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/819599/819599/4/check-not-vote/ext-glance-zuul/008f303 : FAILURE in 24s","accounts_in_message":[],"_revision_number":4},{"id":"5a134e1183952aebc26bc56a68450e2f5bdb3685","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-05 12:21:46.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/4281bca8a4974b48b260414a86a52188 : SUCCESS in 1h 10m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/494a54fa51724f668f7a28a4f3d26373 : SUCCESS in 1h 10m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d1ca1ad45ff64a96b146838322619d51 : SUCCESS in 48m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ed0e3a50bc74550a9c8ba4ee6cb6aab : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/31fb3869a1584da2a72087132b270376 : SUCCESS in 6m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/81b24e5b20df47d5bc658d5f66a3671b : SUCCESS in 5m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/947d6267a88f450ca87ac55d00d8f8d1 : SUCCESS in 6m 45s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ba5c285e40a84c27bf9efc676d8402e9 : SUCCESS in 8m 50s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/83c28fa88e514ffe994fe0e9ea251401 : SUCCESS in 8m 25s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/a04c4ef0051e48a6a1dbb14a7284a358 : SUCCESS in 8m 45s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/7894e4e881a349a38443b5b3738dc549 : SUCCESS in 4m 58s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/b414c5d253e44544b9dedeafa7f8a881 : SUCCESS in 1h 18m 05s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d7c191085bbc41009fa0ba545f060c88 : SUCCESS in 1h 16m 10s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/19799b9efafb4055bee241674aaff1f3 : SUCCESS in 1h 15m 30s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/9c79901538d0499495ddb2c000fec279 : SUCCESS in 1h 38m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0f628fa71ad14651bf79d4bd6ca483dd : SUCCESS in 51m 03s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d63c770d759a4ff79c7d78b32cd7b319 : SUCCESS in 1h 20m 36s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3580cc582811423ba06e994fc92735d7 : SUCCESS in 36m 02s","accounts_in_message":[],"_revision_number":4},{"id":"34a00994d26f5e2ea2c5fdab81775597f02ebfe3","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-01-10 05:19:51.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"e48808f7342cbfd73f3c69534cfbe299c88718f2","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-01-19 01:58:31.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7f8d0bb7983c4b5b74fa827acbc4444baedebc9a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-19 02:36:21.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"32baae8747862ea10c4c8f32a4ce5da50ed84244","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-19 02:38:27.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"bb55b7b0c966d04eefa456b27ad0efc7d8787296","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 04:25:17.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/fc7272e2d2394d76b734b0eb22d8c398 : SUCCESS in 6m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/4d673726e6c44ea18e8fab370b51f480 : SUCCESS in 56m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6b1cd035748b4cd3865f8f04c995bc2d : SUCCESS in 1h 08m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c31e957a67354203b91b76ee5773384c : SUCCESS in 51m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b60a911debf645f4ac3de87f81ec43c0 : SUCCESS in 7m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/226ae13e44044fb782b8ed16165723bb : SUCCESS in 7m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ada5cde840c441ffbb02a9f4380a24e7 : SUCCESS in 8m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0117b6c8400f4bbca28a45f1b87fa59f : SUCCESS in 7m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ac5b61733e32491bba3b62b06cd91346 : SUCCESS in 7m 48s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/55e40d2ee7344ea68771b48be810df3f : TIMED_OUT in 31m 41s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a723febce1c4401ba819e58ebb0d9924 : SUCCESS in 9m 50s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/3b4214db2f2a449d846f05bad52d3976 : SUCCESS in 13m 11s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/43a7fae24e484d5088da24d0da02aa6f : SUCCESS in 7m 16s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/501e1da99f0b48fd97c28d55ad93afb7 : SUCCESS in 1h 26m 07s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4515a981da8749d5b7b483ba51aeaff3 : SUCCESS in 1h 21m 59s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/a6c4023762a54d72b50d62a363c5dbde : SUCCESS in 1h 11m 05s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/4c36af1a6c5545a681075c54530d9170 : SUCCESS in 1h 26m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/52aa0871dd5d404399bc71f7eb70eec9 : SUCCESS in 56m 54s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/56c274ed437f4512a41b366009b34ba7 : POST_FAILURE in 1h 11m 41s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f4b527f31d664b06af6f82530b13ca13 : SUCCESS in 34m 41s","accounts_in_message":[],"_revision_number":5},{"id":"a7881aa44d5844786d7db1d643d3b4f378cb81b4","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-19 04:39:38.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"98d17443a16754510202a4a2b89aa97c03b53591","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-01-19 05:14:33.000000000","message":"Patch Set 5:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/819599/819599/5/check-not-vote/ext-glance-zuul/3a1b0fa : FAILURE in 17s","accounts_in_message":[],"_revision_number":5},{"id":"8ac888b041a9be0555ee92265cf3cc660841b9b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 06:10:29.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e1ce58c2b4ee4d6e8098dc201d79df88 : SUCCESS in 5m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/379bf8bdf3324692a8b7a80ffb470dd2 : SUCCESS in 55m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a8f01b97e62f49cdb64ae870628b7776 : SUCCESS in 1h 10m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/87c861a4ee32495a98ace560926fba85 : SUCCESS in 53m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d8ab88b4e21429684e2e144161e3fc9 : SUCCESS in 7m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eba98d611e534340b542396c9d2b1624 : SUCCESS in 6m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/af9d415230bd482985d2f0242f69b5bc : SUCCESS in 7m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/217ad7aa16f04b7f906e75d21f452639 : SUCCESS in 7m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/55b7f43829894f8499673add4a6cc5fb : SUCCESS in 7m 19s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7c7ddc18acd040988a4d773df93129ca : SUCCESS in 10m 12s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4d45116a1a8e42c6b8f05c0f792517b5 : SUCCESS in 14m 38s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/35b3a20b597b4f9abbe6bbea05ab14ea : SUCCESS in 14m 07s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/06137fc5c06942c1b9fcf855344d606d : SUCCESS in 5m 48s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/b94c0746705b491a8f2832461f3ab492 : SUCCESS in 1h 04m 59s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/853b96f1438b44899a8e33663d1b316e : SUCCESS in 1h 16m 20s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/4082ccc60f4d4d32a47cc3974a77ad8d : SUCCESS in 1h 24m 41s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/35ffac41be0842599350ec761ea14f41 : SUCCESS in 1h 25m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/be1504e04fef43eb9ef75088c2d6be2d : SUCCESS in 46m 11s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4ad77e9a77e34e569e09eac2a4f65374 : SUCCESS in 1h 17m 44s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a7bbecfe63f549efbe81e383adc8ab09 : SUCCESS in 31m 08s","accounts_in_message":[],"_revision_number":5},{"id":"437fe8208ebf1f1c2721e5dbc8ec6c3ca454fa3d","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-01-19 16:03:08.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"73ef5c10495caf9d840110cd13508e47ff68b544","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-01-20 05:41:30.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"71b7663570ab1547f363fce4a668aeb3116f0810","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-20 05:42:13.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"686335d76686ed45648ff83bea1272a73fd0b0e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-20 07:15:51.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/a00ef0924cf646119b5e9b5b17286932 : SUCCESS in 5m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/aab9358836b44e03bfba18a85fb8cd57 : SUCCESS in 49m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/55c31895e01b48408b880762b4ed4ebd : SUCCESS in 48m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/935b137dc2dd4809bdb65d07d6ecbc59 : SUCCESS in 35m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25a0aa1e6c9d411c8e589d8f9e846c68 : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/59ae423bf217462aa57c27f3bc2a43aa : SUCCESS in 5m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/570bd32c670a41feb084e7985920cddd : SUCCESS in 7m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3f854b0e51a4960a2a4d5b24e2250a8 : SUCCESS in 5m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/84bec3a491e04e4aa50e3db229d52598 : SUCCESS in 4m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/dafe16ce183f453990b3c0db41aa2fe1 : SUCCESS in 8m 34s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4dbbbc97516e49f28577f452a470d4e0 : SUCCESS in 8m 03s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4c7fbb6e37d24464bb112ca552094fd8 : SUCCESS in 1h 17m 28s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/689f73979a024f7caffd97fa80ec8f03 : SUCCESS in 1h 13m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bb8ebff8c9744d7d94e2371a8bcd1083 : SUCCESS in 56m 47s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/45b1b1d08b204774bf29dd268f38b6dd : SUCCESS in 1h 29m 11s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2d6841ecd008457c988f1afd621d870e : SUCCESS in 38m 20s","accounts_in_message":[],"_revision_number":5},{"id":"a2e1d5260eaa059aa5e1027fd71264b99142c615","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-20 07:15:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"3978e0b822aa4c90789f226e658622181e881613","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-20 07:17:22.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/304285aa6df545be9009eff9704ff3c3 : SUCCESS in 56s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8bfa56564cb488b8e89830b8553b24c : SUCCESS in 59s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/37948980ece941d0ab9acf660f27576a : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"0a5b5117291ff1fba4838ac873a4cfb7225c8ea7","revisions":{"17377064ed46b8ef4065252f02218fc812e374d5":{"kind":"REWORK","_number":1,"created":"2021-11-29 07:38:58.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/99/819599/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/99/819599/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/819599/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/819599/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/819599/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/99/819599/1"}}},"commit":{"parents":[{"commit":"6fd949a966fea443bb4b0b9351b3fa04662a7ecb","subject":"Refactor glance cinder tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6fd949a966fea443bb4b0b9351b3fa04662a7ecb"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-29 07:37:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-29 07:38:49.000000000","tz":540},"subject":"Use LOG.warning instead of deprecated LOG.warn","message":"Use LOG.warning instead of deprecated LOG.warn\n\nThe LOG.warn method is deprecated[1] and the LOG.warning method should\nbe used instead.\n\n[1] https://docs.python.org/3/library/logging.html#logging.warning\n\nChange-Id: Ie0cac63dedf69728392f293f4551e495aebf4d40\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/17377064ed46b8ef4065252f02218fc812e374d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/17377064ed46b8ef4065252f02218fc812e374d5"}]},"branch":"refs/heads/master"},"d98c3b9e5a841a37d3394b4ce8576c1f25b3c7c7":{"kind":"REWORK","_number":2,"created":"2021-11-29 08:08:15.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/99/819599/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/99/819599/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/819599/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/819599/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/819599/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/99/819599/2"}}},"commit":{"parents":[{"commit":"6fd949a966fea443bb4b0b9351b3fa04662a7ecb","subject":"Refactor glance cinder tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6fd949a966fea443bb4b0b9351b3fa04662a7ecb"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-29 07:37:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-29 08:07:43.000000000","tz":540},"subject":"Use LOG.warning instead of deprecated LOG.warn","message":"Use LOG.warning instead of deprecated LOG.warn\n\nThe LOG.warn method is deprecated[1] and the LOG.warning method should\nbe used instead.\n\n[1] https://docs.python.org/3/library/logging.html#logging.warning\n\nChange-Id: Ie0cac63dedf69728392f293f4551e495aebf4d40\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d98c3b9e5a841a37d3394b4ce8576c1f25b3c7c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d98c3b9e5a841a37d3394b4ce8576c1f25b3c7c7"}]},"branch":"refs/heads/master"},"054216304ab1be1df1004f1ac6c78dd724e40c50":{"kind":"REWORK","_number":3,"created":"2021-11-29 08:08:48.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/99/819599/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/99/819599/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/819599/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/819599/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/819599/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/99/819599/3"}}},"commit":{"parents":[{"commit":"6fd949a966fea443bb4b0b9351b3fa04662a7ecb","subject":"Refactor glance cinder tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6fd949a966fea443bb4b0b9351b3fa04662a7ecb"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-29 07:37:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-29 08:08:36.000000000","tz":540},"subject":"Use LOG.warning instead of deprecated LOG.warn","message":"Use LOG.warning instead of deprecated LOG.warn\n\nThe LOG.warn method is deprecated[1] and the LOG.warning method should\nbe used instead.\n\n[1] https://docs.python.org/3/library/logging.html#logging.warning\n\nChange-Id: Ie0cac63dedf69728392f293f4551e495aebf4d40\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/054216304ab1be1df1004f1ac6c78dd724e40c50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/054216304ab1be1df1004f1ac6c78dd724e40c50"}]},"branch":"refs/heads/master"},"be0d895246af7c7e2b246c64055b6e5aec367128":{"kind":"REWORK","_number":4,"created":"2022-01-05 10:40:55.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/99/819599/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/99/819599/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/819599/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/819599/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/819599/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/99/819599/4"}}},"commit":{"parents":[{"commit":"502fa0ffc8970c087c5924742231812c0da6a114","subject":"Remove six.iteritems usage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/502fa0ffc8970c087c5924742231812c0da6a114"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-29 07:37:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-05 10:40:35.000000000","tz":540},"subject":"Use LOG.warning instead of deprecated LOG.warn","message":"Use LOG.warning instead of deprecated LOG.warn\n\nThe LOG.warn method is deprecated[1] and the LOG.warning method should\nbe used instead.\n\n[1] https://docs.python.org/3/library/logging.html#logging.warning\n\nChange-Id: Ie0cac63dedf69728392f293f4551e495aebf4d40\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/be0d895246af7c7e2b246c64055b6e5aec367128"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/be0d895246af7c7e2b246c64055b6e5aec367128"}]},"branch":"refs/heads/master"},"0a5b5117291ff1fba4838ac873a4cfb7225c8ea7":{"kind":"REWORK","_number":5,"created":"2022-01-19 02:36:21.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/99/819599/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/99/819599/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/819599/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/819599/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/819599/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/99/819599/5"}}},"commit":{"parents":[{"commit":"502fa0ffc8970c087c5924742231812c0da6a114","subject":"Remove six.iteritems usage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/502fa0ffc8970c087c5924742231812c0da6a114"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-29 07:37:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-19 02:36:12.000000000","tz":540},"subject":"Use LOG.warning instead of deprecated LOG.warn","message":"Use LOG.warning instead of deprecated LOG.warn\n\nThe LOG.warn method is deprecated[1] and the LOG.warning method should\nbe used instead.\n\n[1] https://docs.python.org/3/library/logging.html#logging.warning\n\nChange-Id: Ie0cac63dedf69728392f293f4551e495aebf4d40\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0a5b5117291ff1fba4838ac873a4cfb7225c8ea7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0a5b5117291ff1fba4838ac873a4cfb7225c8ea7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
