)]}'
{"id":"openstack%2Fglance~821379","triplet_id":"openstack%2Fglance~master~I3265ba6169bf0788a6b523057342db9b4e0e8b7b","project":"openstack/glance","branch":"master","topic":"secure-rbac","attention_set":{},"removed_from_attention_set":{"19138":{"account":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"last_update":"2026-07-07 13:25:06.000000000","reason":"Change was abandoned"},"5046":{"account":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"last_update":"2026-07-07 13:25:06.000000000","reason":"Change was abandoned"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2023-03-30 06:18:42.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2026-07-07 13:25:06.000000000","reason":"Change was abandoned"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2026-07-07 13:25:06.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I3265ba6169bf0788a6b523057342db9b4e0e8b7b","subject":"PoC SRBAC project-manager support","status":"ABANDONED","created":"2021-12-10 09:43:18.000000000","updated":"2026-07-07 13:25:06.000000000","total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"31683c8981342dda898174e1d408f48c641bc773","_number":821379,"virtual_id_number":821379,"owner":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-12-13 12:07:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":-2,"date":"2023-03-30 06:18:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-10 11:04:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-13 08:19:11.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2021-12-16 15:45:05.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2021-12-16 15:45:05.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"REVIEWER"},{"updated":"2021-12-16 15:45:05.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"}],"messages":[{"id":"d3c84a00a5864039c757c9670951b62919cc85e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-12-10 09:43:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"de073c49695cc5b3b75dc61040d00a640cb09536","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-10 11:04:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/bc05d7a5756e432da95b3fe7cf209da8 : SUCCESS in 57m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bb46465ca4df405aa98a1685ae50e112 : SUCCESS in 1h 13m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4e4082abd8dc481fb3a27f7137f847bd : SUCCESS in 55m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c79cdb141114a71b68873d7370b7ff9 : SUCCESS in 5m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2d1a373bf45d4ce4a1136575fdbd78ed : SUCCESS in 6m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/041581f886e343e6952de54b56ee0230 : SUCCESS in 6m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98f1cdb8669e4cf2b65e228035d8b264 : SUCCESS in 6m 36s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f20dceb2c4a64037a6202f70f0c8559f : SUCCESS in 9m 25s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/72dd0f6c68284343860a7a320a28885c : SUCCESS in 8m 19s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/51489677ee8f4e16b1d3334bfcb44591 : SUCCESS in 10m 26s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/c95f7275940f4e1c932f25e05a8c46e5 : SUCCESS in 4m 47s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/024b568147f34d7381fe4b9961ba2901 : FAILURE in 23m 16s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c48fadbda466440e97c5f762fd48546e : SUCCESS in 1h 16m 03s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/ee6f754e721e43caa7de6222c382b2f4 : SUCCESS in 1h 05m 59s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/af069d3d406d46c692ebbe1f7a77f357 : SUCCESS in 1h 13m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f0ac5cc9d5f043dab2b430ca7bb84465 : SUCCESS in 1h 19m 24s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/dd583c597b6c4183af5c50e18b17b93f : FAILURE in 16m 52s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/4997ce52024344729ceb630672efb0f4 : FAILURE in 38m 49s","accounts_in_message":[],"_revision_number":1},{"id":"945df807b6a42ff268601b9ebb0a98c8865af199","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-12-13 08:19:11.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f1fc10eb914faa31f32a4eb150b41a074d0e526d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-12-13 10:36:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9f51c8a93dae56be615be99627d31abcc2ca2257","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-12-13 10:40:12.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f61dcd509ec5a0254f1eaec0f626b9215de7b860","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-13 12:07:13.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/8f91d0fb3cc245fda1130317f0e81828 : SUCCESS in 1h 00m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/eb1e18f411034c4fb5f116ce414a7714 : SUCCESS in 1h 07m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b03c3534cb0342d4b22b78a6c2049af9 : SUCCESS in 57m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8045a558371542e8897cbd3afad8edd2 : SUCCESS in 5m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/37a7486a77a8444ab85b4b8604b08f00 : SUCCESS in 6m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2cbfd26a42a940faa8d16b5c31bfab65 : SUCCESS in 6m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7255ab2bebf847399349c3f16934c3d2 : SUCCESS in 7m 30s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/97dfae2b4ebe48699448523ef0f0c328 : SUCCESS in 8m 58s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/daadef5db79444738b79bbf2ecf1d033 : SUCCESS in 8m 27s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/2193de9d6619455ba6cfaf84d0bb90af : TIMED_OUT in 31m 23s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/492b4c1d54fa49aba0c55548df46c9df : SUCCESS in 5m 03s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/512bec3aedc644b5b301a26058d41b16 : SUCCESS in 1h 19m 08s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/79f430bd644e46889101ea223011f9ca : SUCCESS in 1h 24m 26s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/9dc90640c9974692b3519a021941b6ae : SUCCESS in 1h 13m 34s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/5a7df1e2e15e4a1ca7654515503ab1a9 : SUCCESS in 1h 21m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/03293c4a7ca0430db7e635a74baf57f2 : SUCCESS in 51m 16s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/bf1f1aaecd71443192e6fdf4979cdf88 : SUCCESS in 1h 18m 47s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c5e40637a81644778992c51943cf157c : FAILURE in 35m 56s","accounts_in_message":[],"_revision_number":2},{"id":"343d5eb1d6ed8f7244b265125941e95a345e9624","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-03-30 06:18:42.000000000","message":"Patch Set 2: Code-Review-2\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"31683c8981342dda898174e1d408f48c641bc773","tag":"autogenerated:gerrit:abandon","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-07-07 13:25:06.000000000","message":"Abandoned\n\nNot needed","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"55a2297bea9cf57b355a4e988ea56325afc8de9b","revisions":{"a995efce3acee81749cb6476f6e0603cdd9f7a0b":{"kind":"REWORK","_number":1,"created":"2021-12-10 09:43:18.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/79/821379/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/79/821379/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/79/821379/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/79/821379/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/79/821379/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/79/821379/1"}}},"commit":{"parents":[{"commit":"4ee7799aa7f6a7172e361392ebb8d3da03e0bf7f","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4ee7799aa7f6a7172e361392ebb8d3da03e0bf7f"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-12-09 13:37:15.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-12-10 09:39:08.000000000","tz":0},"subject":"Poc SRBAC project-manager support","message":"Poc SRBAC project-manager support\n\nAdd support for a project scoped manager for below image\nsharing APIs as per Yoga Secure RBAC community goal,\n1. add_member\n2. delete_member\n3. modify_member\n\nNOTE: The support of default manager role has not been added in\nkeystone yet, it\u0027s currently in progress.\n\nTo test this, create new role \u0027manager\u0027 in keystone and add it to\nhierarchy which should be like,\nreader --\u003e member --\u003e manager --\u003e admin\n\nDepends-On: Ida889aa30d462443b801c0f524c51f54b8b756d5\nChange-Id: I3265ba6169bf0788a6b523057342db9b4e0e8b7b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a995efce3acee81749cb6476f6e0603cdd9f7a0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a995efce3acee81749cb6476f6e0603cdd9f7a0b"}]},"branch":"refs/heads/master"},"55a2297bea9cf57b355a4e988ea56325afc8de9b":{"kind":"REWORK","_number":2,"created":"2021-12-13 10:36:07.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/79/821379/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/79/821379/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/79/821379/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/79/821379/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/79/821379/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/79/821379/2"}}},"commit":{"parents":[{"commit":"4ee7799aa7f6a7172e361392ebb8d3da03e0bf7f","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4ee7799aa7f6a7172e361392ebb8d3da03e0bf7f"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-12-09 13:37:15.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-12-13 10:29:52.000000000","tz":0},"subject":"PoC SRBAC project-manager support","message":"PoC SRBAC project-manager support\n\nAdd support for a project scoped manager for below image\nsharing APIs as per Yoga Secure RBAC community goal,\n1. add_member\n2. delete_member\n3. modify_member\n\nNOTE: The existing tempest plugin tests are failing here because\ncurrently the support of manager role and it\u0027s implication is not\nthere in devstack.\n\nChange-Id: I3265ba6169bf0788a6b523057342db9b4e0e8b7b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/55a2297bea9cf57b355a4e988ea56325afc8de9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/55a2297bea9cf57b355a4e988ea56325afc8de9b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Code-Review\u003dMIN"],"failing_atoms":["label:Code-Review\u003dMAX"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
