)]}'
{"id":"openstack%2Fglance~822731","triplet_id":"openstack%2Fglance~master~If0d2ad8357ea759638c6b83c04ec7a2f3cf9f6bd","project":"openstack/glance","branch":"master","topic":"remove-six","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2022-01-04 19:24:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If0d2ad8357ea759638c6b83c04ec7a2f3cf9f6bd","subject":"Remove six.add_metaclass usage","status":"MERGED","created":"2021-12-22 16:14:34.000000000","updated":"2022-01-04 19:29:29.000000000","submitted":"2022-01-04 19:24:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"822731-remove-six","meta_rev_id":"eb0020a65e0a930d36d8efe7916a9362400efed4","_number":822731,"virtual_id_number":822731,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-04 19:24:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2022-01-04 07:09:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2022-01-04 17:54:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-01-04 11:31:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2022-01-04 17:54:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-22 16:15:20.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-12-22 18:01:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-04 04:12:42.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2022-01-04 07:09:40.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2022-01-04 11:31:30.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"},{"updated":"2022-01-04 17:54:04.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"}],"messages":[{"id":"50fbd201a5b94ab3acb0258087f875476127e1b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-12-22 16:14:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"197ef154c7108573ef9f32a3096b21effeb34b1d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-12-22 16:15:20.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":1},{"id":"b7592c431e2434889791283754191f81292417d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 18:01:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/03df9be25767419bac3d11faabde5ad7 : SUCCESS in 1h 02m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8ae481bb83de456ebe2c0b6ae524bd04 : SUCCESS in 1h 33m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7bacef39896840c29e4e8e9ba9764f7d : SUCCESS in 56m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d62aa4b1afb6471a8148fb5326662fab : SUCCESS in 6m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9c1e01db2205452ab068a7303c4c193b : SUCCESS in 6m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f3c45530803c43a9b8d72ef5cceef6f8 : SUCCESS in 6m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10ebff0bc14841e399e71f8c39e2f6cb : SUCCESS in 10m 43s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d1108994ca4a48f5b3695d6d380751a2 : SUCCESS in 9m 53s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/94abc18e13134e638a4235c3bd9b9b34 : SUCCESS in 10m 31s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/58b741df24414f5594b5c8e28ad8dae4 : SUCCESS in 9m 50s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/bf72cbcdd8e34afc97511b24b61bf612 : SUCCESS in 5m 42s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/55fded81a3724edebf5f0766459d4ed5 : SUCCESS in 1h 24m 21s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/06335c88c1954fd2856d26fddc54ed4e : SUCCESS in 1h 14m 02s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/c39c7fa6a5e34d1595ac04c04bf160ba : SUCCESS in 1h 17m 10s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/949509efbd954b23acf54423b3ca469d : FAILURE in 1h 17m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/21c78f4f52e54ee794460911fedeffe7 : SUCCESS in 52m 08s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f26f5ac9d7bb4b9380c8e0e7741a3874 : FAILURE in 32m 04s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3f8e98d3b10d4696a1575c65c4245a79 : SUCCESS in 36m 14s","accounts_in_message":[],"_revision_number":1},{"id":"4b37b7bc9eb1278d0726c231d058992301984dee","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-01-04 04:12:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8d93cfa034757e52c04e123379b3b5248ba59c9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-04 05:35:28.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/468e66c64019464494577723679b10aa : SUCCESS in 47m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/26b7d45df62d40b7ab7b51a747d14768 : SUCCESS in 1h 03m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c0e43afc8fe848458499ee8dcaba8fb0 : SUCCESS in 44m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a326498cab4b4dee9fac69b2af28b237 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/41488b4b6e034c16bf78e78a2c0abdbf : SUCCESS in 5m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/432561a84eb64be1b8d3e6e4b04640c0 : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5292654974be4ded8552fdf54994ee19 : SUCCESS in 5m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0f43c14a79264e349f5d720625c8fd35 : SUCCESS in 9m 00s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/59c068bede104a92a56205f104e5ab3f : TIMED_OUT in 32m 10s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/799572bd78054193a154e97c808d20f3 : SUCCESS in 8m 05s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/29d08f424fb34a109ddca597c873698f : SUCCESS in 5m 49s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/9d66faf0efb6498490c98480e8d803c2 : SUCCESS in 1h 21m 24s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/429ab7a32f554b8c8e0a102ca820ad9b : SUCCESS in 54m 01s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/7886e2835f75480695e61d384cbe0a00 : SUCCESS in 1h 07m 02s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e52281e3cbb944e3af8e1086ade05711 : FAILURE in 1h 03m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8a37582de5a7484aab272e06ea7b7d15 : SUCCESS in 52m 19s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6e0f9572ec9e4321b1cf04aea7be3c3b : SUCCESS in 1h 09m 21s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/cc85a98cff114295bb4489c43d7a085a : SUCCESS in 22m 16s","accounts_in_message":[],"_revision_number":1},{"id":"50da9c87176312f37875b999af44a39a4e4964c5","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-01-04 05:37:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cd5cfa294a91b6cacbaa9bbc64b3a1c0006394d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-04 07:06:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/78a746cb1ca9491bb1e37faa9e091f2a : SUCCESS in 1h 02m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d24752c68ced4d5390c1dea5554f8b27 : SUCCESS in 1h 23m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/26c3d70a0b6d42089ece620f87552155 : SUCCESS in 49m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fab0b27bfbc64547ae602e2f05fbfab5 : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f37b6d7b4f074bcca80f958a6b263839 : SUCCESS in 7m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a739ae90139d4bbda6876f2c533ad211 : SUCCESS in 6m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e68114de48f0423994a297981a42d4dd : SUCCESS in 6m 18s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4541fac1458d4c68bc7167859b8ee146 : SUCCESS in 8m 59s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/df560e54b5594c3f80bdd6f17942696f : SUCCESS in 8m 37s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/afd4f77120dc44e3a39ea2ff68c2a2d3 : SUCCESS in 9m 04s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/04d16d2285224ecf8c714b81c482ed59 : SUCCESS in 5m 00s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/a2e9e29ae392433986ac8be15598cf0c : SUCCESS in 1h 24m 51s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/97840fc09aa94f87a4f8cc44ecba7064 : SUCCESS in 53m 10s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/6cd0690f691e48a8b07ae545eab24496 : SUCCESS in 1h 00m 20s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/94994d833a864200b0d11f3fb737b51c : SUCCESS in 1h 20m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f51dab7fa48447db849fa0b2408a3db1 : SUCCESS in 39m 04s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8d7f91d81cee451295f2d935e3e53e09 : SUCCESS in 55m 16s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c5201ea3f57044ecac77e9f862f663fc : SUCCESS in 21m 58s","accounts_in_message":[],"_revision_number":1},{"id":"ecd80c1c2915847a9e6968aba4693c791b63ea19","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-01-04 07:09:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d5bc3aadc53608058f1be8dffa5baf34531a4264","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2022-01-04 11:31:30.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"16198e649dc0d6496cb76a1e47956ed80aa04a02","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-01-04 17:54:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c6f7a45f99748352fdc65fa4bec79557b7b5d787","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-04 17:55:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"23f7950b0c7185d8d35f9f9a93d3e1589c89dc44","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-04 19:24:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/7aff45c34f3446058d59ba0745fed5cb : SUCCESS in 59m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/72081aa3384141f3ba8d5c7dad2beacc : SUCCESS in 1h 13m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b7d51b1bc351496b9011db6daa3beea2 : SUCCESS in 48m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d17aa4c171b4ccc816872a88f54d27d : SUCCESS in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dda0f45f1a2246b3bacb28797e956baa : SUCCESS in 6m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/707fa0d366734b26b134d75e41e867ff : SUCCESS in 8m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/011885ab489440f78dbaac9d7164c689 : SUCCESS in 7m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/96ddf8dd6c624d3eb9a9e18f2dae101e : SUCCESS in 11m 33s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b6f789e9815c49a289d9d627672f60d2 : SUCCESS in 9m 53s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/70c5abcc897a44dea2988d96bd513dfb : SUCCESS in 1h 17m 16s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/e969aa84622147129926c643a2a12d2f : SUCCESS in 1h 13m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f472ba759a574b48ae5398d25595c8de : SUCCESS in 54m 27s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/de25a68d19d747ee9e8a0e2616361b87 : SUCCESS in 1h 18m 57s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/900e410b13f744bca9f79c3497f9efc4 : SUCCESS in 37m 23s","accounts_in_message":[],"_revision_number":1},{"id":"e8cca07a7a2dbe7ff651449b2f4578ded1154670","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-04 19:24:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"eb0020a65e0a930d36d8efe7916a9362400efed4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-04 19:29:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0404a11dce9446a79cb10cba2d84d2b1 : SUCCESS in 1m 26s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0562e3335bd5ee5b51a56cfe002c1895b52ef52a","revisions":{"0562e3335bd5ee5b51a56cfe002c1895b52ef52a":{"kind":"REWORK","_number":1,"created":"2021-12-22 16:14:34.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/31/822731/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/31/822731/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/31/822731/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/31/822731/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/31/822731/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/31/822731/1"}}},"commit":{"parents":[{"commit":"4ee7799aa7f6a7172e361392ebb8d3da03e0bf7f","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4ee7799aa7f6a7172e361392ebb8d3da03e0bf7f"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2021-12-21 11:37:31.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2021-12-21 11:38:47.000000000","tz":0},"subject":"Remove six.add_metaclass usage","message":"Remove six.add_metaclass usage\n\nChange-Id: If0d2ad8357ea759638c6b83c04ec7a2f3cf9f6bd\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0562e3335bd5ee5b51a56cfe002c1895b52ef52a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0562e3335bd5ee5b51a56cfe002c1895b52ef52a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
