)]}'
{"id":"openstack%2Fglance~822732","triplet_id":"openstack%2Fglance~master~If5b46580078eb756651ac6118f502eccdc693646","project":"openstack/glance","branch":"master","topic":"remove-six","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2022-01-04 19:28:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If5b46580078eb756651ac6118f502eccdc693646","subject":"Remove six.iteritems usage","status":"MERGED","created":"2021-12-22 16:14:34.000000000","updated":"2022-01-04 19:31:00.000000000","submitted":"2022-01-04 19:28:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"822732-remove-six","meta_rev_id":"3e1c609b405e1b297a87d9780c46b27c9ec934a2","_number":822732,"virtual_id_number":822732,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-04 19:28:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2022-01-04 07:09:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2022-01-04 17:59:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-01-04 11:31:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2022-01-04 17:59:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-22 17:50:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-04 07:09:50.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2022-01-04 11:31:11.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"},{"updated":"2022-01-04 17:59:20.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"}],"messages":[{"id":"599e90853254088fcd9a5802625a68d89c0a8e54","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-12-22 16:14:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c0e4ebfefee7fcdda0c68ded2ba024ff1a63d227","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 17:50:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b7ff6e7bfc8b4602b44fc68c26aa0ee8 : SUCCESS in 7m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/d64b611bfab343cfa643c8ca553dad05 : SUCCESS in 1h 05m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fd4b9360e060420dba6146ef61876e37 : SUCCESS in 1h 15m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/01f85f505e7249f9ac1bb80302a72d7a : SUCCESS in 47m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e8d7da4f2154cb19da77ea79cf3f2c6 : SUCCESS in 6m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d92ded1652da41fe901dd0a707ee8a89 : SUCCESS in 5m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ebfa0cc77d0e4398aaf0199279c55545 : SUCCESS in 7m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8cb897f96dfe4c2c8f1382705805786e : SUCCESS in 6m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b37d774d1ef048d38732d385cf590925 : SUCCESS in 6m 58s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/37273a3833a0426c8c6ef28baae539f8 : SUCCESS in 10m 43s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/64fc8d6fe64a452e94b09b16df6a5c25 : SUCCESS in 9m 55s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/15a5adcaf0f3447cba3c8065025476ed : TIMED_OUT in 31m 44s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/192297758bc44ebdbb30d005660fc373 : SUCCESS in 7m 06s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/02d317e53d184143a5a9806f059895c7 : SUCCESS in 1h 21m 26s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/1152ec0801bc4858bf5fdd1e5808640e : SUCCESS in 1h 20m 09s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/ef73b22a64e6487e9c2047494e053263 : SUCCESS in 1h 13m 30s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f4ded5292473408298ce6f37a6e25b77 : SUCCESS in 1h 22m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3f889eca00e34c03857405e760726d82 : SUCCESS in 54m 31s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/200f6809191c43f3a1e8e39a7b410944 : SUCCESS in 1h 26m 39s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/390793d19bc94813aad0fc98b7d9cc15 : SUCCESS in 39m 43s","accounts_in_message":[],"_revision_number":1},{"id":"55c566510b7923f469d0e6fa535d28a1efd3a330","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-01-04 07:09:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8343a7874122810cc23037429de660873222f64f","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2022-01-04 11:31:11.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e4864e6666dacd248034dd029f5e451c32403532","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-01-04 17:59:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"20627cfb12f9357aefcfb0904dc1fb4dc10f1ecf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-04 18:00:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f9872dfc33f28029f88a8dfcb0bb1d055415b9a1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-04 19:28:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/090301360e1e4d43ab7a560a60ab9b74 : SUCCESS in 7m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/46cbef2930df4de7a67eeb4bf14afc34 : SUCCESS in 51m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cf0bf5aa65e74186a025950042b02161 : SUCCESS in 1h 08m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b96b6dd30cb34df8baf5aabf99703456 : SUCCESS in 53m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a24c417eb728403391cf0c19873fd7db : SUCCESS in 7m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4c265a5e46c444f5a61c6c906293bc21 : SUCCESS in 9m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/034fc89bddc048cab468446856541c3d : SUCCESS in 8m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98aca774a03c427489974fbef1ef0e54 : SUCCESS in 9m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c432bd654bf14e67a2392bae01c194c8 : SUCCESS in 6m 21s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ec75296c404c439996c88a36725d8f7f : SUCCESS in 11m 15s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ee19864b21d4470a8e381ad2ddcd3922 : SUCCESS in 9m 15s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/32a4d3f9fd0245c0aac3f3232a19291a : SUCCESS in 1h 10m 25s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/da9bbee2def94579a5a14ccb593b5367 : SUCCESS in 1h 19m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9422c0496ab44e37892375dda6d0dd86 : SUCCESS in 53m 52s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/661b11d5ae6344239871fcc136394681 : SUCCESS in 1h 22m 01s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/329fe19387314779ad867159b01ab851 : SUCCESS in 37m 16s","accounts_in_message":[],"_revision_number":1},{"id":"35d02a25f61211a7a8f0cadec046fca7ce5dee86","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-04 19:28:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3e1c609b405e1b297a87d9780c46b27c9ec934a2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-04 19:31:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/10ecec1b7cb041c0af27e3063b5bc514 : SUCCESS in 1m 04s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8b98271a4a2415881345c66098e7bea : SUCCESS in 1m 12s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1a74a498299b48d59111bffed816408d : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"502fa0ffc8970c087c5924742231812c0da6a114","revisions":{"502fa0ffc8970c087c5924742231812c0da6a114":{"kind":"REWORK","_number":1,"created":"2021-12-22 16:14:34.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/32/822732/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/32/822732/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/822732/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/822732/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/822732/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/32/822732/1"}}},"commit":{"parents":[{"commit":"0562e3335bd5ee5b51a56cfe002c1895b52ef52a","subject":"Remove six.add_metaclass usage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0562e3335bd5ee5b51a56cfe002c1895b52ef52a"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2021-12-21 11:42:17.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2021-12-21 13:01:56.000000000","tz":0},"subject":"Remove six.iteritems usage","message":"Remove six.iteritems usage\n\nWe also remove hacking tests for this, along with those for iterkeys and\nitervalues (no usage of these).\n\nChange-Id: If5b46580078eb756651ac6118f502eccdc693646\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/502fa0ffc8970c087c5924742231812c0da6a114"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/502fa0ffc8970c087c5924742231812c0da6a114"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
