)]}'
{"id":"openstack%2Fglance~822733","triplet_id":"openstack%2Fglance~master~I2ed464202f8b645aed11490e111c61d3c7423c11","project":"openstack/glance","branch":"master","topic":"remove-six","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2022-02-02 21:55:05.000000000","reason":"Change was submitted"},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2022-01-27 17:50:36.000000000","reason":"Dan Smith replied on the change"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2022-02-02 19:52:32.000000000","reason":"Cyril Roelandt replied on the change"}},"hashtags":[],"change_id":"I2ed464202f8b645aed11490e111c61d3c7423c11","subject":"Remove six.text_type, six.binary_type usage","status":"MERGED","created":"2021-12-22 16:14:34.000000000","updated":"2022-02-02 21:57:23.000000000","submitted":"2022-02-02 21:55:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"822733-remove-six","meta_rev_id":"290c3545ce6ebc362996b1a6d6580de6aaa5933b","_number":822733,"virtual_id_number":822733,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-02 21:55:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2022-01-27 17:50:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":2,"date":"2022-02-02 19:52:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":1,"date":"2022-02-02 19:52:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-22 17:41:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-04 07:09:57.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2022-01-04 11:31:03.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"},{"updated":"2022-01-04 18:14:59.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2022-01-26 20:03:35.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2022-01-28 03:53:37.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"4447c532cac747b0e785971daef1a44b02ff4383","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-12-22 16:14:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a0280bc092b77d83fa52b5ab8a2c908fa1f6f872","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 17:41:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/7955c9f4e11841cd869c9ce8545f9abb : SUCCESS in 6m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/e4862bfbf0c648098014d2dbc377e563 : SUCCESS in 1h 01m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d4930e0e5a7b490e981ffcec8de12470 : SUCCESS in 1h 14m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/12373e9b3cfc40c49b7e569fb1f73988 : SUCCESS in 49m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc28997aded44c0695d9fbc059dbfba2 : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ba42f51d5e8f4fc9889208c51959bf39 : SUCCESS in 7m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c6b40bc3fb7943efa5d31c498e696f8e : SUCCESS in 6m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e83c6a8b7f674b0f8ad8f36c84f02c9d : SUCCESS in 7m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/877fc3572a274998af8a7e7b22160a5e : SUCCESS in 7m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4473d851054848f4a8d673f824ab982d : SUCCESS in 9m 40s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/fa76f992da0e4698b3b69debb6c2e7eb : SUCCESS in 9m 26s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/3045dc4295554b2eb406680d88767e00 : TIMED_OUT in 32m 10s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/7bb9a02d48e545c4aeaeaf65a13ea007 : SUCCESS in 5m 05s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/b1a49ecb540140af91659d8b21def6b9 : SUCCESS in 1h 12m 45s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/64ddf9058db54c349574e71a32e7de87 : SUCCESS in 1h 09m 01s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f0affe264b364a00bb90f9848b383606 : SUCCESS in 1h 09m 18s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/71d9c5a5ab2d4acab732ebce8565243f : SUCCESS in 1h 13m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/291edff40eeb4cf599d8c7d7e65d28f0 : SUCCESS in 55m 13s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ddc40b5fd7e4484286ee1ef61e2588b2 : SUCCESS in 1h 17m 08s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/dcdcb5d7acf948a28b350adc2a548422 : SUCCESS in 33m 19s","accounts_in_message":[],"_revision_number":1},{"id":"5b134478c39fbdc78b4a50dcf8de33dc6dcd8ec0","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-01-04 07:09:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"082cff2df10311866dbd808a834088f1beb50ded","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2022-01-04 11:31:03.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"51f66e0f7d79a6d68398ca4007db9036e56a8b9b","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-01-04 18:14:59.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"7e4faa23b9de68bba9b0e50d40a851d946595236","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-01-26 20:03:35.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"0858f3e6aebcb9c54b1a222f5c0d876c842ea876","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-01-27 14:54:36.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"4714fcd824f66f0125c5b105674e328e39205d11","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-01-27 16:39:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"211fa7689f218236befab8a4eb43cd3a8ed6d97d","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-01-27 17:50:36.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d997041faa4d090760a1bf5674a3b7b2c7d71824","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 19:13:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/23950fa7c4a340e3a277e9e0e9068261 : SUCCESS in 6m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/eeef26ef975a4e76ab54e4689f07201a : SUCCESS in 58m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2d4d52c2df484a0b80b2be066d931bc9 : SUCCESS in 1h 04m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/01556c7781b444beaed3a2f6d7ab4e2a : SUCCESS in 57m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a4e774879fd40dc8f25b6223d45fa39 : SUCCESS in 6m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/57eba54669ab42959d6cbae5249cf056 : SUCCESS in 6m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bd8e63bc009f4c279358eb93ae43fbce : SUCCESS in 6m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7d4a028ab8f4987a09460e3602ea191 : SUCCESS in 8m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7a5353b2a7f74a73a439f66df77ffceb : SUCCESS in 7m 49s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/75739c9ccb014056b842b5120e9de451 : SUCCESS in 11m 31s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/26803565c0e14e94b84641cdfd7c6c78 : SUCCESS in 11m 42s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/cc4198b701fe4ee191da48cc672ea2c7 : SUCCESS in 11m 49s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/bd5b9c5660e14a8fa98fc3284de8f50f : SUCCESS in 6m 05s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/a15b0eaca218448a8c3328e0d124fae1 : SUCCESS in 1h 35m 30s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/37340aa2be2949ad874fcfcf65c32077 : SUCCESS in 1h 19m 36s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/83ce668882fe41009a612e2985bc0499 : SUCCESS in 1h 13m 52s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/cb982f47bd9142e09ac2a6921905e5ef : SUCCESS in 1h 36m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/380faa4d03af468cb5c8813630ce10df : SUCCESS in 55m 26s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9810b2c4cb5340f5ae0f08b4ec486191 : SUCCESS in 1h 22m 34s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/4d73868d5e58475a8d80a82bf12234b8 : SUCCESS in 36m 50s","accounts_in_message":[],"_revision_number":2},{"id":"1be66f3208ff15375374fbd531069babc765b8b4","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-01-28 03:53:37.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/822733/822733/2/check-not-vote/ext-glance-zuul/9926649 : FAILURE in 31s","accounts_in_message":[],"_revision_number":2},{"id":"5e8998614f77feaa142390adf78f2b4845ec0389","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-02-02 19:52:32.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"632b0fa5cc60e16ff72d0fb7dd9fad7405f8de0d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-02 19:53:51.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"dba9ff21447886dbe85cb5350a125aa2eecc3a78","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-02 21:55:05.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/964a7cbc35cc4f9db6e90e12e75c5f36 : SUCCESS in 5m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/8e301c64ebcf4d3499e90a4d824f5126 : SUCCESS in 51m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/77d7dac1d7684e7d9e28c5ddb86d8300 : SUCCESS in 1h 04m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/da423ae9047546fea08eb0a5f8c2a3e7 : SUCCESS in 44m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b54a03344fb454783b26fd6bb04d6b9 : SUCCESS in 6m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/681846f924d84c4fbd330932644b32f0 : SUCCESS in 5m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4c2ef1b96e4340d38aef968da84978d7 : SUCCESS in 6m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88f34bb17fee4f8c9e2e64459fc28153 : SUCCESS in 6m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b4b1edb0012c40e392eb62d676b0795e : SUCCESS in 5m 53s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/eef7639aa4534c10803b30e93290cc8a : SUCCESS in 10m 32s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ec0d2b9f8a494485a56ae9a298f0adb2 : SUCCESS in 8m 38s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6d83a50f134c468ab032f37e6fcf0728 : SUCCESS in 1h 02m 39s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/6d5f9fe0f80547318024540eed68c085 : SUCCESS in 1h 05m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/09b47143ee144cddaf311e0d1bdf254c : SUCCESS in 42m 52s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6af30b0233e54994893f666fa8647a60 : SUCCESS in 1h 06m 03s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/42421d9c136b45e1b5115526d1c29e48 : SUCCESS in 32m 22s","accounts_in_message":[],"_revision_number":2},{"id":"d8202263f17f66dff40d648dde8b16b8a97c2548","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-02 21:55:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"290c3545ce6ebc362996b1a6d6580de6aaa5933b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-02 21:57:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/c0fb6d535a744f959062c419297745db : SUCCESS in 1m 22s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2605579e4094a529a3ec1d57189afaf : SUCCESS in 1m 25s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/20301855af8241c299e0018f273b4865 : SUCCESS in 1m 23s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6bd7c188ee87f65c6ac32aa85f24e50b099e1088","revisions":{"fec3f3c438a1ea3ce10e35bdb4dd81e3173f4d10":{"kind":"REWORK","_number":1,"created":"2021-12-22 16:14:34.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/33/822733/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/33/822733/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/822733/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/822733/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/822733/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/33/822733/1"}}},"commit":{"parents":[{"commit":"502fa0ffc8970c087c5924742231812c0da6a114","subject":"Remove six.iteritems usage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/502fa0ffc8970c087c5924742231812c0da6a114"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2021-12-21 11:50:50.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2021-12-21 13:03:37.000000000","tz":0},"subject":"Remove six.text_type, six.binary_type usage","message":"Remove six.text_type, six.binary_type usage\n\nChange-Id: I2ed464202f8b645aed11490e111c61d3c7423c11\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fec3f3c438a1ea3ce10e35bdb4dd81e3173f4d10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fec3f3c438a1ea3ce10e35bdb4dd81e3173f4d10"}]},"branch":"refs/heads/master"},"6bd7c188ee87f65c6ac32aa85f24e50b099e1088":{"kind":"REWORK","_number":2,"created":"2022-01-27 16:39:44.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/33/822733/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/33/822733/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/822733/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/822733/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/822733/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/33/822733/2"}}},"commit":{"parents":[{"commit":"71c1e4b18a47f94f3302f96c359a6d3e1453ff11","subject":"Bump SQLalchemy to version that supports PY3.9","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/71c1e4b18a47f94f3302f96c359a6d3e1453ff11"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2021-12-21 11:50:50.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2022-01-27 14:54:00.000000000","tz":0},"subject":"Remove six.text_type, six.binary_type usage","message":"Remove six.text_type, six.binary_type usage\n\nChange-Id: I2ed464202f8b645aed11490e111c61d3c7423c11\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6bd7c188ee87f65c6ac32aa85f24e50b099e1088"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6bd7c188ee87f65c6ac32aa85f24e50b099e1088"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
