)]}'
{"id":"openstack%2Fglance~822735","triplet_id":"openstack%2Fglance~master~If3aa008522c24e870b7bf13de32b8ed1b27cb519","project":"openstack/glance","branch":"master","topic":"remove-six","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2022-02-03 08:22:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If3aa008522c24e870b7bf13de32b8ed1b27cb519","subject":"Remove six.moves.range usage","status":"MERGED","created":"2021-12-22 16:14:34.000000000","updated":"2022-02-03 08:23:58.000000000","submitted":"2022-02-03 08:22:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"822735-remove-six","meta_rev_id":"8b44d590b7d0a146799ccb5fcaa3c4e43f1a0365","_number":822735,"virtual_id_number":822735,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-03 08:22:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"recommended":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"all":[{"value":1,"date":"2022-02-03 08:23:58.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":2,"date":"2022-02-03 05:00:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2022-02-02 19:53:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-02-03 08:23:58.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2022-02-02 19:53:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-22 18:09:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-04 07:10:12.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2022-01-04 11:30:42.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"},{"updated":"2022-01-21 09:05:00.000000000","updated_by":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"reviewer":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"state":"REVIEWER"},{"updated":"2022-01-26 20:05:12.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"e44c8c2b0ed94d507cb4f0c0802ada5b67143d4f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-12-22 16:14:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dee045d372e6d85e4410c12cdfe4d574966cbfe8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 18:09:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/20ab59a803224ba7bdc32d4078984a5a : SUCCESS in 54m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f1c9a37044ad4c8bae3b806b49c9c2b9 : SUCCESS in 1h 07m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/936848b5a4554bbfb59e36abfa247c1b : SUCCESS in 50m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6cbb95528ef425b9ecdf349cb76a856 : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9377c69c129e45139bf66aa82296bcd7 : SUCCESS in 7m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7c448df62d854c22a25850518142b6c9 : SUCCESS in 7m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f091e269aba747139f6ef814ebf51756 : SUCCESS in 6m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d524d4e1dd2e4a07a8b975459d265c96 : SUCCESS in 9m 09s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/bc04f11d88964c338b8ebeafcfb4c648 : SUCCESS in 10m 42s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/30db13d685e549e4b9c7b5d53bb467c0 : TIMED_OUT in 31m 34s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/bf77b985bcf54492b122be49ec4e90fa : SUCCESS in 5m 49s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/2a7ea4e4216f4852ad2ffae3027a3f10 : SUCCESS in 1h 44m 43s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7cadd931b1a649baa66056d5c175c27d : SUCCESS in 1h 08m 00s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/5e0ed46983b34c3b914ed318377556bc : SUCCESS in 59m 29s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/ecce5efc0c28455fa8cdb3d71a6d66f5 : SUCCESS in 1h 18m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c3061e4d218e497e8b9cc2b6b4058254 : SUCCESS in 45m 01s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5a379980a51840f5a1332a6bece1d8bd : SUCCESS in 1h 09m 48s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d708240d7cc3421aa784b62b08274a94 : SUCCESS in 31m 30s","accounts_in_message":[],"_revision_number":1},{"id":"7b0f2ba5d6672df4cf2afd36c174b0e6ff076c08","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-01-04 07:10:12.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b82e5ab3b6c7d5267bbf27f3b910200ec18678aa","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2022-01-04 11:30:42.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"57cff0e861304895c09c0416c16644994a2ac7ee","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2022-01-21 09:05:00.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a2105fec5f7878489ae0c0abdee4f8128f9f67b8","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-01-26 20:05:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"469ee6a0060f3a33e4129ff6bc83cdba26be3f4a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-01-27 16:39:44.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"bcdd9bcb578ddd8a59313734b748c0102d9df3a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 19:13:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/a7d0b69802fd4f2abe293278128435bc : SUCCESS in 1h 11m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/84f9dc4fc4024513b348cca037bcb326 : SUCCESS in 1h 07m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4147e40d917d4c47aa54893c027e86d6 : SUCCESS in 40m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2cd600da1d064294a4d98fdf4fb166ae : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4fcd1be01e4749b1851d2723a49fca74 : SUCCESS in 7m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4fa506c29f4e4078bbae0990504f3926 : SUCCESS in 8m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ba91ac77715465db56b934f0541cb14 : SUCCESS in 8m 28s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/822e1950d0f443b09139c801943f7740 : SUCCESS in 9m 15s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a231222b825649e6978058da4fef0ec6 : SUCCESS in 11m 20s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/2fac22f1309244778c905f996c1c2019 : SUCCESS in 11m 31s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/b82f1806ba8641b3bfae91acf9be4214 : SUCCESS in 6m 54s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d6accb7abe2542d98861a7889e035674 : SUCCESS in 1h 35m 59s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ceeada7da9c84d63934c001bea373eed : SUCCESS in 1h 19m 49s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/d1eacef2e30442d59d060dc1921f938f : SUCCESS in 1h 17m 21s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8a119bb44a8c4059a49eb487985550dd : SUCCESS in 1h 38m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/933f75fc8dfe4e8eb6db5fba0c00a381 : SUCCESS in 55m 31s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/791bf858c2af4027a09e6c453da5114a : SUCCESS in 1h 15m 02s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/46458249d80d43b1999bb3eaaeec183b : SUCCESS in 41m 46s","accounts_in_message":[],"_revision_number":2},{"id":"ecd63530767a7ca907cfb550229d18e1c4f210a7","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-02-02 19:53:23.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"353919242ec4009c699c21019610cb2940d87f9a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-02 19:54:40.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"454a8d25f84c4382ba6fa514c970315bb418c684","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-02 22:08:19.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/2355557079414959af9bbe7bd89a15ec : SUCCESS in 48m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b3e650d9b98140b1a16662c26f203a11 : SUCCESS in 56m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3eccb9da2eb84d8a8a8674d8ceb451c5 : SUCCESS in 44m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9408753e1b0c4e68ba49f2a9b9e8f41b : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7961c27f79154871bc287dae832d9921 : SUCCESS in 6m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/af6eab91bc744208a8f53cb7a4a04f40 : SUCCESS in 7m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fabbcb8993ed4a908c36602e4ecedaec : SUCCESS in 6m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a4275ddcf3aa45efb5bdca7bf9a8398f : SUCCESS in 11m 20s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5481ce604e9642f3be91f58be2e14552 : SUCCESS in 10m 36s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/1a87fb51310942879d65c263c682baaf : POST_FAILURE in 19m 12s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/1469a60cb8b9471f96f30d30603fcc15 : SUCCESS in 1h 20m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fbd48dcdcf34405cb21fe1703a4cd18e : SUCCESS in 48m 34s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/cf15879cbc8b4fd8af48c3bd20e193d0 : FAILURE in 1h 09m 49s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e5689c93fe694a31a089ca66ccbd20bb : SUCCESS in 35m 46s","accounts_in_message":[],"_revision_number":2},{"id":"7e19fbfc397efad62d219a9b9a82fe509b625c31","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-02-03 05:00:09.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6fad2909e50702919eed8620191a0c94bdd8405f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 06:46:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/7a0ee40fe9734fbdb08e93f18edb07e4 : SUCCESS in 1h 06m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/dfa8a0c888f344b8ba87f71e3d17a74c : SUCCESS in 1h 45m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2cf87472030942ad89271670a2cb2ae4 : SUCCESS in 55m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c4e991c086f40a8a6ca8f3eb43bb0dd : SUCCESS in 6m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d9b860e807214afd986173add871d3e5 : SUCCESS in 6m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f45f08ae0f5c4b61823bce8bef3f0746 : SUCCESS in 8m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36e13641d9ca402498c9fc94ed60ccbc : SUCCESS in 6m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3ea033fa56154bcc8d2a61deca6993f2 : SUCCESS in 10m 26s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c9397db71b9e426892d1438a97c7dbc1 : SUCCESS in 12m 09s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/90a36308631042d5b86a069cc830c86b : SUCCESS in 10m 36s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/80b23a81d092439487c2dd5d8841499a : SUCCESS in 5m 22s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f4711030dd3b4cfe9001de44b08baf42 : SUCCESS in 1h 11m 07s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/10b4cc0d2f714f85b7b5bedcc10843e2 : SUCCESS in 1h 21m 06s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/2d92d9a9102f400a9b5c0afdb05b76be : SUCCESS in 1h 19m 32s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/9852cb05cbad4e99a0a473b858a84609 : SUCCESS in 1h 25m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c29cc0a201ce4691ae260b220a6b6fb3 : SUCCESS in 52m 34s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/cd86302d41ce40b3b9260640fc6bdf5c : SUCCESS in 1h 16m 06s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/220a27c9babe4a9280351cc17435563d : SUCCESS in 37m 43s","accounts_in_message":[],"_revision_number":2},{"id":"fdcf51ccfd1241e1a4179eaa479aa16564cdb7ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 06:47:57.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ca8fd5a08ee7d18fafc0d96e54814a7c2d375b4a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 08:22:10.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/5cfb47b51f1042f688fb2c3e9603aa7c : SUCCESS in 53m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7ea231040ede4e7f8c3997348ab2bb12 : SUCCESS in 1h 08m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2395381ed7784768950dbbffac94f238 : SUCCESS in 45m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5ad2ffb330b4aa09b67fdcfbfacf2e7 : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3a1bf00da38a416cbefbc2a36f6fcced : SUCCESS in 6m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dc500087d5e646aba16bb8407003224e : SUCCESS in 6m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4177c4172c2b475ab4cd1e2795b74867 : SUCCESS in 6m 06s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/55f595eaf64e4f7c8b6813336e1cc7a1 : SUCCESS in 10m 16s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/61f7ab7ed39b4e74a5b064e21ca4110e : SUCCESS in 10m 26s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/eaba773aa1bb43568f954eaef0b56490 : SUCCESS in 1h 08m 17s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/543fb9b66f86449db047cc4077204262 : SUCCESS in 1h 11m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cd4cd1723e6a425b873aece8d78995e7 : SUCCESS in 48m 19s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d14b27e7cdc8492da2e472dd009a0cf5 : SUCCESS in 1h 10m 45s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1e3b47b13b2f4aee937a6650ad43a18b : SUCCESS in 29m 04s","accounts_in_message":[],"_revision_number":2},{"id":"6920cdbdf84d92a06822abbb065a3488947bd719","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 08:22:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"e577da2bf04c63e02566b79e05405a650eca8851","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 08:23:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5112c3253fa4583b3bd13198c53d56f : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"39e667a14527f194a1905f10bf53eac14cf98f5d","revisions":{"a22e202bdae745774fecc31ef84e783311d473f6":{"kind":"REWORK","_number":1,"created":"2021-12-22 16:14:34.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/35/822735/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/35/822735/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/822735/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/822735/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/822735/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/35/822735/1"}}},"commit":{"parents":[{"commit":"36b9645c9dc8c9252d9fdea46b51dfa5345297c8","subject":"Replace six.BytesIO, six.StringIO usage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/36b9645c9dc8c9252d9fdea46b51dfa5345297c8"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2021-12-21 13:07:51.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2021-12-21 13:07:51.000000000","tz":0},"subject":"Remove six.moves.range usage","message":"Remove six.moves.range usage\n\nThis is the same as the \u0027range\u0027 keyword in Python 3\n\nChange-Id: If3aa008522c24e870b7bf13de32b8ed1b27cb519\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a22e202bdae745774fecc31ef84e783311d473f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a22e202bdae745774fecc31ef84e783311d473f6"}]},"branch":"refs/heads/master"},"39e667a14527f194a1905f10bf53eac14cf98f5d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-01-27 16:39:44.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/35/822735/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/35/822735/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/822735/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/822735/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/822735/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/35/822735/2"}}},"commit":{"parents":[{"commit":"67b531b143b0c95a974c37036d7ef457051199ea","subject":"Replace six.BytesIO, six.StringIO usage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/67b531b143b0c95a974c37036d7ef457051199ea"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2021-12-21 13:07:51.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2022-01-27 14:54:05.000000000","tz":0},"subject":"Remove six.moves.range usage","message":"Remove six.moves.range usage\n\nThis is the same as the \u0027range\u0027 keyword in Python 3\n\nChange-Id: If3aa008522c24e870b7bf13de32b8ed1b27cb519\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/39e667a14527f194a1905f10bf53eac14cf98f5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/39e667a14527f194a1905f10bf53eac14cf98f5d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
