)]}'
{"id":"openstack%2Fglance~822736","triplet_id":"openstack%2Fglance~master~I7badeeaca438b0291f4ed86670e7f217e6372c61","project":"openstack/glance","branch":"master","topic":"remove-six","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2022-02-03 09:12:25.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2022-02-02 19:57:13.000000000","reason":"Cyril Roelandt replied on the change"}},"hashtags":[],"change_id":"I7badeeaca438b0291f4ed86670e7f217e6372c61","subject":"Remove six.moves.http_client usage","status":"MERGED","created":"2021-12-22 16:14:34.000000000","updated":"2022-02-03 09:17:42.000000000","submitted":"2022-02-03 09:12:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"822736-remove-six","meta_rev_id":"914961a33c4df77523eb6e65e760f41116e4956f","_number":822736,"virtual_id_number":822736,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-03 09:12:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2022-02-03 05:00:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2022-02-02 19:57:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-02-03 09:17:42.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2022-02-02 19:57:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-22 18:04:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-23 10:44:15.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-04 07:10:20.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2022-01-04 11:30:30.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"},{"updated":"2022-01-26 20:11:54.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"df472e96c5465607bbb84b6d67da08f305c455f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-12-22 16:14:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"45bb2afb7b7ffc8e03a38461500db4117c93b036","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 18:04:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/1be087860c7140009fedcfb035cd7cd6 : SUCCESS in 56m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/16dd60cf77004020adbf9d2e75ed983e : SUCCESS in 1h 39m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/80f4df2aa50d44599c40543cfca976e8 : SUCCESS in 50m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/378c96f6dede4ba4853a68b413bf8512 : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6e6eebae863a42779c600d315e0e1b76 : SUCCESS in 6m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fe34f0c63fe84a6ea978b7084a2f2fde : SUCCESS in 8m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/729f2060bda0487d934041627c0a0fd7 : SUCCESS in 6m 25s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c372c3c2ef9f4ca29a215aa3ebb24836 : FAILURE in 9m 30s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a9383657b4974abbbc4b48c4d899f6c8 : FAILURE in 11m 03s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/c889a421723c40438408f66e0b9f3927 : TIMED_OUT in 32m 15s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/7dc742eb5ac3411ea778e29398b6232c : SUCCESS in 4m 37s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/ff61a1ea4f104c90a8192afaee058103 : SUCCESS in 1h 23m 59s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c183fb0dddf7456f86645925825cd9f0 : SUCCESS in 1h 10m 01s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/916cc0e70e8344debc710a0cb92140fd : SUCCESS in 1h 09m 34s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/0d05b0c994ae43e281b903be1e3072a2 : SUCCESS in 1h 19m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c247c0897b08408e8bee2c0ec5c1b45c : SUCCESS in 53m 37s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/655c88a4757f406cb0225c79a5b055fc : SUCCESS in 1h 22m 00s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/05b2934477b343dc821a7f6764203b5a : SUCCESS in 39m 41s","accounts_in_message":[],"_revision_number":1},{"id":"a270e86b9fa7d52e00470fc0d944235506db0551","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-12-23 10:42:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ff07beb8355b9f5b8776440431571b38fba36940","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-12-23 10:44:15.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/822736/822736/2/check-not-vote/ext-glance-zuul/56aaad8 : FAILURE in 27s","accounts_in_message":[],"_revision_number":2},{"id":"f6530f409d66e7dfd6f7f2ac0e2ad3dce3761320","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-23 12:35:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/399ddbb2b6fc407eb211721d7b20065e : SUCCESS in 1h 07m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8a116617f2434d6a8042f8cd51d53587 : SUCCESS in 1h 14m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6209ef08825f470e9ed2b2bdbcf68308 : SUCCESS in 52m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e143bf1c6bc442d86a5927df9ebcb61 : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fad569b70f57413d95737a386cd0457c : SUCCESS in 5m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d608aa94ce2042b58afd0a561cc4425d : SUCCESS in 6m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe23865b96674c389fb049616a85ee79 : SUCCESS in 6m 31s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5aa2c22fa15a4868a38d5ed8fd43e92c : SUCCESS in 9m 05s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8287ac62443f4794956dc500acd7e4e3 : SUCCESS in 10m 12s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/dac3c86d302f452b92e59507d83ea8fc : SUCCESS in 9m 54s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/005b8cdb4563454b90f6f76623b702c2 : SUCCESS in 5m 19s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4905f2db057f499fbbcc5f6951430538 : SUCCESS in 1h 13m 45s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0216c62f995443df82e6ce3b536d24fe : SUCCESS in 1h 22m 58s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b7e43a04698e4eec88a1cbccb5538844 : SUCCESS in 1h 23m 31s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/38a347b39e104ab690e9499def9702d9 : SUCCESS in 1h 41m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/680473c48f624b2496da617d5d07ac67 : SUCCESS in 58m 27s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/cf98fa5d2e0747269d78266c2eb8c554 : SUCCESS in 1h 21m 40s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f71bf3d71edc44bf966a8ea96c772b37 : SUCCESS in 38m 15s","accounts_in_message":[],"_revision_number":2},{"id":"292adb28f2782a886c9535b53ef3b56f880da972","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-01-04 07:10:20.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"29e9438c5926a31537b92b93d70a00ae9f90276f","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2022-01-04 11:30:30.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"73d0a2a6e17ad78095ea764d9606a2a9ed8da938","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-01-26 20:11:54.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"899aded38746b8ed401c453364eebd3d705fafbf","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-01-27 11:15:38.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"c639f18922c87a6c7ac2fdd86d4c76db4012ff5f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-01-27 16:39:44.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"ab4a5f92340a0745f719ff05eef968730b077b24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 19:09:59.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/fb38af7b101e4998a95a71fe239a879a : SUCCESS in 58m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/319859cf7489493c9d766237e8ee2d9a : SUCCESS in 1h 11m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1559447d184e4d58903e21b56a180a58 : SUCCESS in 41m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a049eb2a9ca4b50a441f047b21b7a3e : SUCCESS in 6m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f8749610d57a456e8fcbb9c7c02a6b62 : SUCCESS in 6m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9d80f5e60b7245ca919c6075e0307a92 : SUCCESS in 6m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0373a34ae5046829e2ac30c78b291ed : SUCCESS in 7m 56s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a6beffac2d044c79bcc87c96f40ea6a0 : SUCCESS in 11m 50s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4cc60074d010430da9cc1344f1a52a7f : SUCCESS in 12m 24s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/831a71c9cfa84cc4a9c27339fdaf39dd : SUCCESS in 13m 02s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/a347023affca4af091e7ce42f07775a5 : SUCCESS in 7m 32s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/dce33065aea74f6eaed84d5d8fb2d510 : SUCCESS in 1h 17m 47s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/28b129aa726b47ebb0670da6b7ab1942 : SUCCESS in 1h 27m 56s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f35bd3673ece43e0b4d57db3ae0b6096 : SUCCESS in 1h 25m 52s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f5c824a0820440fa87e73a9e6e7c8841 : FAILURE in 1h 15m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7b9940d838fb4fbdb3c3e8c30b2e10d0 : SUCCESS in 56m 31s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/cfedcdefcc294c3a8bfeef07ce3509aa : SUCCESS in 1h 06m 21s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/34d486c6c245458d8b67798e90661ad8 : SUCCESS in 36m 05s","accounts_in_message":[],"_revision_number":3},{"id":"774c94266fb871fc8c177417431990975061c9f1","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-01-28 10:32:28.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"aeb9ab8989fc70a38406691e53bae9d24679b96d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 11:58:01.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/537fb7bb372543eca64100925928d9c8 : SUCCESS in 49m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0f062283ce644747b2c389784cbab772 : SUCCESS in 1h 09m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/485aaccd056c4af8af7866e95c1ab662 : SUCCESS in 49m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e4fe3bb89bc402b8b17e526196c3ea8 : SUCCESS in 6m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7b8a66af19244ef1aa650fb9ed061823 : SUCCESS in 5m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3d54df90fea14bdebc6fc712a8eba001 : SUCCESS in 7m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a06ea625bac4c4aa0ce9bc1abe3b3b6 : SUCCESS in 6m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/204a303d34074ae2991b009df2240fd2 : SUCCESS in 9m 42s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c6d72222c2b4475fb9cc8e758137363a : SUCCESS in 11m 22s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/0e27fa6804814219b2abba8fb0c129df : SUCCESS in 8m 46s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/9799e486dfb74c9da62142ceb1dbf753 : SUCCESS in 4m 46s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/b8ba7c68ddf34704b7dd6293e9fe90ea : SUCCESS in 1h 13m 18s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/be84333918a449118b3f99e91a8b1873 : SUCCESS in 1h 11m 24s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/53bfd01f91c449e6848f4c472c9dab4e : SUCCESS in 1h 18m 00s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c9711561a1734ea0ad2b22689e303f68 : SUCCESS in 1h 19m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/78c6fc8e813b4af79c51440c9aeb07df : SUCCESS in 49m 09s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/aad588de3b4e4a2783abac6976cffba3 : SUCCESS in 1h 09m 50s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/88be590317fd4bc09ec850f9285807e5 : SUCCESS in 34m 36s","accounts_in_message":[],"_revision_number":3},{"id":"acd59b3e39754d67416d5e00d730d4d1c11bbd16","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-02-02 19:57:13.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"db6ca96adf168fb7da5716b18651a00f10e73800","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-02 19:58:54.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c7ed576ce156af85eae79d72257f512be3ef0a84","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-02 22:08:22.000000000","message":"Patch Set 3: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"dfa528484137b89731f0943c9b9cfd0af884fd16","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-02-03 05:00:20.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"155310ef1b7abc421f9c296f05711592388373b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 06:22:16.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/ba664a1159da4ef399e210779bc2b404 : SUCCESS in 51m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/31f838a15fae40e4870b4b2292321738 : SUCCESS in 1h 13m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7a8af88dbe8546bca876f1fe2420a473 : SUCCESS in 42m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38f11ad1a63e466e83c275aa80e3453e : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/11bb49a59ca14232bf9db659afeed7c0 : SUCCESS in 7m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/81386f9046044a5d9e2163a291f84c3c : SUCCESS in 7m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2020075e24f24f0ba93ddd5224a413e9 : SUCCESS in 6m 58s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fbc85d4bd11e4d73a35fe64c4fe29cfd : FAILURE in 10m 36s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ae95640eb44842bcad43e827bddb6c1a : SUCCESS in 12m 37s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/91a49f7929294acfbdce7cbcd940d664 : SUCCESS in 10m 23s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/8d0718153bf846d69b5120710c3722f1 : SUCCESS in 5m 40s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/e91a100b9391481abcbcad5c0e08cbe4 : SUCCESS in 1h 14m 53s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6a90ce7e6f8d44b3aa5376007a3782fe : SUCCESS in 1h 06m 30s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/88de63a56bce408ba289e8b93e68a55c : SUCCESS in 1h 16m 48s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/fb5986fc0dff4180869008945c218ec9 : FAILURE in 1h 12m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/81b9137c37374ac2a4973053bc9edb74 : SUCCESS in 56m 00s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4b2512f6bffc49dd8554254ea63a56f0 : SUCCESS in 1h 15m 29s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8acf3f744d0c42fd8425a8ef8c88b000 : SUCCESS in 35m 52s","accounts_in_message":[],"_revision_number":3},{"id":"77e71365d40c8720b8ddf4390263db163dda9197","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-02-03 06:30:57.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"35ba11127f31b644ab394f2c4ae80558fc752845","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 07:56:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/2e6b7e843f824c2d831ec01752dc1e32 : SUCCESS in 50m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/81b20351c3c2453a8ef9d23e8795ad86 : SUCCESS in 1h 06m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6d3cf90a4c784a90a0059fd6bb1147ad : SUCCESS in 42m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae391be8fdfa44fbb1ebacd5cbd5decf : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/31e9a6c50070468398714b77efea6843 : SUCCESS in 5m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5a04bf02ed5047d18d7697fca67643c4 : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67a88291472b4093a6c12be41eec94ee : SUCCESS in 6m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9b576ebeea144d5daa93261d14ee8fd3 : SUCCESS in 8m 02s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a7405cd3585846c1850a575d16f9115e : SUCCESS in 8m 39s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/a746edeee56840aabff13e1970f7095f : SUCCESS in 9m 12s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/6c331a84c69b423bbbe6eba59551498a : SUCCESS in 5m 16s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/a6162a0f47274524a5c1a788f44ced98 : SUCCESS in 1h 23m 29s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/aa025e2767944e6a93946126412a1ee6 : SUCCESS in 1h 09m 23s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/17d52cb3416c4130a51a90d139c68746 : SUCCESS in 1h 04m 39s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/5c05a4d4e5a44be1a3894bcf8b6bb558 : SUCCESS in 1h 18m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/16b2db43d9ab4fbd8337222cdfa2a38c : SUCCESS in 51m 31s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b37c9e3d0cae4ab381ce7b5e0cd935de : SUCCESS in 1h 09m 54s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0cd14045ad6f40479c93691ea4923e39 : SUCCESS in 26m 38s","accounts_in_message":[],"_revision_number":3},{"id":"8135185d55284808bc9971ac0cf0fbf59b86b5a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 07:57:24.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"668c6e409cbaa3d6476328b01a02676b7c3ed979","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 09:12:25.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/499941ae1e474fc2b28a308cdcf3b749 : SUCCESS in 1h 03m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d4ed6ea72b5243c4bf115f35fdf3d4c8 : SUCCESS in 1h 00m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d360216cf53147e28620595d54fa2de8 : SUCCESS in 40m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a6ba1c7bfb342a4a19f1574446e4f49 : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2404b55b86074a259fde4f5bd6210d69 : SUCCESS in 5m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5e5a838538404025853055dead244797 : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4292da8b0c541e0bf53272da5749a07 : SUCCESS in 6m 08s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4fee482c1c604c3eb7564052ac2c319d : SUCCESS in 9m 09s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e0466ba838ed471b8be2cdcae07d8ea9 : SUCCESS in 9m 18s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/09547c5e57f64b64b654a7ada88d40a4 : SUCCESS in 1h 04m 36s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/275d80b482a2472e98310a50725e7f02 : SUCCESS in 1h 05m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/97910113ee5645eba63999a79bd8b19b : SUCCESS in 39m 33s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/22343a4a18ae4bcfb685a8c7ce3feb62 : SUCCESS in 1h 13m 41s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/9fceb075f1ef4753bd257e432155829e : SUCCESS in 39m 09s","accounts_in_message":[],"_revision_number":3},{"id":"a4818370b577093c27a66f6b7240ab4e95f824c3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 09:12:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"a7f30332577453f533f1a800515a8d43107d8b29","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 09:17:42.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4983df54e3d4087b901acf0d2bd172c : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"9679ffc46349134082f31b2d67214a567f7e8f32","revisions":{"60f30eaba15950dc7700ba1541120b06fc1cdb6f":{"kind":"REWORK","_number":1,"created":"2021-12-22 16:14:34.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/36/822736/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/36/822736/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/822736/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/822736/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/822736/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/36/822736/1"}}},"commit":{"parents":[{"commit":"a22e202bdae745774fecc31ef84e783311d473f6","subject":"Remove six.moves.range usage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a22e202bdae745774fecc31ef84e783311d473f6"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2021-12-21 13:09:22.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2021-12-21 13:13:05.000000000","tz":0},"subject":"Remove six.moves.http_client usage","message":"Remove six.moves.http_client usage\n\nThis is a rather beefy change due to the number of usages of this\nimport. The changes are trivial though.\n\nChange-Id: I7badeeaca438b0291f4ed86670e7f217e6372c61\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/60f30eaba15950dc7700ba1541120b06fc1cdb6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/60f30eaba15950dc7700ba1541120b06fc1cdb6f"}]},"branch":"refs/heads/master"},"ff2a90a24fac13885afab49b96751b0300f826f3":{"kind":"REWORK","_number":2,"created":"2021-12-23 10:42:16.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/36/822736/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/36/822736/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/822736/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/822736/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/822736/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/36/822736/2"}}},"commit":{"parents":[{"commit":"a22e202bdae745774fecc31ef84e783311d473f6","subject":"Remove six.moves.range usage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a22e202bdae745774fecc31ef84e783311d473f6"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2021-12-21 13:09:22.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2021-12-23 10:41:24.000000000","tz":0},"subject":"Remove six.moves.http_client usage","message":"Remove six.moves.http_client usage\n\nThis is a rather beefy change due to the number of usages of this\nimport. The changes are trivial though.\n\nChange-Id: I7badeeaca438b0291f4ed86670e7f217e6372c61\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ff2a90a24fac13885afab49b96751b0300f826f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ff2a90a24fac13885afab49b96751b0300f826f3"}]},"branch":"refs/heads/master"},"9679ffc46349134082f31b2d67214a567f7e8f32":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-01-27 16:39:44.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/36/822736/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/36/822736/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/822736/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/822736/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/822736/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/36/822736/3"}}},"commit":{"parents":[{"commit":"39e667a14527f194a1905f10bf53eac14cf98f5d","subject":"Remove six.moves.range usage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/39e667a14527f194a1905f10bf53eac14cf98f5d"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2021-12-21 13:09:22.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2022-01-27 14:54:05.000000000","tz":0},"subject":"Remove six.moves.http_client usage","message":"Remove six.moves.http_client usage\n\nThis is a rather beefy change due to the number of usages of this\nimport. The changes are trivial though.\n\nChange-Id: I7badeeaca438b0291f4ed86670e7f217e6372c61\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9679ffc46349134082f31b2d67214a567f7e8f32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9679ffc46349134082f31b2d67214a567f7e8f32"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
