)]}'
{"id":"openstack%2Fglance~822737","triplet_id":"openstack%2Fglance~master~I71be65ef7b8f710e9317419d7b38559b39f461b0","project":"openstack/glance","branch":"master","topic":"remove-six","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2022-02-07 04:49:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I71be65ef7b8f710e9317419d7b38559b39f461b0","subject":"Remove six.moves.urllib usage","status":"MERGED","created":"2021-12-22 16:14:34.000000000","updated":"2022-02-07 04:50:53.000000000","submitted":"2022-02-07 04:49:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"822737-remove-six","meta_rev_id":"7a42c38c6e2e0ddff95c6f6d890481b9e3148284","_number":822737,"virtual_id_number":822737,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-07 04:49:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2022-02-07 04:50:53.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2022-02-07 03:20:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-02-07 04:50:53.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2022-02-07 03:20:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-22 18:02:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-04 07:10:27.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2022-01-04 11:30:16.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"},{"updated":"2022-01-26 20:14:00.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"c1c7c32c8287d88c29e353eadbcd763fa79a2371","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-12-22 16:14:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cd0e271d43f4e112686658d92953cefad66aebd3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 18:02:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/2d9f13b85f434dd881bd203d694d59cb : SUCCESS in 1h 00m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b80f769f514040139ebf27430db4bb10 : SUCCESS in 1h 12m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/357769e9510b41b185cab63e93d3eb88 : SUCCESS in 51m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2267315ce8ee462f97f124521a1a9897 : SUCCESS in 8m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a09cd704b9c64864b0f2cf34846b515d : SUCCESS in 9m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b77e0bf241034188bd264da283fba436 : SUCCESS in 6m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae494efd24e34f238ff6f5fffdd74dd6 : SUCCESS in 8m 53s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/646ae1bb35114ccb908d74caffd3a90e : FAILURE in 10m 02s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7ac4a6bcaaa440d682926d666e7cd366 : FAILURE in 11m 17s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/3a3253b7353543439568f6fa2292dc10 : FAILURE in 10m 01s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/d53b687dcf974edfa812d599e62457b9 : SUCCESS in 6m 09s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/ccd0c3f8b32e425e955b87deac68bbf1 : SUCCESS in 1h 33m 39s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3f65a73c244b4de195b412d3e047d017 : SUCCESS in 1h 11m 44s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b78ff22e9aca4e63a9b598cc8528af37 : SUCCESS in 1h 15m 11s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/afeb000a6df347319aa2416b033dc618 : SUCCESS in 1h 24m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7db168d6e0dc4faa81a8139f0d15e007 : SUCCESS in 1h 11m 05s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e3ba7cfd40304c8393025677894ba372 : SUCCESS in 1h 15m 38s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b34d4eab940342a9b9dbc1b637d8a020 : SUCCESS in 39m 41s","accounts_in_message":[],"_revision_number":1},{"id":"339dbe92a7b1f0c044cefd75588b5a0be81917b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-12-23 10:42:16.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"f7debfdbbfacdc0ae0d867c7a53e3e8e95c6f14e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-23 12:11:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/005b38b97da449d48d36adc97bf4e4ff : SUCCESS in 52m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6cca5f56a9544702beee7db890d71010 : SUCCESS in 1h 09m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/34673bb347f04b568457767746714ef1 : SUCCESS in 37m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c79373aa93cf4b6587ca14436ccb73e6 : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1c49bef97bf34af1b7090d8bc4c5d544 : SUCCESS in 6m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/59bb818dfb6c44169f196a6375c2ded4 : SUCCESS in 6m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03822790f3ef43cf964d4e04bb222e22 : SUCCESS in 7m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d44f3346f90a44bdb1c1670b5dc9aac7 : SUCCESS in 9m 39s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/aeadd80282894a219c778677d145a420 : SUCCESS in 9m 02s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/1616811477724994b7263680bd856bf7 : TIMED_OUT in 31m 36s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/ed1c791aae62484cbe367d31cb9b7ee8 : SUCCESS in 5m 32s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c420953e007943a6be4d6ecdd8448796 : SUCCESS in 1h 11m 29s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f3c598a5be884efeb6e52bd35adeb0b1 : SUCCESS in 1h 04m 33s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/c51c27f601d940ed9570969eed3c94d6 : SUCCESS in 1h 05m 39s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/bce5fea699b94490b04ea8ac39a0887e : SUCCESS in 1h 21m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/107ddf2a5d7e49faae70ce7da49f3620 : SUCCESS in 1h 04m 57s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b2666ad654b6438aa9264bf285952b44 : SUCCESS in 1h 22m 56s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6cb0540d49f94ce29b76f3d964638d1c : SUCCESS in 36m 31s","accounts_in_message":[],"_revision_number":2},{"id":"38626a7fefd9c901be5669ed2e82a0faddfddacc","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-01-04 07:10:27.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"abac83425c520e216b47f4537779f03e9b22b6fb","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2022-01-04 11:30:16.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"59dd16390533388e11638be83f7ee6a752c616ce","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-01-26 20:14:00.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5c5a86b5489dd01f77a35393550a70b2298fd1ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-01-27 16:39:44.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"11e27f9a1d34773412b26575bbb09e921bff5ffc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 19:11:52.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/ea516c1a89694f5a9cb05bc2dcf70235 : SUCCESS in 58m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2b138dfa1de64cc69fb777ab9851ea2f : SUCCESS in 1h 16m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/27c1545b067b4e049ac26f3ee9fae60f : SUCCESS in 46m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33a03d4b20d9497db5a1163fe617c965 : SUCCESS in 8m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8b10a4592f1f44fd8f2104c162396970 : SUCCESS in 9m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/95c0337c0f92467383c31ed73917a6a6 : SUCCESS in 9m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/248356d934b34e2f93a4e26747de7c46 : SUCCESS in 7m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a891157bb1874da7ae70d42ae4302ac3 : SUCCESS in 11m 11s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ce338c2f6ece4733bd7e1ff10b23bfd5 : SUCCESS in 11m 51s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/d24e569f830349a5bb939c5f97615d89 : SUCCESS in 11m 45s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/9368024a145a42ccbdf3be30f5749f6b : SUCCESS in 7m 04s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/768f70bd2cff413ba717dbe689ea30c4 : SUCCESS in 1h 24m 23s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7dff6c18e0b54dfbbeed4087486eb178 : SUCCESS in 1h 13m 33s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/66651dba9a4343459271b62600cfbb6c : SUCCESS in 1h 20m 08s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/9be3e986eb3742cfb46b9ad883404e88 : FAILURE in 1h 27m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/666f20148a7b45f18959e2a7d862c275 : SUCCESS in 1h 01m 32s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b267c7b3ec2e40269802841af37d344e : SUCCESS in 1h 21m 24s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/feec7a5362604e51afc6cfad656ac6a3 : SUCCESS in 44m 06s","accounts_in_message":[],"_revision_number":3},{"id":"27cb51bf1e8ccb6a827118c5846a82c289bdea2b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-01-28 10:32:23.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3e01674b1a7b983dc4d5a95f98a5e9ef2d03b3cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 12:10:22.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/b5b025db0d144af8b395de94331dc40c : SUCCESS in 55m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/74eb7bd207ca451da22f27aefdaf99ad : SUCCESS in 1h 01m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/514ba0c7e916446c8fcd154cf3bf42a9 : SUCCESS in 44m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e0bd5b57c5a4b348ea76a3d61f91c8b : SUCCESS in 7m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c2cd4b78ace64afea3c7feb81b76e8d1 : SUCCESS in 7m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0e402b9c62404597a4ee520af7721b5a : SUCCESS in 8m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37b7b95de16d44698b471e5a34bd5b65 : SUCCESS in 6m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9fab547b699041a0aba0930b18bbdf77 : SUCCESS in 11m 07s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f346410a494a4b278e9b112f4d3b528d : SUCCESS in 9m 36s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/8b1efa4d825c447682b6288b4ab1c7aa : SUCCESS in 12m 07s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/3a556adc0a754f87b2ecf5d8c2359148 : SUCCESS in 7m 48s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/504e1a7ac893487b892a6d8fa4548a6b : SUCCESS in 1h 07m 08s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/08f80f0315664276b652b9c82731719c : SUCCESS in 1h 08m 29s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/a445d32e63da42b7985e4bb6d7b983a4 : SUCCESS in 1h 20m 44s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/ba8dbd3c4cc94afc96e22d5fc76010a0 : SUCCESS in 1h 33m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/91f170aa7da74959a10d10b3b80577bb : SUCCESS in 46m 32s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1441d5e19d4c4098991f632b0b509afc : SUCCESS in 1h 20m 02s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/9d4f2065af954733b0517057737fc079 : SUCCESS in 32m 29s","accounts_in_message":[],"_revision_number":3},{"id":"ca451806812d196abf1cd4df0aa9d721d6a4c0b7","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-02-07 03:20:56.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"02f824c3ca34a1960495c14eb2e526dd9a3a684f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-07 03:21:48.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b65713bf3e18ff728c20b0acfbf96a67752b53da","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-07 04:49:15.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/db98f14391a14739b8b88879a652ef55 : SUCCESS in 1h 00m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/31d808af82d143dfbf215356f8dc7397 : SUCCESS in 1h 21m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/48d523d527bf456a9ecd2f4e4c1f5703 : SUCCESS in 1h 15m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71e56a0cfbc24e648c76190aee93f194 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/756dec2845ab4a39b24b2e0dcb8d9844 : SUCCESS in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3b47224beaec423abb80e34a2a15a6d3 : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24eaea6095e246b3a908b21a84d83b6d : SUCCESS in 6m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/17f6cb5a7ac849f99a28b9c1f8d65cc1 : SUCCESS in 8m 39s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2595ed16499d4280a772ec8079e2c102 : SUCCESS in 9m 01s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/78cbd722015b4116953d2342322c2079 : SUCCESS in 1h 07m 06s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/dc10ad38f50544b99bf6414765c8c5d0 : SUCCESS in 1h 07m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/73353e5e5ba0479c9baff7a29ecfbd4e : SUCCESS in 47m 56s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4086961420494d42a99bd667568a027a : SUCCESS in 1h 13m 22s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1d0ecc45a3444467a03714f785d44634 : SUCCESS in 30m 35s","accounts_in_message":[],"_revision_number":3},{"id":"507dc2e85a7d7acfaf5e7dedce26b2f5e90747f3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-07 04:49:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"1d468396c16241a5db5cd6bacc9649de06f05e8a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-07 04:50:53.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41c0b7d56f304df289196c976d11b8d0 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"87ba56161b6e458b5dfb47c39ec09e279c3ac4b5","revisions":{"e65d9a295f1fbfeda2de60d917ba23ddc103d18a":{"kind":"REWORK","_number":1,"created":"2021-12-22 16:14:34.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/37/822737/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/37/822737/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/37/822737/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/37/822737/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/37/822737/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/37/822737/1"}}},"commit":{"parents":[{"commit":"60f30eaba15950dc7700ba1541120b06fc1cdb6f","subject":"Remove six.moves.http_client usage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/60f30eaba15950dc7700ba1541120b06fc1cdb6f"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2021-12-21 12:17:52.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2021-12-21 13:15:57.000000000","tz":0},"subject":"Remove six.moves.urllib usage","message":"Remove six.moves.urllib usage\n\nChange-Id: I71be65ef7b8f710e9317419d7b38559b39f461b0\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e65d9a295f1fbfeda2de60d917ba23ddc103d18a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e65d9a295f1fbfeda2de60d917ba23ddc103d18a"}]},"branch":"refs/heads/master"},"48c128634ade9d4118b199486372624281ac4364":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-12-23 10:42:16.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/37/822737/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/37/822737/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/37/822737/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/37/822737/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/37/822737/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/37/822737/2"}}},"commit":{"parents":[{"commit":"ff2a90a24fac13885afab49b96751b0300f826f3","subject":"Remove six.moves.http_client usage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ff2a90a24fac13885afab49b96751b0300f826f3"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2021-12-21 12:17:52.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2021-12-23 10:41:29.000000000","tz":0},"subject":"Remove six.moves.urllib usage","message":"Remove six.moves.urllib usage\n\nChange-Id: I71be65ef7b8f710e9317419d7b38559b39f461b0\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/48c128634ade9d4118b199486372624281ac4364"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/48c128634ade9d4118b199486372624281ac4364"}]},"branch":"refs/heads/master"},"87ba56161b6e458b5dfb47c39ec09e279c3ac4b5":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-01-27 16:39:44.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/37/822737/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/37/822737/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/37/822737/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/37/822737/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/37/822737/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/37/822737/3"}}},"commit":{"parents":[{"commit":"9679ffc46349134082f31b2d67214a567f7e8f32","subject":"Remove six.moves.http_client usage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9679ffc46349134082f31b2d67214a567f7e8f32"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2021-12-21 12:17:52.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2022-01-27 14:54:05.000000000","tz":0},"subject":"Remove six.moves.urllib usage","message":"Remove six.moves.urllib usage\n\nChange-Id: I71be65ef7b8f710e9317419d7b38559b39f461b0\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/87ba56161b6e458b5dfb47c39ec09e279c3ac4b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/87ba56161b6e458b5dfb47c39ec09e279c3ac4b5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
