)]}'
{"id":"openstack%2Fglance~822738","triplet_id":"openstack%2Fglance~master~Ibca3884e1ea3d0fb170bcc9e70a176d144ee24cc","project":"openstack/glance","branch":"master","topic":"remove-six","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2022-02-14 06:46:12.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2022-02-07 15:11:16.000000000","reason":"Cyril Roelandt replied on the change"}},"hashtags":[],"change_id":"Ibca3884e1ea3d0fb170bcc9e70a176d144ee24cc","subject":"Remove remaining six.moves usage","status":"MERGED","created":"2021-12-22 16:14:34.000000000","updated":"2022-02-14 06:47:47.000000000","submitted":"2022-02-14 06:46:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"822738-remove-six","meta_rev_id":"53bf65b67ea3a0cccf50469f3524b5adb3de8f63","_number":822738,"virtual_id_number":822738,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-14 06:46:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2022-02-14 05:20:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2022-02-07 15:11:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2022-02-14 05:20:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-22 18:09:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-04 04:11:01.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2022-01-04 07:10:33.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2022-01-04 11:30:01.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"},{"updated":"2022-01-26 20:16:31.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"94985f8c7bc9a4f064a3a8a528dfc26564ac2159","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-12-22 16:14:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8be2f9ad695ad2c8f1994301a7cace93b578f1c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 18:09:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/a1c93dd37e6640bfa6ca1d4f0440b6a3 : SUCCESS in 6m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/dfab93a42b474b19bfb3d2db434c6180 : SUCCESS in 56m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/35be97b622fb4193bb6eac177d37a66b : SUCCESS in 1h 38m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/19974d21d1e44d35bcb024f15bf7dc2e : SUCCESS in 54m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fcd23176cb447e8afd38d5285cabc8d : SUCCESS in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/301871fe8b2e4bb3aa00721c2cb7da5f : SUCCESS in 6m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e8e4b9e84eb044a3a2f9397c90323db5 : SUCCESS in 7m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fb9a8f2378f4fcebd92e4d06b6f2ba9 : SUCCESS in 7m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9c1212731e6a43bd96213261f719ac74 : SUCCESS in 7m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f6bdcf26042e42ea9e7b84cf9d965144 : FAILURE in 10m 17s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a89269be2321478aacaede7b64511fcd : FAILURE in 11m 30s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/07c0cdeb52044d9b877e114207e81ac7 : TIMED_OUT in 31m 43s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/ee2132d32147415abd13f86edb3fe898 : SUCCESS in 7m 03s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/efbe5d5f4bfc413fa169dc6e3a5e727e : SUCCESS in 1h 17m 37s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/dcb1acc0a7ec42b18a96cbadcb16ee2f : SUCCESS in 1h 11m 54s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/325bbeffb34f49ad807e5cd466426e87 : SUCCESS in 1h 34m 53s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d904dad092cc485e91ad72de58bcc46b : FAILURE in 1h 15m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0e221dbc93704542b6504b7ea1ef7c23 : SUCCESS in 54m 14s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1c480be5f9c14017b0a2e20916144676 : SUCCESS in 1h 32m 11s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/26b12fe8c95041d7ba6eb25db2554469 : SUCCESS in 43m 52s","accounts_in_message":[],"_revision_number":1},{"id":"b6a338d16aed1d2487837675110c2c204a2dcec4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-12-23 10:42:16.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"2efdc25ce7a8169d0c2e99b410d1e902f1fd3c9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-23 12:00:40.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d755d89d3eef47dda64bd0aa95597c28 : SUCCESS in 5m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/bc7d7887e24f4e9fb698674210396512 : SUCCESS in 53m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e4404f3a57c845ea83099cafb698806f : SUCCESS in 1h 05m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b5e3ddd7166e45139357d95a0094c2bc : SUCCESS in 48m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4ceb67e54c746ca995ce2d16dad26f7 : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/69b14821adff4e1096aee30c1cc81092 : SUCCESS in 6m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3c295b0a9c8c47808df2a3dbb27b2528 : SUCCESS in 6m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cab1210e6dea4f65b986771d9689f207 : SUCCESS in 7m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/62981fb36fe949d8bc0517a432b52921 : SUCCESS in 5m 32s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d6eff6c2017346fea9106c1c58011714 : SUCCESS in 11m 14s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ea12fe7ac8ae4c8fa294c94045426199 : SUCCESS in 8m 58s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/03e0c04674cc41e59ba9d35ac690aa91 : SUCCESS in 8m 56s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/60c81d6c74b048229fc166921be614b0 : SUCCESS in 6m 11s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/9c2f20fe5b3a4c14a62abdb0ed75e1de : SUCCESS in 1h 10m 10s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/65fc2587f2a940c59c625b18b20ee0e9 : SUCCESS in 1h 10m 43s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/cbbdcbb2a7c340da81c8ab7031667a24 : SUCCESS in 1h 05m 00s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/003459acd98c4a73ba0f06f13ebb66db : FAILURE in 1h 06m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/37bdfd994ed14c4da4c3f07b8778a19c : SUCCESS in 47m 43s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/510c50a7f39f45ba9dbeb9ca9c0a360d : SUCCESS in 1h 11m 58s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f8b776424eff43d797e49be1c236ff27 : SUCCESS in 33m 08s","accounts_in_message":[],"_revision_number":2},{"id":"089cfad91c26084dc6987b677b63a42cc5bfa01d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-01-04 04:11:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b9922737da50138d673637564a76eeb302194dc0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-04 06:04:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/947a904c6a8f4370868a8c4899d762e6 : SUCCESS in 5m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/b42d6a8414f5468f95d692b3ce3639c4 : SUCCESS in 52m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a6f9381160ef46ca857e106319324bbb : SUCCESS in 1h 10m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f762441293094ca1b3a3964655f094b5 : SUCCESS in 43m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25237d4b42c14e4cbaba4d641e2be158 : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6430d2bc05f14623ba5d9d360a02f17f : SUCCESS in 6m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2b42a0c9b226443a996ae5312441da6a : SUCCESS in 6m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0476bc3780b84b348a72ceff34f270f4 : SUCCESS in 6m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/224efe4c69394647bd87093993cefaf1 : SUCCESS in 6m 19s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4cd9bbd2bee7450a8123874e58b9eb21 : SUCCESS in 9m 36s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/024bb693479f41afbb43848f57bca83b : SUCCESS in 10m 23s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/782b4a20d477433bbe8f1b3ded4370b3 : SUCCESS in 10m 03s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/60f86475450c470782e05f7398f75add : SUCCESS in 5m 50s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d96c44a5d3264cda9e070778f966b1dc : SUCCESS in 1h 13m 01s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/16048e32231342068866c800d790e7c9 : SUCCESS in 1h 18m 05s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/05d65acc90174a93bcc0c3fcba179904 : SUCCESS in 1h 13m 58s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/2816eef87ede496485683edc68fe3b06 : SUCCESS in 1h 47m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9d3ca1ca6b104767b2e666342b825c8c : SUCCESS in 1h 00m 26s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/80e30f78aac34829b895659f896ba4ee : SUCCESS in 1h 16m 25s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/41be91c5cee44c2fb2bf078478acfd6c : SUCCESS in 36m 11s","accounts_in_message":[],"_revision_number":2},{"id":"3200665e0bc8e31dfebf082b714d80412d312b18","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-01-04 07:10:33.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"83caf9004a1503d80f837aad0093b0bbe792e374","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2022-01-04 11:30:01.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"17f44e66cf0a428f8b74a4e3546a95a555ce4bf7","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-01-26 20:16:31.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6059c1008d17a46f2c03488b3b2ed772cf4db78b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-01-27 15:02:00.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2879b2166252a8ddd59820607cd8530ef2614280","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-01-27 16:39:44.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9d2a1fd88d03108b557f3b71b4470df490c7ac7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 19:11:53.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ce6813068dc2493783403fe1cee41a26 : SUCCESS in 6m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/3ea35f4e10e24bfa87253fd5585e6700 : SUCCESS in 1h 03m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cd9735d4afd94dceb02d4021258efa42 : SUCCESS in 1h 07m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5750922bec0d4f0b9d1ebbd42463de82 : SUCCESS in 46m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79c4475cc1344a10a1e1b4045023de45 : SUCCESS in 7m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/293e42de51014c18bb6b25d4ea994891 : SUCCESS in 8m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0d4b90ccc29040ad839ba6923aa8c838 : SUCCESS in 9m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79360600a67248c6a8b7bb2006c7e598 : SUCCESS in 9m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7c0d8fdea74c43eab12255a6354f576c : SUCCESS in 7m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/369ccb77d4ce487aae03a46f619a79be : SUCCESS in 12m 48s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6e6efbd558da41c781318c3658a2d5a3 : SUCCESS in 12m 24s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/d10fc3a1c5f74e9eb77a6aa6382f0afc : SUCCESS in 12m 03s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/e8eed68282e64298be75843c8fd64e8e : SUCCESS in 6m 19s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/1db5104ff11f48ffb392530823d84141 : SUCCESS in 1h 12m 08s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/cf466dc79e8349d88b12ab44f6aff932 : SUCCESS in 1h 16m 25s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/5ba59b93695f4072a939267df7fbbe42 : SUCCESS in 1h 20m 22s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/cef05351767348ac9d495ed53d52182b : FAILURE in 1h 29m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/33821495c2554423a9393d67f2bd4514 : SUCCESS in 1h 09m 28s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b41f1a3d03734dd399efe30e77eac881 : SUCCESS in 1h 18m 28s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0901a7bd96164bc395b9d5b18c22f7bb : SUCCESS in 32m 47s","accounts_in_message":[],"_revision_number":3},{"id":"0f45059372a32b4d197599f8bf4dc325fa3597e7","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-01-28 10:32:16.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"99c934a6f319a4e9669f0048bd95a6124fb55ad4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 12:03:44.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c6141b1899d647d3b9de5cf9a2f8fd91 : SUCCESS in 5m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/d47a4d0a83784348a1d37c61e90067db : SUCCESS in 47m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/da261343cae74610bd2cb6290a3b01c3 : SUCCESS in 1h 16m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9d3d6f3783c8427595466e68fe0de899 : SUCCESS in 51m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81e1701f05414bcaa407d54e1f3da7fe : SUCCESS in 6m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/977c86c914234a2ca43ba273b42ad632 : SUCCESS in 6m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/683c37d92fd444b89dd8766ee99b699c : SUCCESS in 7m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d623b8ea2d6341039606ccf897a541f3 : SUCCESS in 7m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fcb31cdd984643c2964c069a6f0bd4de : SUCCESS in 5m 58s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7212313902df4ff7a9a9c695dad355f0 : SUCCESS in 10m 12s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a853f624533b4703a91a2e9e9e586c3c : SUCCESS in 13m 15s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/66ee6ddd48744509a72e314d4209723e : SUCCESS in 12m 17s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/f174fb7e9c22475da69a0c0c3b847fbf : SUCCESS in 6m 46s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/91f3bc485d304774822611d86526a62f : SUCCESS in 1h 21m 39s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/43ead8c260c54bb891f55b568e04009c : SUCCESS in 1h 14m 14s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/49d82d2aded44f95bb1b72be70af1a7e : SUCCESS in 1h 12m 40s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a91e40aadb8d4ecca5e9f250e4be7d9e : SUCCESS in 1h 26m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/559fa817663c4fae8d95bf2f5dac92d6 : SUCCESS in 48m 06s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c86b70b04b2d470f9c07f7228ee7191a : SUCCESS in 1h 03m 51s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/49a00da7316148f89ba60ec3d5d3f897 : SUCCESS in 33m 29s","accounts_in_message":[],"_revision_number":3},{"id":"7dc86a152b81f1de8add893a3d4b86cf0e2d65c2","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-02-07 15:11:16.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"f48752d103754267a91064f462f8980e8df8471d","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-02-11 17:17:36.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"79dc575787306d68f3347f4bbfe90b5d7e5de5c9","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-02-11 17:21:31.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"583b21fee064a2b3dbfb91235ad97f4d7be22035","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-02-14 05:20:29.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"8e831b5214c3d3d06302ed253d7b627815bfb87b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 05:21:18.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c605e949328bf7f7d8d9b9e72dc79e09ab79e36b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 06:46:12.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/dd1c66bef2b84b129afb662c17e5bba8 : SUCCESS in 4m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/7966e27264604bc58d10b3607f2f1665 : SUCCESS in 1h 00m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2a5e9e4715664319aab92da678e4587b : SUCCESS in 1h 01m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/da0a614a7e7148c885e134409f56b1ee : SUCCESS in 51m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/706c58e09b8d4831a90dd3a53a21584d : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eb4153765ae840c083591efae6488274 : SUCCESS in 5m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9ee0e77c2e9d40ed9ac7849ce1e82764 : SUCCESS in 6m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfec3e333f574a4cbde16df35a92aa2c : SUCCESS in 5m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f2d37a1cb8ff46f39428c1e59d02d703 : SUCCESS in 5m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/84d85848d6c8412da1e1812257c98d99 : SUCCESS in 9m 44s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/cec1fe862c7e45d29f00e6f2ed301526 : SUCCESS in 8m 59s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/387fec79423b46b59c23616279970c63 : SUCCESS in 57m 31s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/9ef0c59490fa4290960512cef787850f : SUCCESS in 1h 20m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f69129f30f994943b806d82223789a5c : SUCCESS in 42m 44s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/af763040fb264ac59ff0f05421a3c0fd : SUCCESS in 1h 06m 03s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/cc7c42753ccd40428b4026cd87de2269 : SUCCESS in 30m 51s","accounts_in_message":[],"_revision_number":3},{"id":"082be8c77f66c7ee0eea8ca9741cd6bc76172e5e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 06:46:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"53bf65b67ea3a0cccf50469f3524b5adb3de8f63","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 06:47:47.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/e62d26443c1a47a7ae31808f96e8245a : SUCCESS in 57s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb550545eee94418908deaebaf9d5463 : SUCCESS in 1m 08s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9afb1d5d1ddd46afb3b83bf130f48fe0 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"931809c0379ff3bc08f9caabd40416c00c556392","revisions":{"dd55cbe2d93435315a0e455de581296a7ae292ee":{"kind":"REWORK","_number":1,"created":"2021-12-22 16:14:34.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/38/822738/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/38/822738/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/822738/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/822738/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/822738/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/38/822738/1"}}},"commit":{"parents":[{"commit":"e65d9a295f1fbfeda2de60d917ba23ddc103d18a","subject":"Remove six.moves.urllib usage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e65d9a295f1fbfeda2de60d917ba23ddc103d18a"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2021-12-21 12:22:04.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2021-12-21 13:17:58.000000000","tz":0},"subject":"Remove remaining six.moves usage","message":"Remove remaining six.moves usage\n\nChange-Id: Ibca3884e1ea3d0fb170bcc9e70a176d144ee24cc\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dd55cbe2d93435315a0e455de581296a7ae292ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dd55cbe2d93435315a0e455de581296a7ae292ee"}]},"branch":"refs/heads/master"},"cb53bf501b48639d6f52ac18ed661752a1928f29":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-12-23 10:42:16.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/38/822738/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/38/822738/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/822738/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/822738/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/822738/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/38/822738/2"}}},"commit":{"parents":[{"commit":"48c128634ade9d4118b199486372624281ac4364","subject":"Remove six.moves.urllib usage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/48c128634ade9d4118b199486372624281ac4364"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2021-12-21 12:22:04.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2021-12-23 10:41:29.000000000","tz":0},"subject":"Remove remaining six.moves usage","message":"Remove remaining six.moves usage\n\nChange-Id: Ibca3884e1ea3d0fb170bcc9e70a176d144ee24cc\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cb53bf501b48639d6f52ac18ed661752a1928f29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cb53bf501b48639d6f52ac18ed661752a1928f29"}]},"branch":"refs/heads/master"},"931809c0379ff3bc08f9caabd40416c00c556392":{"kind":"REWORK","_number":3,"created":"2022-01-27 16:39:44.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/38/822738/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/38/822738/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/822738/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/822738/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/822738/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/38/822738/3"}}},"commit":{"parents":[{"commit":"87ba56161b6e458b5dfb47c39ec09e279c3ac4b5","subject":"Remove six.moves.urllib usage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/87ba56161b6e458b5dfb47c39ec09e279c3ac4b5"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2021-12-21 12:22:04.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2022-01-27 16:37:11.000000000","tz":0},"subject":"Remove remaining six.moves usage","message":"Remove remaining six.moves usage\n\nChange-Id: Ibca3884e1ea3d0fb170bcc9e70a176d144ee24cc\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/931809c0379ff3bc08f9caabd40416c00c556392"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/931809c0379ff3bc08f9caabd40416c00c556392"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
