)]}'
{"id":"openstack%2Fglance~822740","triplet_id":"openstack%2Fglance~master~I7c24a1aa3545f3499a7a2ce30b73e2656666c764","project":"openstack/glance","branch":"master","topic":"remove-six","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2022-02-14 20:05:38.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2022-02-14 20:05:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7c24a1aa3545f3499a7a2ce30b73e2656666c764","subject":"Remove final six usage","status":"MERGED","created":"2021-12-22 16:14:34.000000000","updated":"2022-02-14 20:07:27.000000000","submitted":"2022-02-14 20:05:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"822740-remove-six","meta_rev_id":"ae76c174ff7c098fbc12e4cb61ecd185436cf04d","_number":822740,"virtual_id_number":822740,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-14 20:05:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2022-02-14 05:21:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2022-02-14 20:07:27.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":-1,"date":"2022-02-14 20:07:27.000000000","post_submit":true,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-02-14 20:07:27.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2022-02-14 05:21:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-22 18:00:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-04 07:10:46.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2022-01-04 11:29:36.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"},{"updated":"2022-01-04 14:46:04.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2022-01-26 21:13:47.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"4959c9d7cdec60a6fe4c65e758197fea0406f641","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-12-22 16:14:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c785f35de68e25fb18d99d5201363b6fe7b9d733","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 18:00:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/987fd99cda9d46559d9b024cea79297b : SUCCESS in 4m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/122a6ae2e0d940f68a46d04c86e39d72 : SUCCESS in 54m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/928ad169fbe34ea1be00c98deefd5386 : SUCCESS in 1h 10m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7a398c7390ec491a86343a839a19bc2d : SUCCESS in 1h 01m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3197b2292e641dfb38fcf48e35991a7 : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5dbc34673815473ea3f4ad7f8e7a7bc8 : SUCCESS in 6m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/feebab3f9f4f45c79772b6fbf00f468f : SUCCESS in 7m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d241ea5a43964f9991174c8c2d719370 : SUCCESS in 10m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/aed90773a149499ea616136eb86a59f3 : FAILURE in 10m 46s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/cecb538af83c416185f3425168ddfc4f : FAILURE in 9m 56s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/27b8b798552544f985d8ff1fea49aa77 : FAILURE in 9m 33s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/968897dc82fd477598ee770edf9f4f90 : SUCCESS in 6m 04s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4fec09011e92468cbee78c0cf44dc19e : SUCCESS in 1h 08m 57s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/74890c5217834b32bb27380f926a050b : SUCCESS in 1h 20m 30s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/46999f9cf9b54260aa9545ba73d5b068 : SUCCESS in 1h 26m 41s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/bbdb99e13c4d400d915e81d33d59953c : SUCCESS in 1h 26m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2085fac0dcad4f8ea903353f94ab8940 : SUCCESS in 56m 03s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0c2f903acc9e4aeab1756d0ee5908102 : SUCCESS in 1h 12m 22s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/daa7de9d7d4b4929813fa5ab898d4fc4 : SUCCESS in 37m 16s","accounts_in_message":[],"_revision_number":1},{"id":"c9f832523fe5c149aae99ebba06b3b34ec13fb8b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-12-23 10:42:16.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"6a1a5be46dc25085cdb9903f593cdd49d828496e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-23 12:22:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/8c11ae0dc279413ba5c82397e6dd436d : SUCCESS in 3m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/4bd9bbea90c7432e8655de0711a6f2bb : SUCCESS in 1h 02m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ca931f7226ea41fca4e059a9165d0235 : SUCCESS in 1h 21m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cd2c64d34e194eecb9d40dd3e253da81 : SUCCESS in 54m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ac603dcaad243548a43e26bce77e34e : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f4e0650f281d4425a14847d59d619159 : SUCCESS in 5m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/261f6f0260dc48b98d69b0f0bf43b623 : SUCCESS in 6m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fe3724eb02949409712170825af826b : SUCCESS in 7m 39s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4158c2b4969347a88258ee571cbfda7f : SUCCESS in 9m 17s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6457b572b4c84fb39b41ed47cf66fce0 : SUCCESS in 9m 20s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/f1c96e455b1a4926910d0ba0752630fa : SUCCESS in 9m 11s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/54416e1cd3424b70b16b582fafc06101 : SUCCESS in 5m 12s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/9d30537c382048bf9635455fd4016aa6 : SUCCESS in 1h 28m 01s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/2aa97b3f0e314d0784c1c609ade5a768 : SUCCESS in 1h 19m 11s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/2ab8016cae0c43a49a3bdf6bc85574c7 : SUCCESS in 1h 11m 51s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/fb0a785d5de642afae350aae42db7113 : SUCCESS in 1h 29m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/836d60d595ca418fb81a80c273e6d124 : SUCCESS in 1h 01m 50s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/304ad1251c314c35bf9fc9f46b51799f : SUCCESS in 1h 30m 12s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ce17e902d9f143ed963fe7faabe28ce5 : SUCCESS in 41m 11s","accounts_in_message":[],"_revision_number":2},{"id":"38ef756d91ec7742ee3e136820b0dd3deb750f77","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-01-04 07:10:46.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f2a46087d388f2b8b5f3a47de59364e9332e3254","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2022-01-04 11:29:36.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"0b8b474d5d33c962b5eff74c18b4fe8ec1984298","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-01-04 14:46:04.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ca5435877f4361ea28dcb45bf44fee1e9a8503c6","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-01-26 21:13:47.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"c427e1dd5829f24b89339ad983b859743761c8dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-01-27 16:39:44.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"3fda0f7a625ec352d7deacec40881ff3c809c7a1","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-01-27 17:34:49.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"404cec79c9006b1f1eeb3039b12e06a82b1d459f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-01-27 17:35:00.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"803e0b5db7170a68b851e74b18e8be7374429420","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 19:04:15.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/93b5b1cc281f4cfbb6f6112690ec4b60 : SUCCESS in 4m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/c075f2c278b04b08bb548626b61d3cfe : SUCCESS in 58m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4a256ebd55234e768dc1115b72ebd7ae : SUCCESS in 1h 13m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5ab9f1907b68455bb4bef566cfbe5975 : SUCCESS in 42m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd81214e282946b3a08db11de617fb83 : SUCCESS in 8m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4885729cc6f24e32b6da83da4b8d6e24 : SUCCESS in 7m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7e43d09ea7464154bc7057591cb65677 : SUCCESS in 7m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b124a7a790146f29bb8027947fd1a71 : SUCCESS in 9m 08s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/85467678bea0414dadc72e200d2fc4bd : SUCCESS in 13m 33s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3a4e0811e2d146baa3dc7598b815d0b6 : SUCCESS in 11m 43s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/77b7699cb4714a1b86fdcf3b2987933d : SUCCESS in 11m 46s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/72cacdcc1921435c8ffa026393c96d14 : SUCCESS in 7m 15s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/643b377b73054a7f8bb02133e3ea08a3 : SUCCESS in 1h 20m 13s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/78b9f3f0ee214908a749cd08c1f61e1d : SUCCESS in 1h 19m 50s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/8328517f95d14e10a2ae700bf778903f : SUCCESS in 1h 21m 17s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/6b1483993c86448aa090d5f603df82db : FAILURE in 1h 13m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/90e4d65a1a02487ead05209cd39d85aa : SUCCESS in 45m 46s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/516527162687420ca6b519a81818d350 : SUCCESS in 1h 18m 09s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f7cc5f5d88f64e73b3f7cd342697276a : SUCCESS in 41m 12s","accounts_in_message":[],"_revision_number":3},{"id":"8e61415d2ed41d7b7ab3fef3f8c092ed83a76b02","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-01-28 10:32:45.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"744330714f6678dd5ac482fec68084a7f5c5e8d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 12:00:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/38a9eab26ed244909802260b1c3da67f : SUCCESS in 3m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/7d2b59659b024a17a1126b1779e56d13 : SUCCESS in 54m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1366f4af8fcd455cb4a7852cc6b5e30c : SUCCESS in 1h 01m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cf595d0176b34701866c2515aa9f660a : SUCCESS in 43m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0edd61b34ed457a8edf580a57ab49f6 : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d21ed91451a4265b9da36e59aaffd9a : SUCCESS in 5m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/39628514b9d6429faf1cecadfa713ad4 : SUCCESS in 6m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f24f0444cc2b4f079e958dfb3c3ea6ff : SUCCESS in 7m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/197d73d475d0462d89ef4c473a055eea : SUCCESS in 8m 09s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b6fbe859a3764474afedd969409a11ab : SUCCESS in 8m 26s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/eafd190d45dd40bdb0d820fa70f93b89 : SUCCESS in 8m 36s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/22262d7815eb4cebb0b0da7cbe4b501f : SUCCESS in 5m 44s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4ed005437282437f920561e5c8b205b5 : SUCCESS in 1h 11m 25s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/86fb2cba64ff4938ab57c868444a0869 : SUCCESS in 1h 10m 14s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/90b75b5c3a6d4184b3944471285dfba1 : SUCCESS in 1h 06m 24s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/6c87fdb270da4430b92f2144152dfac1 : SUCCESS in 1h 07m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b6d4288d35424802aea999d672ee1224 : SUCCESS in 48m 49s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/32902455636944b6876d3ee361364abc : SUCCESS in 1h 09m 21s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e184af6210174cc6869f18465eb9ba8b : SUCCESS in 38m 30s","accounts_in_message":[],"_revision_number":3},{"id":"a9fd46cfa5be9e4983213d2555e2ec60e7f2f5d3","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-02-14 05:21:08.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"0971641cf385f3cfc828fbc4b8659c1dab656a57","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 05:22:00.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"cdfe2ffb02b583e2b63369a797468ed7fff829e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 07:34:33.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/92b769cc700043cab6346d6b2db7c44d : SUCCESS in 4m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/9c6dec934c2146e09a6a7b64b541c981 : SUCCESS in 44m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6ce034abc5f14bc3b7e5b52fd0f516e9 : TIMED_OUT in 2h 11m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/670ee2d96b564a199662067c5a847ffd : SUCCESS in 41m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75be728988394fe0989079900c1a27f3 : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d9973f867034401ea55ebdfaf0d70147 : SUCCESS in 5m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b146f949535044dc9d5acf25b62d1d76 : SUCCESS in 6m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/297f724de88e4dda80799a22cccb9a13 : SUCCESS in 5m 31s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c3f9902d7c9f410f8c1b6071a6f04ce1 : SUCCESS in 11m 05s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6bf4864c25d345b19811906b11c6fd30 : SUCCESS in 8m 12s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4c81088aef1d4e6394b2289015c0f934 : SUCCESS in 1h 05m 30s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/91b270d970b34eaa98a446f3fa6520a2 : SUCCESS in 1h 06m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/50f7f4bc72734437a057985f7b6bf1b3 : SUCCESS in 44m 36s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4e95779feea04f3681ce3ad3e42fca1c : SUCCESS in 1h 00m 35s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c4df3254aa6947a1997b63e17a08e4a2 : SUCCESS in 30m 00s","accounts_in_message":[],"_revision_number":3},{"id":"e9d8399e1d198f800e41b5598fe88a5df1339ecf","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-02-14 07:58:31.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4eda9a208ef025dd51170d345b07e7e2516f4f24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 10:10:26.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/eaf633178ec54080b67110643c28ccd4 : SUCCESS in 3m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/1976079496784fd196e55ad2cff86504 : SUCCESS in 1h 03m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cce7f43cc0e64787af214cd45911756b : SUCCESS in 1h 01m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/67e1a2920aae4a9eb280f6578464e9a5 : SUCCESS in 1h 06m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d974cd6266784344ba869b9e43f7de55 : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cc6bd6da866d497bb04dc29bf7b4bfce : SUCCESS in 5m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/62cb7d7570864740acd50050c76546df : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/375cf925bc2f42cbbe3f96412b8e56f9 : SUCCESS in 5m 51s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/37f53c04c8064e9f8dd3b9cdbf32c4ec : SUCCESS in 12m 22s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/658897d37ec9485a834576199e10c070 : SUCCESS in 8m 58s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/906573423d734d7883d46b7a9a695721 : SUCCESS in 9m 04s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/bfaf6e90af2941de9a744814919172a1 : SUCCESS in 4m 31s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/0196078d84764abd978d2434af240889 : TIMED_OUT in 2h 04m 55s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/035897d61dd54aebae46566edc5a63ec : SUCCESS in 1h 16m 08s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/3377154ccfb44ba9b149fb77ae0d99e1 : SUCCESS in 1h 14m 16s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b04f6e3ac5fd47629f667540b7b66813 : FAILURE in 1h 20m 01s\n- glance-multistore-cinder-import-fips https://zuul.opendev.org/t/openstack/build/f355bf73430a4ddab16d6b245a9b7e5e : POST_FAILURE in 18m 28s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/30b45691e0914b29874587ee28eb1559 : SUCCESS in 56m 59s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8d0a7b6e6e764c2f8f4c1c0ec5d91fcd : SUCCESS in 1h 15m 18s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/65d033f0e4e0490d94a8bf118deeceb6 : SUCCESS in 37m 42s","accounts_in_message":[],"_revision_number":3},{"id":"7d24ee8781cc67375aebd0f30a97056576e67002","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-02-14 10:21:36.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"80f328532d7c67dbd42ba3c9dd50a83b4a2fb5a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 11:57:21.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/3c0a6a00d91644558f13248d8d1a99ce : SUCCESS in 3m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/4d52739a949f405dad311f8507d7fb07 : SUCCESS in 59m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b9dde7edc3e14ecea025609044c659b0 : FAILURE in 1h 25m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e3adf090fc6b439283bb8a4b0d3537bd : SUCCESS in 47m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/374a49fab28e405e883e2923a193ae08 : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/40ce540afa6a422b81c3c398c89c0394 : SUCCESS in 5m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/384c0c5407024ac7be52a9525bb6e1e7 : SUCCESS in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33fb2720247949079d92ff5dbd255c28 : SUCCESS in 6m 13s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/01bbbc30c54e43ffa35ff3eb06ef396a : SUCCESS in 11m 33s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d354e9b0e7284bb497afee942d21e98d : SUCCESS in 8m 58s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/471ac2d50aea444eafcb6a68af0fd41a : SUCCESS in 8m 56s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/9e1be78f7c7547cbbe1895d07ed3778e : SUCCESS in 4m 41s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/845ddc09e8ea4028980708a10c577121 : SUCCESS in 1h 10m 30s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/db545669b6424640a35450be5c07fc23 : SUCCESS in 1h 12m 48s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/3eb40fcc78644e86a995176f29c884e4 : SUCCESS in 1h 12m 36s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e39f9fd3c77f48279519b65f2295503b : FAILURE in 1h 16m 22s\n- glance-multistore-cinder-import-fips https://zuul.opendev.org/t/openstack/build/635ef5f485a14a9da6d69ae1564665ab : FAILURE in 1h 30m 34s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3479f7365a6d4c969ef26bde848f34ba : SUCCESS in 51m 58s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b341e1983242428abb672e741462d97e : SUCCESS in 1h 14m 33s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b0f1327fcc10468c9e10824ea3fcae03 : SUCCESS in 35m 10s","accounts_in_message":[],"_revision_number":3},{"id":"617904ae8725bdd145e08525053145de8667a60e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-02-14 14:17:29.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b5fa1a090430e2a3468f082c1e1b7bf64c6a8843","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 16:06:34.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/2aa004ea5883465788d3fca6dbe72370 : SUCCESS in 3m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/fd3aaf02c38d4d07bdfdeb017263e95f : SUCCESS in 41m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/448fcfbfe83c4bc689ea8f5cc3d0c427 : SUCCESS in 1h 05m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3c0152e3c6ea43dc987834dd4227d77e : SUCCESS in 41m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f90b3e7085ae4e73b363fd57048c6a02 : SUCCESS in 6m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/08d8d93d3499428cbd763ec4c534616c : SUCCESS in 6m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d55a3899b1fd4ec2a22ab4b9114b2191 : SUCCESS in 6m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a17e20d142b741219b705c74798905d1 : SUCCESS in 6m 37s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/bcf89467f1e24cdfa758655f26d0ae1d : SUCCESS in 11m 17s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/13d8d356002b48adb22a3d35d2ba63bf : SUCCESS in 8m 20s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/54fde8fc71414c26993cf7842e993645 : SUCCESS in 9m 58s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/ed3b6d34242548a8be10158104492480 : SUCCESS in 5m 30s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/922ecc29281a4c41ba7111ca94d05ed7 : SUCCESS in 53m 02s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/2ec38ac91d3b443e827e24868577d8bc : SUCCESS in 59m 43s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/6dfdd84e65be4d90847e43bdf4ffd69b : SUCCESS in 1h 07m 08s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/bfdaf9647447472ab2cae799e0596187 : FAILURE in 25m 47s\n- glance-multistore-cinder-import-fips https://zuul.opendev.org/t/openstack/build/a755d821f29b41efafa6c3e240fe4618 : FAILURE in 1h 44m 10s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8c4273a7e9c7489da1333f4f0692f8b3 : SUCCESS in 37m 43s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d30ee5d46df94f949f4fb8241a08bddc : SUCCESS in 1h 06m 04s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ac560b74d545498f8b7aa17af95ece30 : SUCCESS in 34m 19s","accounts_in_message":[],"_revision_number":3},{"id":"104886b94bb8c4ccdc5bfda9852dcadc8651af24","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-02-14 16:20:16.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6efb95a24a5993ef69d77ee51d0c5030e78971a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 17:56:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/3cff8795e5484e81a45b15768758079e : SUCCESS in 3m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/0161ae5dd98e411190f9afc7ed5e8c87 : SUCCESS in 51m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2caf94328bd244bb92517c51e651e893 : SUCCESS in 1h 13m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cb8a40bce71045c598f1d171e8c271f5 : SUCCESS in 41m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d36f7ceed414660838c7f8425f50318 : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/92c560cea3ef433da45d046bbdbfee2b : SUCCESS in 6m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f0e68c33a3974d0490e99f5f2f2289f3 : SUCCESS in 6m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f193b3e3cbf84838aa77ac0d6df57c82 : SUCCESS in 6m 49s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/e6ea29e90c5d4550a08d57c0d7108a6f : SUCCESS in 11m 17s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/26e1e9c59b8449c9a8d8610be7838fe5 : SUCCESS in 10m 14s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/5e08634957e24a8f9cb5d423bd23505b : SUCCESS in 11m 36s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/3881fce1d60c40c2a2b18296e58c4f81 : SUCCESS in 5m 51s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/b7ac183dad834a369f1f0f5a3cfcc8e4 : SUCCESS in 1h 14m 15s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/dd5efe490f234a6aa9e8484695b00b90 : SUCCESS in 1h 10m 49s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/587d3af470ee44a58d830fc7387a03ad : SUCCESS in 1h 12m 30s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/eea9f8d74e7a4864855b90067c484a72 : SUCCESS in 1h 27m 21s\n- glance-multistore-cinder-import-fips https://zuul.opendev.org/t/openstack/build/06527249bd3845d68983dcbbed95d89d : FAILURE in 1h 33m 41s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9688f996616e4863ac7213c288bb5fd1 : SUCCESS in 53m 07s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/54b73871bc784e1ca8054e72de3ebdfc : SUCCESS in 1h 12m 45s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/17a4afc3efa34aa9953de55ab6ad5bf2 : SUCCESS in 26m 03s","accounts_in_message":[],"_revision_number":3},{"id":"f879584ebc84595602377196e5ff56e95c542aa8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 17:57:59.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"950c0c39bf422f0dced1dc23444e34eef372aa5c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 20:05:37.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/194f7fd066d8426fa495a429f6c261fe : SUCCESS in 3m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/3f91f3f1acce4ec38363d2e37eba6605 : SUCCESS in 1h 03m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/da2317e189394827a6e1cbb089d92ee2 : SUCCESS in 1h 12m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a4dfbc2bea4f4616945a44d7c6f8a265 : SUCCESS in 57m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3b9d37313804bed8a0b04372afbf319 : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e983229849146bb8fa79e19833b746c : SUCCESS in 5m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f03c132cf49946a0b65786cfd41ad915 : SUCCESS in 8m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3753a765fcc34f72ad44575f238c060e : SUCCESS in 6m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ac899337e7db4dd4b6ee785b4321e952 : SUCCESS in 9m 23s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4ccb1025167d43b98ca49cb795286108 : SUCCESS in 8m 28s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f0f88b39736343e2a92d28f140d590ba : SUCCESS in 1h 09m 18s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/83e4c3bc605741bca59fa2e19ad29978 : SUCCESS in 1h 05m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/db3d5f1a85d543e6906d95b072081d5a : SUCCESS in 45m 33s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e868e0ec71a94180b121a806e140ecea : SUCCESS in 1h 05m 20s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f3e9f67972384f7d91495eee0200f158 : SUCCESS in 31m 37s","accounts_in_message":[],"_revision_number":3},{"id":"8536c35e85e97760c3a0478d09c77daed3e122b6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 20:05:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"8d07382a0a1be646d5e765bce526599fd40b3561","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 20:07:27.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/228c5d4f879f4371b6b7c132247b5a0b : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"33741138d92b7806700c843de9a885ae821db663","revisions":{"b85eea5a33b44951407ffe18b9c8aa8eaca5bab7":{"kind":"REWORK","_number":1,"created":"2021-12-22 16:14:34.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/40/822740/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/40/822740/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/40/822740/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/40/822740/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/40/822740/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/40/822740/1"}}},"commit":{"parents":[{"commit":"109608fc3bcbeaad974fa3b6f2490d0c1449b33e","subject":"Remove six.assertRaisesRegex usage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/109608fc3bcbeaad974fa3b6f2490d0c1449b33e"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2021-12-21 12:36:37.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2021-12-21 13:20:49.000000000","tz":0},"subject":"Remove final six usage","message":"Remove final six usage\n\nWe also update docs since guidance has necessarily changed here.\n\nChange-Id: I7c24a1aa3545f3499a7a2ce30b73e2656666c764\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b85eea5a33b44951407ffe18b9c8aa8eaca5bab7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b85eea5a33b44951407ffe18b9c8aa8eaca5bab7"}]},"branch":"refs/heads/master"},"407b7d63c19041150bde074263cbffb698402e83":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-12-23 10:42:16.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/40/822740/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/40/822740/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/40/822740/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/40/822740/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/40/822740/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/40/822740/2"}}},"commit":{"parents":[{"commit":"9d21844e211c26f653b300814ac4a2d2323b27dc","subject":"Remove six.assertRaisesRegex usage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9d21844e211c26f653b300814ac4a2d2323b27dc"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2021-12-21 12:36:37.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2021-12-23 10:41:29.000000000","tz":0},"subject":"Remove final six usage","message":"Remove final six usage\n\nWe also update docs since guidance has necessarily changed here.\n\nChange-Id: I7c24a1aa3545f3499a7a2ce30b73e2656666c764\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/407b7d63c19041150bde074263cbffb698402e83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/407b7d63c19041150bde074263cbffb698402e83"}]},"branch":"refs/heads/master"},"33741138d92b7806700c843de9a885ae821db663":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-01-27 16:39:44.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/40/822740/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/40/822740/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/40/822740/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/40/822740/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/40/822740/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/40/822740/3"}}},"commit":{"parents":[{"commit":"f68e04f26837cbebd8593c6ca87ec9fe9bca0ea4","subject":"Remove six.assertRaisesRegex usage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f68e04f26837cbebd8593c6ca87ec9fe9bca0ea4"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2021-12-21 12:36:37.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2022-01-27 16:37:20.000000000","tz":0},"subject":"Remove final six usage","message":"Remove final six usage\n\nWe also update docs since guidance has necessarily changed here.\n\nChange-Id: I7c24a1aa3545f3499a7a2ce30b73e2656666c764\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/33741138d92b7806700c843de9a885ae821db663"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/33741138d92b7806700c843de9a885ae821db663"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
