)]}'
{"id":"openstack%2Fglance~824438","triplet_id":"openstack%2Fglance~master~I6882fd2381e6ae245fd8c61bf9f4d52df2b216f5","project":"openstack/glance","branch":"master","topic":"bp/expose-store-specific-info","attention_set":{},"removed_from_attention_set":{"33765":{"account":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"last_update":"2022-02-16 17:44:21.000000000","reason":"Change was submitted"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2022-02-16 17:44:21.000000000","reason":"Change was submitted"},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2022-02-16 17:44:21.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2022-02-16 13:47:42.000000000","reason":"Cyril Roelandt replied on the change"}},"hashtags":[],"change_id":"I6882fd2381e6ae245fd8c61bf9f4d52df2b216f5","subject":"Added a new API to expose store info","status":"MERGED","created":"2022-01-12 13:27:27.000000000","updated":"2022-02-16 17:48:26.000000000","submitted":"2022-02-16 17:44:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":26,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"824438-bp/expose-store-specific-info","meta_rev_id":"3fa7dd05aad6e11846a276090b11c789d088232f","_number":824438,"virtual_id_number":824438,"owner":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-16 17:44:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":2,"date":"2022-02-16 05:56:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2022-02-16 13:47:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2022-02-16 13:47:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":1,"date":"2022-02-16 05:56:05.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-12 14:34:44.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-12 18:43:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-13 07:58:39.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2022-02-02 15:47:36.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2022-02-14 14:08:49.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2022-02-14 14:08:49.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2022-02-15 12:34:55.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"REVIEWER"}],"messages":[{"id":"0f057a540c3387b8598a3a04a9d0f9e44da7c7b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"date":"2022-01-12 13:27:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"54395374f088f1bab3d1a18df608ab98abb5334e","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"date":"2022-01-12 13:27:56.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"da958a6de8bd2e507258d7cbef68ca09840ed61f","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-01-12 14:34:44.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/824438/824438/1/check-not-vote/ext-glance-zuul/612f744 : FAILURE in 24s","accounts_in_message":[],"_revision_number":1},{"id":"8f566c4b0ec3d978b4354c0a5e2d75853917e7ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 18:43:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/42eef27ba6e64fd69097af952510da2c : SUCCESS in 55m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/11dad71589964a2c913f8cecc5ecd762 : SUCCESS in 1h 14m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/78f4f0fe09804635bdbd29c3f2349d2a : SUCCESS in 50m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df46efc1c9834e6f814f7cccb445e4b8 : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7bb657a64a9c458c9b811ce07e97db22 : SUCCESS in 6m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a43dd96c847449f191ae68074cfaed82 : SUCCESS in 6m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3abbf5c88e14e95a469f94c4927814b : SUCCESS in 7m 10s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3642a06e822e44698fb940744bb709ad : SUCCESS in 9m 53s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0e3ca1a69327430ca4c3cdf64b0e74db : SUCCESS in 10m 07s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/e535d8e3c06e43a0afbab52819a43827 : SUCCESS in 13m 44s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/4407317aa0994c768c8a3981bbfdaec4 : SUCCESS in 5m 40s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/139c12ec97204bc58e1796254e961a9d : SUCCESS in 1h 26m 29s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/09718faf498a4c53b7626cb04f90dfa8 : SUCCESS in 1h 21m 53s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/9d0d5d18c21f4cb5b8b3665b6b25559c : FAILURE in 1h 19m 11s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a6a688dbf61c47c7ae364d6b1df21bb5 : FAILURE in 1h 24m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c092521be3fb4523a0986951a66fb325 : SUCCESS in 56m 46s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9591413657bf48dfaaf7aed3f920de2e : SUCCESS in 1h 20m 04s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a17dbadeb8b04c92bf554241946b3b8d : SUCCESS in 36m 22s","accounts_in_message":[],"_revision_number":1},{"id":"31b9e544ea10377b549b12657d950c22a11bcfa6","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-01-13 07:58:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"30ca369901ac56a68bb1d2385a55304a09e44ea6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"date":"2022-02-02 14:38:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6876c14baf784a929fb4a8e296aed2c4e2702ae1","author":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"date":"2022-02-02 14:39:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d07cc0710f9bf346d203789bf1070025d5a6b6d1","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-02-02 15:47:36.000000000","message":"Patch Set 2: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"948aac30e230df3c8fa7ad310d7988427469dd49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-02 16:14:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/f9047640625f42b0925984c1e0d1ad1a : SUCCESS in 54m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/abf735c315994fef896677d978ee20d1 : SUCCESS in 1h 21m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fb7d535a3ac641f38bd2806f53186335 : SUCCESS in 48m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6257319c10e420e8fc5549883b180f7 : SUCCESS in 7m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/53be715990f5417e933160e6ec0de87a : SUCCESS in 5m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/300b215f778c4bea8acc7379733633e3 : SUCCESS in 6m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b81c2741878c4480950a501208b25956 : SUCCESS in 8m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/78653657d5314c4aafbdd666c335a3d7 : SUCCESS in 9m 42s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/636ff6554d3d4483adc3c244935f60cf : SUCCESS in 10m 03s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/5577cf4c5aba464589b037d96b69118d : SUCCESS in 11m 12s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/9ba3572e3c5f44edb6b3c303ae2a8475 : SUCCESS in 6m 42s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/b766c0fd189f4b6e83ae543fae4559a2 : SUCCESS in 1h 08m 02s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b771d25fff914b3b9476b2334d495ed6 : SUCCESS in 1h 09m 46s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b9cd2688f87e44b0b868715c363e68da : SUCCESS in 1h 07m 43s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/fbc5bee8f6ff4ed88dbaf80b57ef8907 : SUCCESS in 1h 27m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/370baa5de2024344aced37571e80e0af : SUCCESS in 46m 22s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/40119fa3f0034849811568917636e947 : SUCCESS in 1h 10m 21s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d302308de60f43c0a15303c528abdccc : SUCCESS in 44m 16s","accounts_in_message":[],"_revision_number":2},{"id":"2a76f3d9e669840c77ee0335595f118c1fdb2b5f","author":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"date":"2022-02-03 05:34:27.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2219285db30c4913c1059b98d6c67165efb9a08b","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-02-03 09:50:13.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4caeef1421b636c3f12ef8f678a855a050bb3018","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"date":"2022-02-09 10:48:45.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6dde6ee77a7fc01b5b29322d727916b7e59ecdec","author":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"date":"2022-02-09 10:49:30.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"721f68410362d7cb5a8aabbf6263d321e7aa4f8f","author":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"date":"2022-02-09 10:49:39.000000000","message":"Patch Set 3:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"20a538773d6ce96f0d6ce1ad2eef39a20e2c8edb","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-02-09 10:49:46.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/824438/824438/3/check-not-vote/ext-glance-zuul/60741c5 : FAILURE in 27s","accounts_in_message":[],"_revision_number":3},{"id":"2fd369d2069cf0215547f430dd449134dc73118c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 13:07:19.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/0ec6785b2aa94174bf8ececd302ac8de : SUCCESS in 6m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/32f759c904444cd38d24788131d45de5 : SUCCESS in 59m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d20f48155bc44697895c8cc9eaba6c8b : SUCCESS in 1h 07m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/25c3f4b282e048b69dcb16cb36e3bee2 : SUCCESS in 50m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8c915042a104cb286fc15c4ddde7ef2 : SUCCESS in 7m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/45e1be7abf4b42728963a958b525aed6 : SUCCESS in 6m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/de021ff334e34d71a772c33e8260d428 : SUCCESS in 6m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03e6e8dc61f949fa90e3ae1e2d05fc0c : SUCCESS in 7m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7b421cfb9e3d4d3abed719b65acadbd1 : SUCCESS in 7m 14s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/06adf6c86815424a9ca112bedae69b0f : SUCCESS in 13m 44s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b01a0dbe754a4fef889af0e6747ab47c : SUCCESS in 11m 27s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/de39870a81d4478ca1227378b75f8439 : SUCCESS in 9m 15s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/2f22028d59894f4c8e63d4830b2d194e : SUCCESS in 6m 19s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d64e1297f341494aaf2d92a336569ef7 : SUCCESS in 1h 36m 25s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/fc13f8f9f58c4a7d9a5cfeacd4c1ea09 : SUCCESS in 1h 01m 03s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/3660292df03447e3abfb6cd03fee9f3d : SUCCESS in 1h 15m 57s\n- glance-multistore-cinder-import-fips https://zuul.opendev.org/t/openstack/build/fa9d3affdd094e65bbfd7bc5c6106dee : FAILURE in 2h 09m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3768d56edae84b8299de069c8be5e95e : SUCCESS in 51m 17s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0ae9d81325584410881ad553edf4f95b : FAILURE in 1h 09m 42s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/75b150842d1d42d989c43f2e61927009 : SUCCESS in 36m 13s","accounts_in_message":[],"_revision_number":3},{"id":"4dacc2bf38d60dd30bfbdb9ff30d5e49a18e0c2e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-02-09 13:53:18.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"425dd6d130fb31b8e3ca99d7a90e2c20c4c7f471","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-02-09 14:05:01.000000000","message":"Patch Set 3: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":3},{"id":"9e3848c2c9a2bd344a088c2a06eeb052b6a035b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 16:05:04.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8451b7bf718145da95bc883096fa7dd0 : SUCCESS in 7m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/6f3f4eaf91fc4097ba1f82438c1b1e90 : SUCCESS in 1h 00m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/898dda3fe786431eae4f0c28fd5f88f0 : SUCCESS in 1h 21m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/99e8d54b377f434eab65bc38aad82a28 : SUCCESS in 48m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2b0e1609e7d4fec9e4d2a8a64b5b983 : SUCCESS in 6m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc8744e5a39f47ce98b214b7c957db58 : SUCCESS in 6m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7f3fc5959345438ba6f87a852dc5f017 : SUCCESS in 6m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31947f7a0f5142bbb9fd94b322b826e9 : SUCCESS in 8m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/85705c26074640e6a6d4c7bc8dea469a : SUCCESS in 6m 47s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/663202c5dfda4893b7b61190f0061bb8 : SUCCESS in 12m 07s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7a1f4398872f4231bce2009fbebedb12 : SUCCESS in 11m 49s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/8f788562f2d74dfa9b8151109d868bf3 : SUCCESS in 12m 09s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/854998499ca7483d9972ebd2668e2a89 : SUCCESS in 5m 21s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/79b213d63b414b04a1a1b1a5cae61a4d : SUCCESS in 1h 06m 00s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/04dee274f8934a04b7a9ff46eb71e500 : SUCCESS in 1h 09m 51s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/3c47b42581b4446b9ff8554a662ed999 : SUCCESS in 1h 13m 52s\n- glance-multistore-cinder-import-fips https://zuul.opendev.org/t/openstack/build/c8c2ee9ca44343799a5bb11297b96f98 : FAILURE in 1h 53m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7b7b08ca34334ec785235cacc8dc111a : SUCCESS in 44m 22s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9570fc0089ee4c4d8bed5c9852c97b86 : SUCCESS in 1h 26m 26s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8a77aee1282645b99b9759c42bdafbf3 : SUCCESS in 36m 36s","accounts_in_message":[],"_revision_number":3},{"id":"926fe3f2f43f47dd3d302412d9e31d5442f7a177","author":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"date":"2022-02-10 06:21:01.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"aec07d9ef003f8a6b570fd25b073eaf9ee8e94ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"date":"2022-02-10 06:22:01.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"9072b223b489ddc80931216880096ce6fdf1a3fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 08:21:10.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/825427388b6241488489d18bb81c522a : SUCCESS in 4m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/0da87b54599f4e3485e7f328ac6dba5d : SUCCESS in 43m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/90e3efb6f55742a1897349f0315fcc2e : SUCCESS in 1h 00m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/27b5660c65974847b2a2a27cbb262b16 : SUCCESS in 36m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60ca4972420d47c1bb061e93a9a2ae48 : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/20651494a767408cacbf6ff65ef5be09 : SUCCESS in 5m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8de5e38cc90d4c3ab5e438690188dc1e : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cf8c9207c8141c8a3428999d527cc84 : SUCCESS in 6m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c90721c2de4f46f38a31d635fa2e27ce : SUCCESS in 5m 18s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/ea145528d6bb45558a556887200e256d : SUCCESS in 10m 08s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1a5a6383d9824ddba89c97ccb5148583 : SUCCESS in 11m 13s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/122b3dce3baf45648629461c0da14ddc : SUCCESS in 8m 54s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/c21f0cb5f7cc4f71b17b4debba15d85e : SUCCESS in 6m 37s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/a9938801a64b4d1a899497d05ef3c9f0 : SUCCESS in 1h 42m 08s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c07143b066c2499d8c3067fefb22d522 : FAILURE in 1h 16m 14s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/0d660e30213f44e8ab39aa5d0dee1087 : SUCCESS in 1h 01m 38s\n- glance-multistore-cinder-import-fips https://zuul.opendev.org/t/openstack/build/101cdd60993f40fb88c3c35ab6e9bc6f : FAILURE in 1h 54m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/adc2adf30ed2405eb259db0c43d935e0 : SUCCESS in 42m 14s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c24a955590784209a1f0795674d36f05 : SUCCESS in 1h 24m 55s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0c5d226a53b949e3909198e91bf27dee : SUCCESS in 26m 26s","accounts_in_message":[],"_revision_number":4},{"id":"45c406cb486ea47ad8a97fbef09c63f47c24ace4","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-02-11 14:05:24.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"8f7a8491595c5fb7012ce32becb0b4a56d9bb630","author":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"date":"2022-02-14 08:59:00.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7a7319b40ba09332e7b95cb0cea40a1ce6beb93f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"date":"2022-02-14 08:59:09.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"556fb8b59539bde7d9030d6732d7f8ac25dd3e74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 10:35:27.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/227f31ff805f436a854816fbdf3de148 : SUCCESS in 4m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/52ef7c2d40d343d2808adfec44babb02 : SUCCESS in 58m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/048a72cf90e3425f9b8ed04dee595060 : SUCCESS in 1h 30m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/50c58f5b22654335b7af315a222b6e55 : SUCCESS in 46m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0263a7199af4adcb091cc4ba8250fda : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ce376cecee284aafa82ec207c933c646 : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/12a94ca81c0c4afdb6273c582ad24cba : SUCCESS in 6m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5eb82ea7df064099882cce6be9293270 : SUCCESS in 7m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6e777ab058544a01bac42532e1db4199 : SUCCESS in 5m 42s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/c3a0cfac277e46c79b5085dd4208e0ed : SUCCESS in 10m 17s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/078d4843eb0d49b4b0050ac49f7afe22 : SUCCESS in 10m 17s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/c098bbe676bd4953a682241cba5d536c : SUCCESS in 9m 34s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/a343cca6c0d6404ab3af232cca5b81ba : SUCCESS in 5m 06s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/99fda463b0aa4f319b3aa4bc57ee1077 : SUCCESS in 1h 03m 50s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/07ff02cf0aac44ca9b1c331689d0237f : SUCCESS in 1h 08m 20s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/150ecfe5c0834c4281bda0c0fd455baf : SUCCESS in 1h 16m 29s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8323c0343e0e4145b782ffa7e6632b28 : SUCCESS in 1h 19m 54s\n- glance-multistore-cinder-import-fips https://zuul.opendev.org/t/openstack/build/f39b81d40ef54674a284d82abd96f3d9 : FAILURE in 1h 15m 31s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8ad7a686c00b4ed583afceb601ca4e5b : SUCCESS in 44m 02s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e78c4e70f4fd457bb92587ecf4e1e441 : SUCCESS in 1h 24m 01s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a752064251734c85bf4ce4a84c5816f5 : SUCCESS in 38m 25s","accounts_in_message":[],"_revision_number":5},{"id":"4ea7883c2ed5bfb27687e35e9daec29f05883534","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-02-14 14:08:49.000000000","message":"Patch Set 5: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":5},{"id":"96161b565e4ba8bf9cad9a0f87575b3905cef789","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2022-02-15 12:34:55.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"3cf0a478836d22960c0f58ec704f9b75c2af56b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-02-16 05:52:43.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"adbe50e5f0f70c4ef413e0939a48d335505a3ee4","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-02-16 05:56:05.000000000","message":"Patch Set 6: Code-Review+2 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"8dd155847fb1d34c38e3a5bfcf1380400cac528b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 07:28:36.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/64d1ebff095b4539994ceb99cd05d350 : SUCCESS in 6m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/255098e258f3453b8b408d138790065b : SUCCESS in 59m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0d03105c7c334ce99d390ec32cea9e43 : SUCCESS in 1h 19m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/32ef56b348cb40f8a500edc8f4e707e5 : SUCCESS in 35m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2e56e2e75be4eee8cd2de2bc91ee1ae : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/09cf6618883d446eaabd9f82e9807e4b : SUCCESS in 6m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dbe42a11285649d8a62e4cb5b928a029 : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f73db217ab041cbb33b05abb571fee1 : SUCCESS in 8m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1dc0b67e0051431a92b7380cfe4f1f8a : SUCCESS in 5m 12s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/ea31a437dbe04eff85fbf78412f0b06b : SUCCESS in 11m 22s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1fc76bc6f5664391907373a9fd39c2ca : SUCCESS in 8m 26s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/0dab56e0237542d59597c7ed80069b7c : SUCCESS in 14m 23s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/1fab2884859e4537ba88505138fe79d1 : SUCCESS in 5m 53s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/1f4681518f3a4da796d3fa1dd52922a0 : SUCCESS in 1h 09m 58s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/2146374e671c49578a5e0a9b854b885c : SUCCESS in 58m 57s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/04f0c3c316fa4f108204921341cbc4b4 : SUCCESS in 1h 12m 34s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c37dbd9bb1724596bfa199d2d42cd476 : SUCCESS in 1h 30m 17s\n- glance-multistore-cinder-import-fips https://zuul.opendev.org/t/openstack/build/74aa6441f033403fbd3afa055a03efe1 : FAILURE in 1h 32m 45s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eb59381f639447ffb2bab19da0fb56ad : SUCCESS in 40m 22s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/55bef1dc419a4af1aa5ce9d3fe96696a : SUCCESS in 1h 08m 13s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/894bed1be1124cb7aa9d49ec7c5636d6 : SUCCESS in 31m 13s","accounts_in_message":[],"_revision_number":6},{"id":"c981d2a81c6021cc9f3a61d5dc50b7aeaa81f26f","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-02-16 13:47:42.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"e1a5904f3988e6da294acbb017e9a4871d3fe7be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 13:48:36.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"ffeced2c8a717a43e17f2e390e2f960c369c8b76","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 17:44:20.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/3c61dea651e74443a454ebdcf8bbd8f9 : SUCCESS in 5m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/5916c37a335d4d85871d5aa1c6637c97 : SUCCESS in 1h 13m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0e7acd0bbdb04119b9782c380b43ee35 : SUCCESS in 1h 39m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1d851ee68d0842c18f01a7611e9cbc09 : SUCCESS in 45m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d63c615705e1439889cc35b2c8907084 : SUCCESS in 6m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0accf9e95e144ac898e0da9c6b171f6a : SUCCESS in 6m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5ddf50d4e1c24ce8a9e61062bd60fcbe : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2680daf7732f4df393d81e368013a69a : SUCCESS in 7m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/49869c79601546e0b5395cdaba44d289 : SUCCESS in 7m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/62febcff0c8c4f7bb63fede400791c41 : SUCCESS in 9m 00s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ba2c7502bafa4358957b2c0c1e7b6aff : SUCCESS in 10m 14s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e7f71c35f93a4c7d85e6913368e25186 : SUCCESS in 1h 02m 15s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f077b60b430c42c4a4d6ee1f880ef834 : SUCCESS in 1h 13m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0fcc5472575e4e518181d90c64b675c7 : SUCCESS in 44m 52s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/317572069a6a400988e748695a13b314 : SUCCESS in 1h 15m 36s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/53827da6e3794b8ab30e26c83314a5e7 : SUCCESS in 27m 10s","accounts_in_message":[],"_revision_number":6},{"id":"2d0421d8a8c1ed12781ccb696fc562303a04e83e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 17:44:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"3fa7dd05aad6e11846a276090b11c789d088232f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 17:48:26.000000000","message":"Patch Set 6:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/ddc35899da2b416c96634139dee239bf : FAILURE in 1m 18s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ae543c3072b4308b938bf09b99abd60 : FAILURE in 1m 19s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/62879c9d4391487cbf95914656d29174 : FAILURE in 1m 27s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"a34764ecac0ab0c3dc8ecd2d923e05ee2275b72b","revisions":{"60cc3346e534e2daeb9e2aea178a9c940b457440":{"kind":"REWORK","_number":1,"created":"2022-01-12 13:27:27.000000000","uploader":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"ref":"refs/changes/38/824438/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/38/824438/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/824438/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/824438/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/824438/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/38/824438/1"}}},"commit":{"parents":[{"commit":"1db6e66f93535120fdd974fce7e6490063f3c863","subject":"Merge \"Remove broken tempest-full-py3-opensuse15 job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1db6e66f93535120fdd974fce7e6490063f3c863"}]}],"author":{"name":"Mridula Joshi","email":"mrjoshi@redhat.com","date":"2022-01-12 13:01:15.000000000","tz":0},"committer":{"name":"Mridula Joshi","email":"mrjoshi@redhat.com","date":"2022-01-12 13:01:15.000000000","tz":0},"subject":"Added a new API to expose store info","message":"Added a new API to expose store info\n\nThis patch adds a new API to glance ``GET /v2/info/stores/detail``\nto expose the stores specific details about the store like store\ntype and other specific store properties.This operation\nwill be admin only and validated by the new policy rule\n``stores_info_detail`` which defaults to admin only\n\nImplements: blueprint expose-store-specific-info\nChange-Id: I6882fd2381e6ae245fd8c61bf9f4d52df2b216f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/60cc3346e534e2daeb9e2aea178a9c940b457440"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/60cc3346e534e2daeb9e2aea178a9c940b457440"}]},"branch":"refs/heads/master"},"abcdfb9f33e6d2572786703934e2e04f344ecfd8":{"kind":"REWORK","_number":2,"created":"2022-02-02 14:38:35.000000000","uploader":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"ref":"refs/changes/38/824438/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/38/824438/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/824438/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/824438/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/824438/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/38/824438/2"}}},"commit":{"parents":[{"commit":"1db6e66f93535120fdd974fce7e6490063f3c863","subject":"Merge \"Remove broken tempest-full-py3-opensuse15 job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1db6e66f93535120fdd974fce7e6490063f3c863"}]}],"author":{"name":"Mridula Joshi","email":"mrjoshi@redhat.com","date":"2022-01-12 13:01:15.000000000","tz":0},"committer":{"name":"Mridula Joshi","email":"mrjoshi@redhat.com","date":"2022-02-02 14:29:26.000000000","tz":0},"subject":"Added a new API to expose store info","message":"Added a new API to expose store info\n\nThis patch adds a new API to glance ``GET /v2/info/stores/detail``\nto expose the stores specific details about the store like store\ntype and other specific store properties.This operation\nwill be admin only and validated by the new policy rule\n``stores_info_detail`` which defaults to admin only\n\nImplements: blueprint expose-store-specific-info\nChange-Id: I6882fd2381e6ae245fd8c61bf9f4d52df2b216f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/abcdfb9f33e6d2572786703934e2e04f344ecfd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/abcdfb9f33e6d2572786703934e2e04f344ecfd8"}]},"branch":"refs/heads/master"},"bbb168767cb695b1e6407f3d331d86ff474413f5":{"kind":"REWORK","_number":3,"created":"2022-02-09 10:48:45.000000000","uploader":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"ref":"refs/changes/38/824438/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/38/824438/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/824438/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/824438/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/824438/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/38/824438/3"}}},"commit":{"parents":[{"commit":"a42fda92dc8fe160bea1dd4d22cb1135fab89022","subject":"Merge \"[APIImpact] Quota usage API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a42fda92dc8fe160bea1dd4d22cb1135fab89022"}]}],"author":{"name":"Mridula Joshi","email":"mrjoshi@redhat.com","date":"2022-01-12 13:01:15.000000000","tz":0},"committer":{"name":"Mridula Joshi","email":"mrjoshi@redhat.com","date":"2022-02-09 10:39:36.000000000","tz":0},"subject":"Added a new API to expose store info","message":"Added a new API to expose store info\n\nThis patch adds a new API to glance ``GET /v2/info/stores/detail``\nto expose the stores specific details about the store like store\ntype and other specific store properties.This operation\nwill be admin only and validated by the new policy rule\n``stores_info_detail`` which defaults to admin only\n\nImplements: blueprint expose-store-specific-info\nChange-Id: I6882fd2381e6ae245fd8c61bf9f4d52df2b216f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bbb168767cb695b1e6407f3d331d86ff474413f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bbb168767cb695b1e6407f3d331d86ff474413f5"}]},"branch":"refs/heads/master"},"a814cbc4335ab993e690c6fb8271a6525ad08ad1":{"kind":"REWORK","_number":4,"created":"2022-02-10 06:22:01.000000000","uploader":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"ref":"refs/changes/38/824438/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/38/824438/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/824438/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/824438/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/824438/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/38/824438/4"}}},"commit":{"parents":[{"commit":"a42fda92dc8fe160bea1dd4d22cb1135fab89022","subject":"Merge \"[APIImpact] Quota usage API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a42fda92dc8fe160bea1dd4d22cb1135fab89022"}]}],"author":{"name":"Mridula Joshi","email":"mrjoshi@redhat.com","date":"2022-01-12 13:01:15.000000000","tz":0},"committer":{"name":"Mridula Joshi","email":"mrjoshi@redhat.com","date":"2022-02-10 06:12:50.000000000","tz":0},"subject":"Added a new API to expose store info","message":"Added a new API to expose store info\n\nThis patch adds a new API to glance ``GET /v2/info/stores/detail``\nto expose the stores specific details about the store like store\ntype and other specific store properties.This operation\nwill be admin only and validated by the new policy rule\n``stores_info_detail`` which defaults to admin only\n\nImplements: blueprint expose-store-specific-info\nChange-Id: I6882fd2381e6ae245fd8c61bf9f4d52df2b216f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a814cbc4335ab993e690c6fb8271a6525ad08ad1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a814cbc4335ab993e690c6fb8271a6525ad08ad1"}]},"branch":"refs/heads/master"},"96c8af40a6bd859e647e827bae1678aaa9813d53":{"kind":"REWORK","_number":5,"created":"2022-02-14 08:59:09.000000000","uploader":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"ref":"refs/changes/38/824438/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/38/824438/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/824438/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/824438/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/824438/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/38/824438/5"}}},"commit":{"parents":[{"commit":"a42fda92dc8fe160bea1dd4d22cb1135fab89022","subject":"Merge \"[APIImpact] Quota usage API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a42fda92dc8fe160bea1dd4d22cb1135fab89022"}]}],"author":{"name":"Mridula Joshi","email":"mrjoshi@redhat.com","date":"2022-01-12 13:01:15.000000000","tz":0},"committer":{"name":"Mridula Joshi","email":"mrjoshi@redhat.com","date":"2022-02-14 08:49:58.000000000","tz":0},"subject":"Added a new API to expose store info","message":"Added a new API to expose store info\n\nThis patch adds a new API to glance ``GET /v2/info/stores/detail``\nto expose the stores specific details about the store like store\ntype and other specific store properties.This operation\nwill be admin only and validated by the new policy rule\n``stores_info_detail`` which defaults to admin only\n\nImplements: blueprint expose-store-specific-info\nChange-Id: I6882fd2381e6ae245fd8c61bf9f4d52df2b216f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/96c8af40a6bd859e647e827bae1678aaa9813d53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/96c8af40a6bd859e647e827bae1678aaa9813d53"}]},"branch":"refs/heads/master"},"a34764ecac0ab0c3dc8ecd2d923e05ee2275b72b":{"kind":"REWORK","_number":6,"created":"2022-02-16 05:52:43.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/38/824438/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/38/824438/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/824438/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/824438/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/824438/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/38/824438/6"}}},"commit":{"parents":[{"commit":"88c43d47151dfb4818933d4c6b98731a2cc9b529","subject":"Merge \"Cache management API endpoints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/88c43d47151dfb4818933d4c6b98731a2cc9b529"}]}],"author":{"name":"Mridula Joshi","email":"mrjoshi@redhat.com","date":"2022-01-12 13:01:15.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2022-02-16 05:46:47.000000000","tz":0},"subject":"Added a new API to expose store info","message":"Added a new API to expose store info\n\nThis patch adds a new API to glance ``GET /v2/info/stores/detail``\nto expose the stores specific details about the store like store\ntype and other specific store properties.This operation\nwill be admin only and validated by the new policy rule\n``stores_info_detail`` which defaults to admin only\n\nImplements: blueprint expose-store-specific-info\nChange-Id: I6882fd2381e6ae245fd8c61bf9f4d52df2b216f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a34764ecac0ab0c3dc8ecd2d923e05ee2275b72b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a34764ecac0ab0c3dc8ecd2d923e05ee2275b72b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
