)]}'
{"id":"openstack%2Fglance~825027","triplet_id":"openstack%2Fglance~master~Ibce11ac002dc900299d478ea875797dd1f8df17e","project":"openstack/glance","branch":"master","topic":"bug/1954321","attention_set":{},"removed_from_attention_set":{"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2022-01-27 16:14:08.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibce11ac002dc900299d478ea875797dd1f8df17e","subject":"Fix tests on Python3.10","status":"MERGED","created":"2022-01-18 00:00:02.000000000","updated":"2022-01-27 16:27:33.000000000","submitted":"2022-01-27 16:14:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"825027-bug/1954321","meta_rev_id":"27ecefc027a17236bb8ab25074e961d4d7c910fd","_number":825027,"virtual_id_number":825027,"owner":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-27 16:14:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2022-01-25 22:26:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2022-01-27 14:37:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2022-01-27 14:37:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-18 01:26:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-18 02:47:15.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-25 22:26:22.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2022-01-27 14:37:19.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"3d01951716f58a3dd378f37a441650fa37ab6fa4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-01-18 00:00:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b713f6c7a47490f22080a879c7563d9f60e916f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-18 01:26:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/d072e5eb9b5a4e1384055b15a8a0a762 : SUCCESS in 1h 05m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/abfeb00decb9425696bab80dc3a448be : SUCCESS in 51m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6aa7d6d0d34f479c9569a0959ea746be : SUCCESS in 40m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbea109ed1984fe7b3f8d85785d4c079 : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be196e5df9ad47e4902b3deca30a9be7 : SUCCESS in 5m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2af535d49bcc4340bed26fb423e770c9 : SUCCESS in 5m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2545820c2fe64b54b2089370089081a5 : SUCCESS in 5m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/053a60d15344485988d344641f60e6f9 : SUCCESS in 9m 21s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0679a024b9f74777b9f5e25c7e819b29 : SUCCESS in 8m 39s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/f59e67648c0d470cbf485c5ce55f964b : SUCCESS in 8m 55s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/615f8ae334df48ed942ff8f4befb5aa3 : SUCCESS in 5m 20s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/0e5d5f698dc2478c97f15bc20e743b4b : SUCCESS in 1h 18m 24s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4749c6b6349747c7b9ba9f16e17e18a2 : SUCCESS in 1h 12m 19s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/802ba0512d4e42db893294075942f61b : SUCCESS in 1h 17m 26s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/29558242baa04dae95bddf41db311556 : SUCCESS in 1h 22m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c328aa84de2544d3bee62929e380b8ec : SUCCESS in 40m 47s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c8074321f7634f789dc7738e33dd21c0 : SUCCESS in 1h 05m 04s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d40e256d81374126aaf3ce6ea3678921 : SUCCESS in 25m 52s","accounts_in_message":[],"_revision_number":1},{"id":"b308b980ffe688aded18430e32730a2a1a8f749a","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-01-18 02:47:15.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/825027/825027/1/check-not-vote/ext-glance-zuul/4a059ba : FAILURE in 18s","accounts_in_message":[],"_revision_number":1},{"id":"f9da6068e48071cbb19e0efa8efdf54267a0ead0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-01-25 19:20:18.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"77f721e8499da67597e97530a051c767a5b12a1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 21:01:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/a577686afc3a4f038f1b817bf3d43824 : SUCCESS in 1h 00m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/85b1db0ecf7a47abafcdce4805381d1d : SUCCESS in 1h 17m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c814bb0abdeb42e68ac804df278e194f : SUCCESS in 1h 02m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b92a3fbe1434af28adbdf3e45a5e053 : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/828b9492d6dd406e95c6a14ab64f8202 : SUCCESS in 6m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1df6c56b80114713a9a0db0e9b492c08 : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a640780aca6b4cae84cf0cd5a73d279f : SUCCESS in 6m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9ea31f8627244d44a57539984468c216 : SUCCESS in 8m 44s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/00bc869dd8074914aa4c116d4f01a04e : SUCCESS in 10m 06s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/ee4454b537e64c37b229ff671772c3ae : SUCCESS in 9m 29s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/75a70b80da254a7ea0b181a1fc873def : SUCCESS in 6m 07s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/0807c18277684002b1155dd22d869bfc : SUCCESS in 1h 13m 53s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a5de9cf251d7496d91d6783a6c5f1d7f : SUCCESS in 1h 16m 40s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/7d41cacdb5214e0294bcda651ed2dc51 : SUCCESS in 1h 20m 02s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/1c08ca5732c04ab89fc75d436323f331 : SUCCESS in 1h 38m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eaf410740b1744f4af5fba38386fe488 : SUCCESS in 50m 42s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e2b4e220b1f04ef7b184e1fe4ae58f2f : SUCCESS in 1h 14m 30s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/cfc575bcd2d34f198eea5c646ded5eb9 : SUCCESS in 28m 57s","accounts_in_message":[],"_revision_number":2},{"id":"560961740b54635ce6896de8f83ee40c2d8e5c10","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2022-01-25 22:26:22.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f3fc77209ab81c9785508b6a4b27c73f4903efdd","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-01-26 07:35:24.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/825027/825027/2/check-not-vote/ext-glance-zuul/b78507a : FAILURE in 22s","accounts_in_message":[],"_revision_number":2},{"id":"d676db01bb0c3c82a3758e19d10ffc93bf59abbf","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-01-27 14:37:19.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a263f7b6af50360ced8281dadec33a49787b7315","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 14:39:15.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"63292764b51965bd29707717b9d2c6b5571e83f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 16:14:08.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/d99b8179d263495f96f676a704ec0862 : SUCCESS in 56m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/44742ed2951c48c98f15ec8461f7ff04 : SUCCESS in 1h 10m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/19d86b5acd1d49218104cf4321bddfff : SUCCESS in 46m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/869b09dfd39345c99e0f7f9e954b934c : SUCCESS in 6m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6470a42c74184ab598a064856d910d92 : SUCCESS in 6m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/99b7f3c9d6da4d72a9e0584828e9e54a : SUCCESS in 7m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb33238e284e42a8a737475d797e495c : SUCCESS in 7m 20s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/11b35f7b15804df4b7e4e259cbed873b : SUCCESS in 11m 32s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9e47f68624aa4d2a96ac433e7e8ddefc : SUCCESS in 12m 13s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d4dd5cb5bdcb4f8fb67d863609127f99 : SUCCESS in 1h 11m 45s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/65776d4bf7b74b80aecf1ef61c99828f : SUCCESS in 1h 16m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4fb6d23bb5744b3d8b1f97b8113912b5 : SUCCESS in 48m 10s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/bfd802acc5634b0790e83afffe2dab92 : SUCCESS in 1h 19m 00s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e44370cd9e6a4c05a7aff85e15a04330 : SUCCESS in 42m 18s","accounts_in_message":[],"_revision_number":2},{"id":"0522c99fea212cea0858c92f31b88dcb49dd8283","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 16:14:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"27ecefc027a17236bb8ab25074e961d4d7c910fd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 16:27:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dec48c84b3d24f7c9f09f5e175a3c868 : SUCCESS in 1m 25s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"800ba9645dc180f8be4ac68e231cd7b7c6e42716","revisions":{"150abaaa11d2ec831aed63a65af6647832314be0":{"kind":"REWORK","_number":1,"created":"2022-01-18 00:00:02.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/27/825027/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/27/825027/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/825027/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/825027/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/825027/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/27/825027/1"}}},"commit":{"parents":[{"commit":"502fa0ffc8970c087c5924742231812c0da6a114","subject":"Remove six.iteritems usage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/502fa0ffc8970c087c5924742231812c0da6a114"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2022-01-17 23:43:08.000000000","tz":60},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2022-01-17 23:43:08.000000000","tz":60},"subject":"Fix tests on Python3.10","message":"Fix tests on Python3.10\n\nBoth test_property_protection_with_missing_operation and\ntest_property_protection_with_misspelt_operation rely on a bug in Python\n3.9: the \"]\" character in a section name is not read, so section\n\u0027^[0-9]\u0027 is read as \u0027^[0-9\u0027. This causes _compile_rule to fail and raise\nthe exception the tests are expecting.\n\nIn Python 3.10, this bug has been fixed and the exception is no longer\nraised. Instead, CONFIG.get() raises a NoOptionError in _load_rules,\nwhich is not caught.\n\nChange-Id: Ibce11ac002dc900299d478ea875797dd1f8df17e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/150abaaa11d2ec831aed63a65af6647832314be0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/150abaaa11d2ec831aed63a65af6647832314be0"}]},"branch":"refs/heads/master"},"800ba9645dc180f8be4ac68e231cd7b7c6e42716":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-01-25 19:20:18.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/27/825027/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/27/825027/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/825027/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/825027/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/825027/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/27/825027/2"}}},"commit":{"parents":[{"commit":"502fa0ffc8970c087c5924742231812c0da6a114","subject":"Remove six.iteritems usage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/502fa0ffc8970c087c5924742231812c0da6a114"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2022-01-17 23:43:08.000000000","tz":60},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2022-01-25 19:19:47.000000000","tz":60},"subject":"Fix tests on Python3.10","message":"Fix tests on Python3.10\n\nBoth test_property_protection_with_missing_operation and\ntest_property_protection_with_misspelt_operation rely on a bug in Python\n3.9: the \"]\" character in a section name is not read, so section\n\u0027^[0-9]\u0027 is read as \u0027^[0-9\u0027. This causes _compile_rule to fail and raise\nthe exception the tests are expecting.\n\nIn Python 3.10, this bug has been fixed and the exception is no longer\nraised. Instead, CONFIG.get() raises a NoOptionError in _load_rules,\nwhich is not caught.\n\nCloses-Bug: #1954321\nChange-Id: Ibce11ac002dc900299d478ea875797dd1f8df17e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/800ba9645dc180f8be4ac68e231cd7b7c6e42716"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/800ba9645dc180f8be4ac68e231cd7b7c6e42716"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
