)]}'
{"id":"openstack%2Fglance~826718","triplet_id":"openstack%2Fglance~stable%2Fxena~I8f1f38df4746cced085fb1cb4e96540763cdda13","project":"openstack/glance","branch":"stable/xena","topic":"fix-sdk-func-job","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2022-01-27 18:34:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8f1f38df4746cced085fb1cb4e96540763cdda13","subject":"[stable-only] Specify openstacksdk stable branch","status":"MERGED","created":"2022-01-27 16:03:11.000000000","updated":"2022-01-27 19:05:35.000000000","submitted":"2022-01-27 18:34:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"826718-fix-sdk-func-job","meta_rev_id":"f89a4993ac380efe650750b27180036cfc6ab85e","_number":826718,"virtual_id_number":826718,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-27 18:34:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2022-01-27 17:14:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2022-01-27 17:14:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-27 16:57:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-27 17:14:52.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"4a74dbcfa545d57c98740bf4976cb0da272858b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-01-27 16:03:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6ce44109f3791dbaa53344229bd1ae3026d61ff1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 16:57:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5a42d1edd2a54b56876172124ddc0d92 : SUCCESS in 43m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b4194ab3c4b45009c17b5126e214e7d : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/500c92e27a17418ebc170f84576554ea : SUCCESS in 6m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/42c3d29ee2364610a5e2c536cface840 : SUCCESS in 6m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/efee0f05ebef4405a1b96f45f30ab0a0 : SUCCESS in 6m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d360719c5884b149105aeb382fd74c9 : SUCCESS in 5m 50s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6cc6bce58e4d4f7caa164734948594d0 : SUCCESS in 8m 37s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/eac4c83ded844cc78b0cabefa03cc017 : SUCCESS in 8m 39s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/1336693ff7e04933ba5f9e974285d6ea : SUCCESS in 8m 01s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/620a3a36c92149ff96d6b867b699ff7a : SUCCESS in 25m 30s","accounts_in_message":[],"_revision_number":1},{"id":"db975a50311afe2b4034706fda4c2d0a7f6fb6b5","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-01-27 17:14:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c9a38e8aff1f290af1d818418682bf6096647b53","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 17:17:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6bab0d32a272a3dec222134383aa4d801f450627","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 18:34:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4bc76e6594a1470487128bac217fb4be : SUCCESS in 49m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8c34b1505704527a877a9cbc4be67b3 : SUCCESS in 7m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/82395c586d284968a1042d6303f2e616 : SUCCESS in 8m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9af71a2736cf4c59a8b2c775c49e98f7 : SUCCESS in 7m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3dd2bad5d69643cb8bd701e3e877c86a : SUCCESS in 8m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0d43c40088a646bca88bcf898d7674ff : SUCCESS in 11m 41s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/bfcdb748d5094bb4bd41ab947bfc6597 : SUCCESS in 10m 02s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ef06bc372c5046a8924a127fc4d2ffcd : SUCCESS in 45m 03s","accounts_in_message":[],"_revision_number":1},{"id":"524db35fbd8b86e0e3c7cc6860762708532725e9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 18:34:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f89a4993ac380efe650750b27180036cfc6ab85e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 19:05:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a1a1a33537a4e0b945cbe176aece019 : SUCCESS in 1m 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b76dd7880cf40f8f2cc9425c54941c5d637401a2","revisions":{"b76dd7880cf40f8f2cc9425c54941c5d637401a2":{"kind":"REWORK","_number":1,"created":"2022-01-27 16:03:11.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/18/826718/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/18/826718/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/826718/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/826718/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/826718/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/18/826718/1"}}},"commit":{"parents":[{"commit":"aa7c4acdade2071084a35e42b35c1ec5a8c9990f","subject":"Bump oslo.policy to version 3.8.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/aa7c4acdade2071084a35e42b35c1ec5a8c9990f"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-01-27 14:56:18.000000000","tz":-300},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-01-27 14:56:18.000000000","tz":-300},"subject":"[stable-only] Specify openstacksdk stable branch","message":"[stable-only] Specify openstacksdk stable branch\n\nTaking the advice given here:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2022-January/026905.html\nto fix the openstacksdk-functional-devstack job.\n\nThis job is not specified in glance\u0027s .zuul.yaml, but is brought in\nby the integrated-gate-storage template:\nhttps://opendev.org/openstack/tempest/src/commit/86db21ea6afb6c26a21fc143a7d061e947c31a93/zuul.d/integrated-gate.yaml#L382-L398\n\nSo we add the job here so that we can override its definition.\n\nChange-Id: I8f1f38df4746cced085fb1cb4e96540763cdda13\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b76dd7880cf40f8f2cc9425c54941c5d637401a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b76dd7880cf40f8f2cc9425c54941c5d637401a2"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
