)]}'
{"id":"openstack%2Fglance~826729","triplet_id":"openstack%2Fglance~master~I809d9ba93ed19a8dd7a7be59e5470c56aca871e7","project":"openstack/glance","branch":"master","topic":"remove-six","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2022-02-16 15:44:41.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2022-02-16 12:23:56.000000000","reason":"Cyril Roelandt replied on the change"}},"hashtags":[],"change_id":"I809d9ba93ed19a8dd7a7be59e5470c56aca871e7","subject":"Remove glance.common.client._encode_headers","status":"MERGED","created":"2022-01-27 16:39:44.000000000","updated":"2022-02-16 15:46:24.000000000","submitted":"2022-02-16 15:44:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"826729-remove-six","meta_rev_id":"2228853b545aca50296797e7d52d4fd3a391395e","_number":826729,"virtual_id_number":826729,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-16 15:44:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2022-02-14 05:21:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2022-02-16 12:23:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2022-02-16 12:23:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-27 19:24:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-14 05:21:33.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2022-02-14 05:21:33.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"db2a85f6590547bd0e4fc6892bb537b240f6c35c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-01-27 16:39:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9a7d46eb67b87a163f57e77cc0ef818402cd04b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 19:24:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/abbe28704623482885ca72e730d6a4e5 : SUCCESS in 1h 01m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c5d311338cca4f89b612f8a57d312412 : SUCCESS in 1h 02m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/52118796dfa7451d8d54270e6fff3a7f : SUCCESS in 52m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ccad12db94a42ef918b8db3048b8e25 : SUCCESS in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/55466459f50b46b5b7ac1f790eb2ae43 : SUCCESS in 6m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9a205065c38d4602ba08dcd95a78d35f : SUCCESS in 6m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f0e42ada3e14934a432c4c25db084a7 : SUCCESS in 8m 21s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/de122962d3904d2fa8a03e5128f61afb : SUCCESS in 10m 59s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/25e3d33e427047ccbb9dd0906ac27382 : SUCCESS in 10m 40s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/1580200a5b334cd3aa06222834fedeb4 : SUCCESS in 12m 14s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/25d5c4d2905f463990f67198b7f4afd6 : SUCCESS in 6m 03s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/9c60c242945d488fb809ab159d71e277 : SUCCESS in 1h 18m 20s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/955c093c3f394a0ea128459428b57def : SUCCESS in 1h 23m 00s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/bed72e053a564d65b45c21a5bede981b : SUCCESS in 1h 23m 12s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/9161a2b1d9ca4f71b6d7c08c2ea53de0 : SUCCESS in 1h 29m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7634b0271a764c8ea88a485f1002b193 : SUCCESS in 50m 17s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/fcc97a7f9d1248fc98bdc9c98d756cdf : FAILURE in 1h 30m 01s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3d708e92997046e083c4983a6782b000 : SUCCESS in 30m 16s","accounts_in_message":[],"_revision_number":1},{"id":"1696e16b16da193788be3cbf364a70c3a10ecf0b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-01-28 10:32:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6a908ea87fddd746ca08cf80b7c7da7fc6cf70cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 12:09:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/7bebc539525c423f8f6aea4f1d32e03e : SUCCESS in 49m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/613f6ec26c814fbfabe310da72f5a8ff : SUCCESS in 59m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6aa3ddff73f74f43a4f08d3fa46d8a37 : SUCCESS in 46m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98894704e06840c5934492b94983d0cf : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/384368fdab27456382b594ff7f132172 : SUCCESS in 5m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b35e2a4ab1d64d4c8e3e1c6f17b76a7e : SUCCESS in 7m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cce1d2c8ce744d5a2618c74e8ce96c4 : SUCCESS in 7m 19s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7dbdc8daed994c628a1897849d636c11 : SUCCESS in 9m 18s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0bdf3a7a6f8f4c8ebe6c3f139bf4e3b8 : SUCCESS in 8m 48s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/5303c495490a44d085ac7f73dbad28ee : SUCCESS in 9m 25s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/dfb1f105b0534794b25bf9b055df8289 : SUCCESS in 5m 30s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4e4293a66f57490a8bc59f0c978ec4e3 : SUCCESS in 1h 25m 11s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/83cc9a5555024c7b924a37969239e0b3 : SUCCESS in 1h 20m 43s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/7153b38195ef4338a29c51bdd65dec1a : SUCCESS in 1h 12m 18s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/140916730a1c42c98fe3aab85c3762f8 : SUCCESS in 1h 29m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3a70e15ff7fe4c06be89aeee6df4ed18 : SUCCESS in 54m 15s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/418c0122dd64407481ebbc3aca9bf98f : SUCCESS in 1h 08m 34s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/91d46fb0fc344dc1aa49ae14f61e7e0d : SUCCESS in 37m 40s","accounts_in_message":[],"_revision_number":1},{"id":"b4cd1edbf7ec265cb66282a7a456dc345fb135b4","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-02-14 05:21:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"768f5e3c501d45f904a679e6e35aa8d3efc33f98","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-02-16 12:23:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3764b9f9720f5d6d97767cb2c7a6d247ffc9e2a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 12:25:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3923ebd3e6d085edc14348b5a9f3d757092d71c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 15:44:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/30f8070bb69b49098cca2bbd4366fea1 : SUCCESS in 56m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/22aead55e21044f8938cb30eda74f0a7 : SUCCESS in 1h 04m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e4fbe3e935b549728587c047761d5fbf : SUCCESS in 52m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a705ed28209418991e6af56e2de58e0 : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1adcde84637b4473a60a6a5f2b0da858 : SUCCESS in 5m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/259abb4464a944e59fa84cb7cb5479dd : SUCCESS in 5m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24a84b20a79b418e8dbd696fd1291810 : SUCCESS in 6m 30s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4e0cd7b63e96451cb25e29c931eb7db5 : SUCCESS in 8m 46s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2a263a4bae154a5bbbb6277539debe86 : SUCCESS in 8m 40s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/2ef7da39e96d43bf8f7b16b39f9d5051 : SUCCESS in 1h 22m 00s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/d2e97410d2fc459ba8b8fde97d409b3c : SUCCESS in 1h 08m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/643b3eea5edb4d3e8cc45033f096e7e0 : SUCCESS in 51m 44s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4c012c1560dd45afaac21756a6e85213 : SUCCESS in 1h 12m 48s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c7b72a77d8614bc7955d95c89ea93529 : SUCCESS in 39m 08s","accounts_in_message":[],"_revision_number":1},{"id":"4aae08b7e40c3769750070c25d954613eb422a19","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 15:44:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2228853b545aca50296797e7d52d4fd3a391395e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 15:46:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/784c8a665cfc4052ac0342c0cf005e7a : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"87b4b33e861747154087a62df32cc440e40021a4","revisions":{"87b4b33e861747154087a62df32cc440e40021a4":{"kind":"REWORK","_number":1,"created":"2022-01-27 16:39:44.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/29/826729/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/29/826729/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/826729/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/826729/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/826729/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/29/826729/1"}}},"commit":{"parents":[{"commit":"33741138d92b7806700c843de9a885ae821db663","subject":"Remove final six usage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/33741138d92b7806700c843de9a885ae821db663"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2022-01-27 16:32:23.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2022-01-27 16:37:47.000000000","tz":0},"subject":"Remove glance.common.client._encode_headers","message":"Remove glance.common.client._encode_headers\n\nThis is no longer necessary now that everything is a unicode string.\n\nChange-Id: I809d9ba93ed19a8dd7a7be59e5470c56aca871e7\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/87b4b33e861747154087a62df32cc440e40021a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/87b4b33e861747154087a62df32cc440e40021a4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
