)]}'
{"id":"openstack%2Fglance~826730","triplet_id":"openstack%2Fglance~master~I3c06744a1de4a65b97556665040bcaa347cdb76b","project":"openstack/glance","branch":"master","topic":"remove-six","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2022-02-16 16:08:51.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3c06744a1de4a65b97556665040bcaa347cdb76b","subject":"Remove \u0027__unicode__\u0027 magic methods","status":"MERGED","created":"2022-01-27 16:39:44.000000000","updated":"2022-02-16 16:11:07.000000000","submitted":"2022-02-16 16:08:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"826730-remove-six","meta_rev_id":"9d1102af88b145052b6ce609859c12fda1ccaddf","_number":826730,"virtual_id_number":826730,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-16 16:08:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2022-02-15 14:28:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2022-02-15 14:28:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-27 19:26:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-15 14:28:17.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"534156510d30b099784d38a8c24f2f81a56e07e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-01-27 16:39:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"25552a1f270ff89deebbdf6a56f94ba37daf0663","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 19:26:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/40b8a31b8d754b2e82742119aa7dc6bf : SUCCESS in 1h 00m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a58ac23c715244e89d392e77137cd1ff : SUCCESS in 1h 16m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dc080044e7ce48bbbed038673e37d132 : SUCCESS in 48m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4bf6c1dede284d4c9b6aaecc2737a384 : SUCCESS in 6m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f4dbaa16c5ff43d6aefdea545e46812b : SUCCESS in 6m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/60e29b9d85ab43dca29a3dd3009fe117 : SUCCESS in 6m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5a610d306f140a78ee955d424f9a70e : SUCCESS in 9m 00s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c08f7e9ba510453da545ab2b6ee0231b : SUCCESS in 11m 22s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/71e002c94a374a9983f371fcad593f40 : SUCCESS in 16m 20s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/902e266e42554a4390cbbc75790ab2ff : SUCCESS in 15m 26s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/c13bba290c354955ac60617c8582ee9e : SUCCESS in 6m 48s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/186169065a5840e989caf3b562ebe4ee : SUCCESS in 1h 19m 45s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/bda047033aad4b6a938fc7edec730174 : SUCCESS in 1h 17m 11s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/a23c710d96f444ed9c1c7c2dc8faf009 : SUCCESS in 1h 23m 57s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/4b24d709f0b04f76ab31d4951bbc2bcf : SUCCESS in 1h 32m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fed895dabca24ecda01d479004e81d12 : SUCCESS in 54m 01s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1cd949cddc784e949c8b714727321cd1 : SUCCESS in 1h 10m 06s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7d44de7417484ea691dcd81bbf65b8f9 : SUCCESS in 36m 45s","accounts_in_message":[],"_revision_number":1},{"id":"4afac9f53f75b2b9e15ac0dc209ec4b2a19ae085","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-02-15 14:28:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"02b003efa200da09aa1a96a27c68a49f194bc759","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 12:24:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3f755db85d8db7980fe8b57d32bf038d923654d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 16:08:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/920cb218c3ea404bab760642c18a1c88 : SUCCESS in 1h 06m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6771e8883f3d443292addb6ab26701c2 : SUCCESS in 1h 22m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9c89324f0dea4fa485c34dcf447f4170 : SUCCESS in 50m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73ef29381d5c4529adee543c917322ba : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/05b89c69645c46b7960ab262a8367779 : SUCCESS in 5m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dc5bbc19da3a472d8c89dde5b02fe6bb : SUCCESS in 6m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25ea524a10364344976fb2ce26530a13 : SUCCESS in 6m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/745f4d531bba428bae88ee2e43983d48 : SUCCESS in 10m 59s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c308d67f94f34e27ba9441c6e39b0b51 : SUCCESS in 9m 58s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/12adbbd6fe014509b9f0924eafb1d7e3 : SUCCESS in 1h 45m 01s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/505e854baeda4ee4ba120aa3f1945696 : SUCCESS in 1h 11m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d93626681b504a7eb0c851c323af766b : SUCCESS in 53m 44s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/678212c54ab54dbba1b8b68cd9d92152 : SUCCESS in 1h 33m 25s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a8437cd285ae4137b0323ef3d25599b2 : SUCCESS in 37m 58s","accounts_in_message":[],"_revision_number":1},{"id":"34b41d36f77946cd87b961f29bc05bdaf02744c8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 16:08:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9d1102af88b145052b6ce609859c12fda1ccaddf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 16:11:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cbaf915a2964f4da25bb38638ff9cf7 : SUCCESS in 1m 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"33874f09cdd40ff6214b534cf9eee5d65c056f7a","revisions":{"33874f09cdd40ff6214b534cf9eee5d65c056f7a":{"kind":"REWORK","_number":1,"created":"2022-01-27 16:39:44.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/30/826730/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/30/826730/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/30/826730/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/30/826730/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/30/826730/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/30/826730/1"}}},"commit":{"parents":[{"commit":"87b4b33e861747154087a62df32cc440e40021a4","subject":"Remove glance.common.client._encode_headers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/87b4b33e861747154087a62df32cc440e40021a4"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2022-01-27 16:33:07.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2022-01-27 16:37:57.000000000","tz":0},"subject":"Remove \u0027__unicode__\u0027 magic methods","message":"Remove \u0027__unicode__\u0027 magic methods\n\nThese don\u0027t do anything in Python 3. We should use \u0027__str__\u0027 instead.\n\nChange-Id: I3c06744a1de4a65b97556665040bcaa347cdb76b\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/33874f09cdd40ff6214b534cf9eee5d65c056f7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/33874f09cdd40ff6214b534cf9eee5d65c056f7a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
