)]}'
{"id":"openstack%2Fglance~826743","triplet_id":"openstack%2Fglance~stable%2Fvictoria~I8f1f38df4746cced085fb1cb4e96540763cdda13","project":"openstack/glance","branch":"stable/victoria","topic":"fix-sdk-func-job","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2022-01-28 07:59:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8f1f38df4746cced085fb1cb4e96540763cdda13","subject":"[stable-only] Specify openstacksdk stable branch","status":"MERGED","created":"2022-01-27 17:08:32.000000000","updated":"2022-01-28 08:06:54.000000000","submitted":"2022-01-28 07:59:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"826743-fix-sdk-func-job","meta_rev_id":"a5567e6f049dd53bf84f35e582df9e1debb9caa7","_number":826743,"virtual_id_number":826743,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-28 07:59:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2022-01-28 05:57:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2022-01-28 05:57:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-27 20:01:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-28 05:57:51.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"2e81817e7b96c5793ca8adda46f8de920408e0d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-01-27 17:08:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"778b88d89294deefdbdc3844cd635f45ce3c0a62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 20:01:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a5d546005e6a4f76b0b6a75e54b468c8 : SUCCESS in 44m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26738d40fa3c452bbef7a70fc890d344 : SUCCESS in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/322671a9d9bd49618a9050c8d17f25ee : SUCCESS in 6m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8b3b88375d9b4924bdd780a4748bee4b : SUCCESS in 6m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edf78b25cdd04069ae01513e54bbe4ef : SUCCESS in 7m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/35440ef1b6d04123ae369d8cbf4b8764 : SUCCESS in 9m 18s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5825d5dbc6014f0cba6ee77ba1ed8c47 : SUCCESS in 9m 19s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/cfc3d977e9f444adb1bd5ea8c23a0559 : SUCCESS in 1h 41m 29s","accounts_in_message":[],"_revision_number":1},{"id":"19539f0889b3d441033fcfc4924f9157792b743f","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-01-28 05:57:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cb50a6cf408c3e9e1f74f2b12b9406632592d4be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 05:58:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"58279c8b7b5e1abea2b5fb026d675668980c1c83","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 07:59:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e6fe3f5a376a484fae34074b869cc486 : SUCCESS in 49m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ea0a120693042f5bb222bdb682bac94 : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c12031bc57294f83896ded86d9db0993 : SUCCESS in 5m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/52d5c6f5fe814f14bcf96c4c070ced98 : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ebecc6fd8974e52a18a25252af03a8b : SUCCESS in 7m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8e3a7b4fcb114e639d78ee2a90ae24ba : SUCCESS in 8m 36s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/63fdd4c7a3784e8b934a9acb0c270684 : SUCCESS in 8m 49s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/75dfe812827a4747bc2f8592445b5541 : SUCCESS in 1h 54m 38s","accounts_in_message":[],"_revision_number":1},{"id":"fa35769e8b6c2d17e6c0178f1bcef2db43b7861f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 07:59:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a5567e6f049dd53bf84f35e582df9e1debb9caa7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 08:06:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1938a0242fb473e8e8649f0b4a647d5 : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"78d63ec150b764f15a48ae9a32890e1c06abc163","revisions":{"78d63ec150b764f15a48ae9a32890e1c06abc163":{"kind":"REWORK","_number":1,"created":"2022-01-27 17:08:32.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/43/826743/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/43/826743/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/43/826743/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/43/826743/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/43/826743/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/43/826743/1"}}},"commit":{"parents":[{"commit":"31a574b4e562fb8d3c00562d63cc5f8ee274221c","subject":"Add 21.0.1 reno","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/31a574b4e562fb8d3c00562d63cc5f8ee274221c"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-01-27 14:56:18.000000000","tz":-300},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-01-27 17:07:29.000000000","tz":-300},"subject":"[stable-only] Specify openstacksdk stable branch","message":"[stable-only] Specify openstacksdk stable branch\n\nTaking the advice given here:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2022-January/026905.html\nto fix the openstacksdk-functional-devstack job.\n\nThis job is not specified in glance\u0027s .zuul.yaml, but is brought in\nby the integrated-gate-storage template:\nhttps://opendev.org/openstack/tempest/src/commit/86db21ea6afb6c26a21fc143a7d061e947c31a93/zuul.d/integrated-gate.yaml#L382-L398\n\nSo we add the job here so that we can override its definition.\n\nChange-Id: I8f1f38df4746cced085fb1cb4e96540763cdda13\n(cherry picked from commit b76dd7880cf40f8f2cc9425c54941c5d637401a2)\n(cherry picked from commit fdf7b80dafa6c1a722b8a3e3bce5be995e652ce7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/78d63ec150b764f15a48ae9a32890e1c06abc163"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/78d63ec150b764f15a48ae9a32890e1c06abc163"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
