)]}'
{"id":"openstack%2Fglance~826975","triplet_id":"openstack%2Fglance~stable%2Fwallaby~I6fcb14ad240105ed0b1b9086c6c6c95034b4dd42","project":"openstack/glance","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2022-01-31 05:27:13.000000000","reason":"Abhishek Kekane replied on the change"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2022-01-31 05:59:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6fcb14ad240105ed0b1b9086c6c6c95034b4dd42","subject":"Fix failing copy_image flow init","status":"MERGED","created":"2022-01-30 19:54:43.000000000","updated":"2022-01-31 06:01:14.000000000","submitted":"2022-01-31 05:59:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"826975","meta_rev_id":"38f04031d2fdc0aea72016d32c6d295d822cfd48","_number":826975,"virtual_id_number":826975,"owner":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-31 05:59:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2022-01-31 05:27:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2022-01-31 05:27:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-30 19:54:43.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"CC"},{"updated":"2022-01-30 19:55:12.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2022-01-30 20:20:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fa610db4b907329eaf44356986ded5cc8537e02a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-01-30 19:54:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bf3f1b3b0fdeea0c64fe17fbc3327a0a22e62dbb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-30 20:20:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b84fd5bf2f8d434aba6d79950fb17e48 : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b6753fca2414507b18a3b897906c9e6 : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/54a45656c882427a965ba70bf1e17760 : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c015b6ea2514e248793d41ff066ba95 : SUCCESS in 4m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b26c69b6205c40d38f430c519d3125e5 : SUCCESS in 6m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51beb0e126094562be589fb590b8f9d5 : SUCCESS in 5m 29s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c9481d66625841b582843644c1d362c3 : SUCCESS in 7m 59s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a3da2b023df3459db67aa7ba23bf4ac5 : SUCCESS in 7m 13s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/3c7a1161206c4733b52f588bd0f918b2 : SUCCESS in 4m 08s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/04a54dc46aaf40efaa49a5f952282100 : SUCCESS in 23m 56s","accounts_in_message":[],"_revision_number":1},{"id":"00da949c1c09a127c9c013152ea74db8ff6b07a4","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-01-31 05:27:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1abb3457effad59b28814315c7746e2d0ca91b5a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-31 05:27:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e798d70493779e119387fd383ede22a6e56797bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-31 05:59:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2e77ff7bc577459f957b3013a418839f : SUCCESS in 6m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3701cf41b5d94a84aca3e655a0806a21 : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e5083689347747a68aa2c4e60161a9c6 : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f2d89fd1db2e4703957afe97e89f7433 : SUCCESS in 8m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/930c189f5b7048b4ab1c56c7377f1d1e : SUCCESS in 6m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d21ea7755e694a5ba54daf4341aac150 : SUCCESS in 8m 09s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d29c4b697d704b29aa91fe15c50a6f42 : SUCCESS in 8m 30s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/062baa9365c44b8887c7a669f1fe7223 : SUCCESS in 30m 19s","accounts_in_message":[],"_revision_number":1},{"id":"f7bb39a18a4b1ef55f85a61348987b41d3d74ba5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-31 05:59:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"38f04031d2fdc0aea72016d32c6d295d822cfd48","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-31 06:01:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/888e7f7d057f4c86b572afc97f5c02db : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b1d635c287f950b15df207455d64d97228963e5f","revisions":{"b1d635c287f950b15df207455d64d97228963e5f":{"kind":"REWORK","_number":1,"created":"2022-01-30 19:54:43.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/75/826975/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/75/826975/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/826975/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/826975/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/826975/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/75/826975/1"}}},"commit":{"parents":[{"commit":"fdf7b80dafa6c1a722b8a3e3bce5be995e652ce7","subject":"[stable-only] Specify openstacksdk stable branch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fdf7b80dafa6c1a722b8a3e3bce5be995e652ce7"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-08-04 15:18:55.000000000","tz":-420},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2022-01-30 19:52:37.000000000","tz":60},"subject":"Fix failing copy_image flow init","message":"Fix failing copy_image flow init\n\nThe new networkx requirement apparently causes us to fail the\ntest_init_copy_flow_as_non_owner test because a None value can not be\nadded to the flow graph. This just mocks out the glance_store call\nthat is failing to get the staging store to avoid us choking there\nso we can proceed. We are just testing get_flow so we never use it\nanyway.\n\nCloses-Bug: #1959361\n\nChange-Id: I6fcb14ad240105ed0b1b9086c6c6c95034b4dd42\n(cherry picked from commit ded8ecf3828fa9afd7382f8ba850505ae0541e2c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b1d635c287f950b15df207455d64d97228963e5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b1d635c287f950b15df207455d64d97228963e5f"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
