)]}'
{"id":"openstack%2Fglance~827538","triplet_id":"openstack%2Fglance~master~I9b8d6141a59f2267a0021bded1794a7b96ace005","project":"openstack/glance","branch":"master","topic":"bp/cache-api","attention_set":{},"removed_from_attention_set":{"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2022-07-07 13:45:45.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I9b8d6141a59f2267a0021bded1794a7b96ace005","subject":"DNM: Test override g-api-r cache interval","status":"ABANDONED","created":"2022-02-02 18:04:27.000000000","updated":"2022-07-07 13:45:45.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ce8f273994d02b2ee18e0b3c03a7189ba5081e01","_number":827538,"virtual_id_number":827538,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-02-02 23:01:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-02 18:04:59.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2022-02-02 21:23:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cf97041296cb2690a642017637f723f469fa3fd4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-02-02 18:04:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1eff47b94d3bf0642af9f717f8cc96dab011fe9a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-02-02 19:23:30.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"22c2c5eb5672604823550aeadddea877dd793e4c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-02-02 20:34:12.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"0010254596c952c5d12fa58111e8d4dc47e1d09b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-02 21:23:51.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/47524626ff274a2f958de45495e14e7a : SUCCESS in 36m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b34ed5aa23b2443ab6e55750031c4f27 : SUCCESS in 6m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a17f1185b23c4920b8852eba3edb4562 : SUCCESS in 7m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b72f022777a64b11b87a1ebc7036e73c : SUCCESS in 6m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a7a3537745446988a1e394716944f07 : SUCCESS in 6m 56s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3e8e2f05d81d4dfd862612fb50df40dd : SUCCESS in 8m 53s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e9953d2c8436408292761e551f2f6284 : SUCCESS in 9m 35s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/f25c1fb4e40b48c7b053b19b62c009d1 : SUCCESS in 11m 19s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0da59824f45540e6ac715974bf66159d : POST_FAILURE in 34m 41s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f0589b87a49a4c3381dc5460ece4d2a8 : POST_FAILURE in 30m 41s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/221a9945320246389d4246797fd6b43d : POST_FAILURE in 30m 02s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/70ac5463ab6a4eb7acec4cd957a23be3 : SUCCESS in 41m 11s","accounts_in_message":[],"_revision_number":3},{"id":"dc7213c3e3b339d857b9eec3e5f04f486d5d48a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-02-02 21:27:37.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"c52d915b243700fc925b3865bdfb1434fa7911c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-02 23:01:06.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7d7840a00d7142e38ba0bb40cf2b3ce5 : SUCCESS in 45m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ada4094e072145d59e1bed41c88fef43 : SUCCESS in 6m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4901b5079d7a49f1a9b08407391ebdb4 : SUCCESS in 7m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6d65ea1e489e477eaf940358ec0c4833 : SUCCESS in 7m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c2dcc28771d4780bc1cb40ff5905570 : SUCCESS in 8m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f06b920d654f42b7948f5d1ed25d1dff : SUCCESS in 13m 06s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c78dab06cb9447618bcbe6898349be1c : SUCCESS in 14m 11s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/e20357cfef3743ea84926b1f2c1eda01 : SUCCESS in 8m 57s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ef041018e85d44e5b2d6bf6b491f9688 : FAILURE in 1h 18m 55s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f91756e3e1a14754b8c65d25b54e9558 : FAILURE in 59m 41s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8116e1bec5864d2e8a94ec3c8259c50f : FAILURE in 1h 30m 14s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/074c146730ab4c9c840c6e5ccff8bb5c : SUCCESS in 37m 08s","accounts_in_message":[],"_revision_number":4},{"id":"e402719da4826705e56df48a4cde267a2d4d4803","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-02-02 23:18:37.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"edd2fc7adad2e47bf51d8616d177670e49fe432c","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-02-03 16:36:15.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ce8f273994d02b2ee18e0b3c03a7189ba5081e01","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-07-07 13:45:45.000000000","message":"Abandoned\n\nNot required any more","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"9934c79b1421a0cd88dc626261412f7959564c8f","revisions":{"bf0dbed3a0084e1c187d6742ae7a9a50b3dc66e0":{"kind":"REWORK","_number":1,"created":"2022-02-02 18:04:27.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/38/827538/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/38/827538/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/827538/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/827538/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/827538/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/38/827538/1"}}},"commit":{"parents":[{"commit":"52f60e159cc04d45a7d27904fa0976ae78d60cb5","subject":"[DNM] Test image caching tempest tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/52f60e159cc04d45a7d27904fa0976ae78d60cb5"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2022-02-02 19:10:19.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2022-02-02 19:10:19.000000000","tz":-480},"subject":"DNM: Test override g-api-r cache interval","message":"DNM: Test override g-api-r cache interval\n\nIf this works, meld into previous\n\nChange-Id: I9b8d6141a59f2267a0021bded1794a7b96ace005\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bf0dbed3a0084e1c187d6742ae7a9a50b3dc66e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bf0dbed3a0084e1c187d6742ae7a9a50b3dc66e0"}]},"branch":"refs/heads/master"},"15a3e146845166fac78147a3e817356d846124a9":{"kind":"REWORK","_number":2,"created":"2022-02-02 19:23:30.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/38/827538/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/38/827538/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/827538/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/827538/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/827538/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/38/827538/2"}}},"commit":{"parents":[{"commit":"52f60e159cc04d45a7d27904fa0976ae78d60cb5","subject":"[DNM] Test image caching tempest tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/52f60e159cc04d45a7d27904fa0976ae78d60cb5"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2022-02-02 19:10:19.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2022-02-02 19:54:01.000000000","tz":-480},"subject":"DNM: Test override g-api-r cache interval","message":"DNM: Test override g-api-r cache interval\n\nIf this works, meld into previous\n\nChange-Id: I9b8d6141a59f2267a0021bded1794a7b96ace005\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/15a3e146845166fac78147a3e817356d846124a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/15a3e146845166fac78147a3e817356d846124a9"}]},"branch":"refs/heads/master"},"efea4200e87e993fae1ffe2cca5f88387c61fdc1":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-02-02 20:34:12.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/38/827538/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/38/827538/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/827538/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/827538/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/827538/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/38/827538/3"}}},"commit":{"parents":[{"commit":"52f60e159cc04d45a7d27904fa0976ae78d60cb5","subject":"[DNM] Test image caching tempest tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/52f60e159cc04d45a7d27904fa0976ae78d60cb5"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2022-02-02 19:10:19.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2022-02-02 21:40:05.000000000","tz":-480},"subject":"DNM: Test override g-api-r cache interval","message":"DNM: Test override g-api-r cache interval\n\nIf this works, meld into previous\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/827537\nChange-Id: I9b8d6141a59f2267a0021bded1794a7b96ace005\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/efea4200e87e993fae1ffe2cca5f88387c61fdc1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/efea4200e87e993fae1ffe2cca5f88387c61fdc1"}]},"branch":"refs/heads/master"},"9934c79b1421a0cd88dc626261412f7959564c8f":{"kind":"REWORK","_number":4,"created":"2022-02-02 21:27:37.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/38/827538/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/38/827538/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/827538/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/827538/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/827538/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/38/827538/4"}}},"commit":{"parents":[{"commit":"52f60e159cc04d45a7d27904fa0976ae78d60cb5","subject":"[DNM] Test image caching tempest tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/52f60e159cc04d45a7d27904fa0976ae78d60cb5"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2022-02-02 19:10:19.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2022-02-02 22:33:47.000000000","tz":-480},"subject":"DNM: Test override g-api-r cache interval","message":"DNM: Test override g-api-r cache interval\n\nIf this works, meld into previous\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/827537\nChange-Id: I9b8d6141a59f2267a0021bded1794a7b96ace005\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9934c79b1421a0cd88dc626261412f7959564c8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9934c79b1421a0cd88dc626261412f7959564c8f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
