)]}'
{"id":"openstack%2Fglance~828139","triplet_id":"openstack%2Fglance~stable%2Fwallaby~I060d8247c7b09f63990ea411e6c6a056bb50410d","project":"openstack/glance","branch":"stable/wallaby","topic":"fips-compatibility","attention_set":{},"removed_from_attention_set":{"5202":{"account":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"last_update":"2023-02-09 14:58:38.000000000","reason":"Change was abandoned"},"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2023-02-09 14:58:38.000000000","reason":"Change was abandoned"},"7973":{"account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"last_update":"2023-02-09 14:58:38.000000000","reason":"Change was abandoned"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2023-02-09 14:58:38.000000000","reason":"Change was abandoned"},"34120":{"account":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"last_update":"2023-02-09 14:58:38.000000000","reason":"Change was abandoned"},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2023-02-09 14:58:38.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I060d8247c7b09f63990ea411e6c6a056bb50410d","subject":"Add fips check jobs","status":"ABANDONED","created":"2022-02-07 18:29:28.000000000","updated":"2023-02-09 14:58:38.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":828179,"cherry_pick_of_patch_set":1,"meta_rev_id":"93a4a0ff93981aa712bd574c97d7dbfeeff96d8c","_number":828139,"virtual_id_number":828139,"owner":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-07-21 12:22:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-07 18:29:28.000000000","updated_by":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"reviewer":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"state":"CC"},{"updated":"2022-02-07 18:32:03.000000000","updated_by":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2022-02-07 18:32:03.000000000","updated_by":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"},{"updated":"2022-02-07 18:32:03.000000000","updated_by":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2022-02-07 18:32:03.000000000","updated_by":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2022-02-07 18:32:03.000000000","updated_by":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2022-02-07 19:58:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1801bf7031a4437c5bc5859637931b44ee164495","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2022-02-07 18:29:28.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.","accounts_in_message":[],"_revision_number":1},{"id":"a234bee89517b7e7f37143f1e158a2c048055839","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-07 19:58:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/01ae72e966d347caac6b465c11955a10 : SUCCESS in 56m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fbb3c4756105403389d8b84324436f16 : SUCCESS in 1h 06m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/564d48072bd3460a99064f8818f600da : SUCCESS in 51m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/50438524670b4f179b913e6a59698633 : SUCCESS in 7m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9531fcadbab442e7a02bf94415502cce : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b32dc7ff0e3541e081a8b588f8f94538 : SUCCESS in 5m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c206a14c37e14fb3ada9e402147369f5 : SUCCESS in 6m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ac67c69cf9b5466c8f4b201ee54b63cb : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e06b512e9bde4b3d933dc560f227835e : SUCCESS in 7m 37s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/f8236aa9de7445bba87ff157e02e7973 : FAILURE in 10m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1c0e325f7fe343c2b608d9db606af4d1 : SUCCESS in 8m 36s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/60a5b867390c41acb4c19e132549bfc6 : SUCCESS in 8m 50s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/601cd4e6b696426a98eda5733ee99645 : SUCCESS in 6m 13s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/2e7ed0a0656b48b5aabec2a98e964a7c : SUCCESS in 1h 26m 24s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/8534e586c7b84c648e9bc95e8edc7548 : SUCCESS in 1h 05m 39s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/1bb1ba7f71754e578f1d56bc12a47263 : SUCCESS in 1h 14m 06s\n- glance-multistore-cinder-import-fips https://zuul.opendev.org/t/openstack/build/02be5417de1f44c6a20e14b5a2b8a571 : FAILURE in 35m 06s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2b087f6f6b0f43cba1e06c0926a975a7 : SUCCESS in 1h 03m 00s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/633b8ac2305c4f0fa7a2daa7607844af : SUCCESS in 1h 09m 16s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/eb4ae2e92a3d44f0828a379f405bf5a6 : SUCCESS in 30m 05s","accounts_in_message":[],"_revision_number":1},{"id":"f4dff950b3b3673b9ea82f24ce72717cc44774e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2022-02-08 10:15:41.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"95bdcff3bfa5f5cb9bab2a101122a57d2af52da1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2022-02-08 10:16:04.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"073e7191743376ab0f8a11734b00dfa3e96a889f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 11:50:58.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/b1b6d100e7e8401d96ae949eb04e2184 : SUCCESS in 1h 04m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7eda1a16b6414d3abce24f48a4f39456 : FAILURE in 1h 10m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f5465a21366c4271a3d817473648782a : SUCCESS in 49m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0dd4c7d8ef884963bfc43f8a6e3d9097 : SUCCESS in 8m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8344acdea44c47019220a656c9537e65 : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/100f2a6096c54da3b1aa7e20468f8d3b : SUCCESS in 6m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4a5a6ddb61aa461ba6d85a4b5e15c02a : SUCCESS in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9793c77e7b524a22988abf51cc26b4ec : SUCCESS in 5m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e59c82f8f60740769df964a9f88212ff : SUCCESS in 5m 48s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/7b5143a43b314d99932402a13735797a : SUCCESS in 10m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/412b6a2ea5df4350a6d2e8c86666e475 : TIMED_OUT in 31m 03s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a9aa295299774eae9b3cb2e89cf8fd54 : SUCCESS in 8m 53s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/db4a8090724943ea9e49955d0af1359f : SUCCESS in 4m 37s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/b647de4d9a044c31ac99454c84c2ebbb : SUCCESS in 1h 15m 00s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b951e6c63bcc444d9f70467d0aa3835a : SUCCESS in 1h 11m 34s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/cf68113246434b9b909310679619da37 : SUCCESS in 1h 27m 57s\n- glance-multistore-cinder-import-fips https://zuul.opendev.org/t/openstack/build/3e4ea83f4b364146997bd0679e666a20 : FAILURE in 38m 59s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/652217ee289a4666aa4ea8e60bdd7dd8 : SUCCESS in 52m 49s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/91762928e1c84778a1a2d0a0d252ed9c : SUCCESS in 1h 05m 31s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8afb4e6a1bb04ad7a7ebb267d5f0c847 : SUCCESS in 31m 33s","accounts_in_message":[],"_revision_number":3},{"id":"c76070c5a5749539cbb6579ce49fc5582f2a5cda","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2022-02-09 14:23:19.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"54dea692030005c7ee8a1072f2f4dc229647c573","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 16:34:51.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/9555fadb90c744d2b378ba6bccb25730 : SUCCESS in 1h 04m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9ca9ded5d601435ea851aacc406afcda : SUCCESS in 1h 10m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/49d4a0508bd849a39ef3abaeee65be71 : SUCCESS in 54m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/66feb7b602614a0bac86e53b92b0f768 : SUCCESS in 8m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c837982617dc49c883edb23948795f18 : SUCCESS in 7m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/efc2d35e4d1d42a09ae9e95b48249a64 : SUCCESS in 7m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e5ae8fcbb83a4a2e852c0a03bb5d9a67 : SUCCESS in 7m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/76789dde96604b0f8981f9176c7b2447 : SUCCESS in 9m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/799f86186160486882cf3b3d073bae77 : SUCCESS in 8m 43s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/962e408698454423a9eaff0ffcd6b23d : SUCCESS in 13m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/29a8bbcbe8604b00bf1ebc62a5c1f4ad : SUCCESS in 11m 04s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/66270ca53bae40e6b682529b2a9729fd : SUCCESS in 10m 54s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/dcc675a66c63409397d3815e0e6330e7 : SUCCESS in 7m 13s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/5b49e524f8424df99ff395609f52c75c : SUCCESS in 1h 28m 22s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7cfc1986c5c2425cb16eb60b4b0d9751 : SUCCESS in 1h 18m 50s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/0a949bf87bbb4c6abedd4a51ead8553d : SUCCESS in 1h 11m 10s\n- glance-multistore-cinder-import-fips https://zuul.opendev.org/t/openstack/build/226fce9095cb42a1b3a36797967996eb : FAILURE in 39m 13s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/13c91ccc7d6549b0b63559bf31b91e41 : SUCCESS in 1h 03m 47s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7c8ea2c587a44b5fa4d2a727ae778503 : SUCCESS in 1h 49m 33s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/aa6d081f5dda4e6db7c7c4382b51077a : SUCCESS in 37m 16s","accounts_in_message":[],"_revision_number":3},{"id":"11f627463d9728a4e08d690060db1ddb7bee8804","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2022-02-10 16:31:36.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"8300ed0956e70677e718a345ac8eb196fc5b2841","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2022-02-10 16:32:02.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"249fa508104ff3e609017528640407f29c1cd482","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2022-02-10 16:33:18.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"da49f736fa5d83b68374149c8a9239abb73121e3","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2022-02-10 16:39:07.000000000","message":"Topic set to fips-compatibility","accounts_in_message":[],"_revision_number":6},{"id":"763eeb2bb68ca1bf2af57854457cf1bb920c4136","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 18:19:50.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/303fb586a525400dbff4e748906eab4b : SUCCESS in 1h 01m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8c7eed80230b44d0a5c7e3e6482f229b : SUCCESS in 1h 26m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ffe0d6bc4fd247c3a01af980d4cc5352 : SUCCESS in 58m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4efcbaf14eeb48c29089980b314767b8 : SUCCESS in 7m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d6e9fee43914be1aa75246674f7c56d : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aa57d263505343df8052babe0a464e71 : SUCCESS in 6m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d3a987a2f0264a628dce796251e6e43b : SUCCESS in 6m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/854ccfbae2c74de0a242c587c624c8b5 : SUCCESS in 7m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0be52f29f5694c939e650554d8887c6f : SUCCESS in 7m 43s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/5a0038e18091459995121ea0ef8fec67 : SUCCESS in 14m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6b5c94a6ebb34d0d8c57a942553b0fa0 : SUCCESS in 8m 26s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6752710dba944b5d92f7591ccb86f368 : SUCCESS in 9m 02s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/e6cc58d5d707460282231c22324778a0 : SUCCESS in 5m 19s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/0183e6c6ccfd4db988aa83ab65d5b780 : SUCCESS in 1h 20m 35s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6bf6c58d31fd4d659f1f2dd01cb7b3de : SUCCESS in 1h 18m 20s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/4bbe83ca21ef4d419c293fa265ebce04 : SUCCESS in 1h 35m 59s\n- glance-multistore-cinder-import-fips https://zuul.opendev.org/t/openstack/build/fbb2c586756748a897633a8d70a8f7c6 : FAILURE in 1h 28m 13s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9f3a488a3b4040fa97ea3a4caaa33a00 : SUCCESS in 46m 24s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a9bd46dc493041faa694133ff58db0a2 : SUCCESS in 1h 11m 00s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/75d8ab1e2b1a451483840839be352157 : SUCCESS in 37m 43s","accounts_in_message":[],"_revision_number":6},{"id":"9dacc52f13f0382ff37d1abf0803a7d01682c06b","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2022-07-20 13:35:09.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"f683f73a20cb591b209f20f8ca22e196ae138043","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 15:12:33.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8bd3d7a4c82c4193b9499605963c710c\n\n- grenade https://zuul.opendev.org/t/openstack/build/e0db4d3c89b041f4b7b9d727908e9397 : SUCCESS in 1h 04m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d8f4ca062b81482d9031a5495ae3f6fc : SUCCESS in 1h 33m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e6ed42148e54495490c72acc2baacc5c : SUCCESS in 49m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ed45e49a8d9548eb985a5c89f89a8df7 : SUCCESS in 7m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41b0a8ff14fe4ce88717f723b4d973ad : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/066de8166894446f8032c720e2edd217 : SUCCESS in 9m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/48d1f1c199a24acd83cbe080ce1b9e30 : SUCCESS in 5m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ec64006a84364668ad35cc3df8ead3f5 : SUCCESS in 9m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/144f2d4c647e4fb1aac51b8d344b5c91 : SUCCESS in 7m 01s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/bf6d1389842c43c8a9fcd5e11d3bb5ce : SUCCESS in 14m 35s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bda7d4e90fc54d06bcbaa3ca060834f7 : TIMED_OUT in 30m 55s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7a7c7b787b7d4b89b9903ca9157770ee : SUCCESS in 9m 21s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/d2958896e4694659b5422d9a6eb55e44 : SUCCESS in 3m 49s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/7690e92e16d8478d895e7f1a5c261ff5 : SUCCESS in 1h 16m 11s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/5c9a1b5ffc6f4d8a804c2368a6abc4e7 : SUCCESS in 1h 30m 59s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/5371cd892d5f498ba5a1ab3c92edee95 : SUCCESS in 1h 12m 59s\n- glance-multistore-cinder-import-fips https://zuul.opendev.org/t/openstack/build/5857d71954dd41b28d50c18fa35cc8bd : RETRY_LIMIT in 14m 19s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a57029d6f79c4e1083d427c393ecaa03 : SUCCESS in 57m 32s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3dce52d4f14c4728bbaa29eb58e82609 : SUCCESS in 1h 31m 28s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0272850ecd514dc086e41efd3aadf502 : SUCCESS in 34m 28s","accounts_in_message":[],"_revision_number":6},{"id":"5c8c4c2475468c0f690e9ca20392da7d85bd0d6a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2022-07-21 11:21:21.000000000","message":"Patch Set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"ce3bf380ac3d6229d107ad6af21b5df64d511b4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 12:22:12.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/db04e3ccd5d444ebbecfc3ee26580327\n\n- grenade https://zuul.opendev.org/t/openstack/build/f3d5a43839244e6992eb920dfe598e2b : SUCCESS in 52m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7a61516e207c4e2ca7a4090e2a05e338 : FAILURE in 12m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/baa2bee860ff4c929c281df622d1e205 : POST_FAILURE in 11m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ed30f15932bc47b8986eb088267ee6d0 : SUCCESS in 8m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2535dba61d3946639a4427e3d29fc725 : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ba1fce42b26449f086ab810aa3f9d6d7 : SUCCESS in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1d611cfb76864b1b83f0ae85413c2750 : SUCCESS in 6m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c9306c625b924f58ab5d84e0e0dce74c : SUCCESS in 10m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f9ff93f748a492c97bdd9082028eaf4 : SUCCESS in 5m 53s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/5879ef5359ba400198301a76651b0692 : SUCCESS in 14m 46s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5e98749bcb01437694379e054909fa51 : SUCCESS in 12m 00s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8cdd3065a11d4ddfaf88d1b0c3b2be5b : SUCCESS in 11m 40s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/a045d101fcfb4d3e830f3568562a1210 : SUCCESS in 7m 38s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/758db8dd1e6348e0bc6e21b60a2462a1 : FAILURE in 18m 22s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/69eac0759c644fcdb413a8af5a5f8bce : POST_FAILURE in 20m 51s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b1609681a84849b8b36084258386e207 : POST_FAILURE in 21m 48s\n- glance-multistore-cinder-import-fips https://zuul.opendev.org/t/openstack/build/30ed6af246d04277bde21ddcdd8842a6 : RETRY_LIMIT in 21m 19s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7a573dfaf28b4ea09a14d837bf33de8a : FAILURE in 20m 59s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/32c5372c05084b60a8690b5c11509006 : FAILURE in 23m 27s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/18ff4fa15d914db084e8d96b03910b0a : FAILURE in 15m 24s","accounts_in_message":[],"_revision_number":7},{"id":"62c9caadaccee023499b4bbacb3df3ebcf4e0607","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2022-07-21 19:11:27.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"02418090f82dae82eb90c50cfdf3543571746ea6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 20:14:09.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2747a61b848444d7879741102fe70b86\n\n- grenade https://zuul.opendev.org/t/openstack/build/f5d76e42aa014dbdb77fd0bc630790c2 : SUCCESS in 47m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/43e47069f1a54bc0bdf510903dd03f62 : FAILURE in 11m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0de9e4fd427e4304bb037f339c8fac19 : POST_FAILURE in 14m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0416604ea129458dbfd8ce7bb2ae2bf7 : SUCCESS in 8m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a59dc6218964650be1fef01f419dfb3 : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c2322c7edb874941a1932d4462163735 : SUCCESS in 5m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2a335f06cb234ea58920d7c06dc250b2 : SUCCESS in 6m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/468d29c535c04336ba729b60633e8502 : SUCCESS in 10m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8f32753052b4918a0f6877158885035 : SUCCESS in 9m 16s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/b7e571bc9c6b468ab373f7766ca0fe66 : FAILURE in 15m 45s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cf54b29e5fae440bad6b2127342fdfae : SUCCESS in 11m 41s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4519d3bf948d4b8dba51f4fc3e25046b : SUCCESS in 13m 12s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/8e3f1f9c1cfe4116b2b035ede3d02b46 : SUCCESS in 7m 58s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f1dbc8b71c884e7381261220e276b27b : FAILURE in 14m 21s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/1b83a4115ecd4141895b759db17996c3 : POST_FAILURE in 22m 30s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/a9bd96cb13784fdfa1670f1994208dd8 : POST_FAILURE in 12m 30s\n- glance-multistore-cinder-import-fips https://zuul.opendev.org/t/openstack/build/2f70a580a35341c2b781c0dddef202ab : RETRY_LIMIT in 20m 52s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2e7f6f237ec24da19e8a2187e3f838e1 : FAILURE in 21m 29s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7f157803844a4e20809675fee44e01a0 : FAILURE in 17m 50s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/48584adfff33486fb77d84c5390b9f34 : FAILURE in 21m 58s","accounts_in_message":[],"_revision_number":7},{"id":"93a4a0ff93981aa712bd574c97d7dbfeeff96d8c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-02-09 14:58:38.000000000","message":"Abandoned\n\nThis branch is too old, we will not merge this.","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"139b5de076c15a9a0eb0b9fa4e1e49030ed0fb65","revisions":{"1f116b6613c0f2672f46fbea9c0e3ed50fcb7ee6":{"kind":"REWORK","_number":1,"created":"2022-02-07 18:29:28.000000000","uploader":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"ref":"refs/changes/39/828139/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/39/828139/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/39/828139/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/39/828139/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/39/828139/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/39/828139/1"}}},"commit":{"parents":[{"commit":"b1d635c287f950b15df207455d64d97228963e5f","subject":"Fix failing copy_image flow init","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b1d635c287f950b15df207455d64d97228963e5f"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-05-11 00:50:29.000000000","tz":-240},"committer":{"name":"Andre Aranha","email":"afariasa@redhat.com","date":"2022-02-07 18:29:28.000000000","tz":0},"subject":"Add fips check jobs","message":"Add fips check jobs\n\nSome jobs have been modified to run to confirm that functionality is\nstill working when FIPS is enabled on the nodes.\n\nAs the FIPS tests currently run on centos nodes, code is added\nto the test-setup script to set up the databases correctly.  Also had\nto increase the swap space on the nodes; see [0] for an explanation.\n\nThe current FIPS jobs run using python 3.6 on centos-8-stream.  We will\nmodify these to run on centos-9-stream and python 3.9 in a\nsubsequent patch.\n\n[0] https://review.opendev.org/c/openstack/devstack/+/803706\n\nChange-Id: I060d8247c7b09f63990ea411e6c6a056bb50410d\n(cherry picked from commit 0f13754f84919ce095cd6962e4b369c926738b77)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1f116b6613c0f2672f46fbea9c0e3ed50fcb7ee6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1f116b6613c0f2672f46fbea9c0e3ed50fcb7ee6"}]},"branch":"refs/heads/stable/wallaby"},"684ce5d5a945fb689d281fcf5cf3bba36c06ba7f":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-02-08 10:15:41.000000000","uploader":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"ref":"refs/changes/39/828139/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/39/828139/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/39/828139/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/39/828139/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/39/828139/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/39/828139/2"}}},"commit":{"parents":[{"commit":"436372e357c3108eb4edb058bf5ffe7d37053404","subject":"Add missing packages to test dependencies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/436372e357c3108eb4edb058bf5ffe7d37053404"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-05-11 00:50:29.000000000","tz":-240},"committer":{"name":"Andre Aranha","email":"afariasa@redhat.com","date":"2022-02-08 10:15:41.000000000","tz":0},"subject":"Add fips check jobs","message":"Add fips check jobs\n\nSome jobs have been modified to run to confirm that functionality is\nstill working when FIPS is enabled on the nodes.\n\nAs the FIPS tests currently run on centos nodes, code is added\nto the test-setup script to set up the databases correctly.  Also had\nto increase the swap space on the nodes; see [0] for an explanation.\n\nThe current FIPS jobs run using python 3.6 on centos-8-stream.  We will\nmodify these to run on centos-9-stream and python 3.9 in a\nsubsequent patch.\n\n[0] https://review.opendev.org/c/openstack/devstack/+/803706\n\nChange-Id: I060d8247c7b09f63990ea411e6c6a056bb50410d\n(cherry picked from commit 0f13754f84919ce095cd6962e4b369c926738b77)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/684ce5d5a945fb689d281fcf5cf3bba36c06ba7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/684ce5d5a945fb689d281fcf5cf3bba36c06ba7f"}]},"branch":"refs/heads/stable/wallaby","description":"Rebase"},"aa83942d71e84496321fa3bb5afb78bb81b3ac91":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-02-08 10:16:04.000000000","uploader":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"ref":"refs/changes/39/828139/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/39/828139/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/39/828139/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/39/828139/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/39/828139/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/39/828139/3"}}},"commit":{"parents":[{"commit":"436372e357c3108eb4edb058bf5ffe7d37053404","subject":"Add missing packages to test dependencies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/436372e357c3108eb4edb058bf5ffe7d37053404"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-05-11 00:50:29.000000000","tz":-240},"committer":{"name":"Andre Aranha","email":"afariasa@redhat.com","date":"2022-02-08 10:16:04.000000000","tz":0},"subject":"Add fips check jobs","message":"Add fips check jobs\n\nSome jobs have been modified to run to confirm that functionality is\nstill working when FIPS is enabled on the nodes.\n\nAs the FIPS tests currently run on centos nodes, code is added\nto the test-setup script to set up the databases correctly.  Also had\nto increase the swap space on the nodes; see [0] for an explanation.\n\nThe current FIPS jobs run using python 3.6 on centos-8-stream.  We will\nmodify these to run on centos-9-stream and python 3.9 in a\nsubsequent patch.\n\n[0] https://review.opendev.org/c/openstack/devstack/+/803706\n\nDepends-On: https://review.opendev.org/c/openstack/glance/+/828240\n\nChange-Id: I060d8247c7b09f63990ea411e6c6a056bb50410d\n(cherry picked from commit 0f13754f84919ce095cd6962e4b369c926738b77)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/aa83942d71e84496321fa3bb5afb78bb81b3ac91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/aa83942d71e84496321fa3bb5afb78bb81b3ac91"}]},"branch":"refs/heads/stable/wallaby","description":"Edit commit message"},"940812e79401147980ca22fd21aa90747f23934f":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-02-10 16:31:36.000000000","uploader":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"ref":"refs/changes/39/828139/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/39/828139/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/39/828139/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/39/828139/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/39/828139/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/39/828139/4"}}},"commit":{"parents":[{"commit":"b1d635c287f950b15df207455d64d97228963e5f","subject":"Fix failing copy_image flow init","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b1d635c287f950b15df207455d64d97228963e5f"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-05-11 00:50:29.000000000","tz":-240},"committer":{"name":"Andre Aranha","email":"afariasa@redhat.com","date":"2022-02-10 16:31:36.000000000","tz":0},"subject":"Add fips check jobs","message":"Add fips check jobs\n\nSome jobs have been modified to run to confirm that functionality is\nstill working when FIPS is enabled on the nodes.\n\nAs the FIPS tests currently run on centos nodes, code is added\nto the test-setup script to set up the databases correctly.  Also had\nto increase the swap space on the nodes; see [0] for an explanation.\n\nThe current FIPS jobs run using python 3.6 on centos-8-stream.  We will\nmodify these to run on centos-9-stream and python 3.9 in a\nsubsequent patch.\n\n[0] https://review.opendev.org/c/openstack/devstack/+/803706\n\nDepends-On: https://review.opendev.org/c/openstack/glance/+/828240\n\nChange-Id: I060d8247c7b09f63990ea411e6c6a056bb50410d\n(cherry picked from commit 0f13754f84919ce095cd6962e4b369c926738b77)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/940812e79401147980ca22fd21aa90747f23934f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/940812e79401147980ca22fd21aa90747f23934f"}]},"branch":"refs/heads/stable/wallaby","description":"Rebase"},"a3076895a62c3a4fe4baaaeb6fe6efde83f1ac3e":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2022-02-10 16:32:02.000000000","uploader":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"ref":"refs/changes/39/828139/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/39/828139/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/39/828139/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/39/828139/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/39/828139/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/39/828139/5"}}},"commit":{"parents":[{"commit":"b1d635c287f950b15df207455d64d97228963e5f","subject":"Fix failing copy_image flow init","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b1d635c287f950b15df207455d64d97228963e5f"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-05-11 00:50:29.000000000","tz":-240},"committer":{"name":"Andre Aranha","email":"afariasa@redhat.com","date":"2022-02-10 16:32:02.000000000","tz":0},"subject":"Add fips check jobs","message":"Add fips check jobs\n\nSome jobs have been modified to run to confirm that functionality is\nstill working when FIPS is enabled on the nodes.\n\nAs the FIPS tests currently run on centos nodes, code is added\nto the test-setup script to set up the databases correctly.  Also had\nto increase the swap space on the nodes; see [0] for an explanation.\n\nThe current FIPS jobs run using python 3.6 on centos-8-stream.  We will\nmodify these to run on centos-9-stream and python 3.9 in a\nsubsequent patch.\n\n[0] https://review.opendev.org/c/openstack/devstack/+/803706\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/828752\n\nChange-Id: I060d8247c7b09f63990ea411e6c6a056bb50410d\n(cherry picked from commit 0f13754f84919ce095cd6962e4b369c926738b77)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a3076895a62c3a4fe4baaaeb6fe6efde83f1ac3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a3076895a62c3a4fe4baaaeb6fe6efde83f1ac3e"}]},"branch":"refs/heads/stable/wallaby","description":"Edit commit message"},"e94c9b9d4f0a4c80211e5029271e1ae42dcaac07":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2022-02-10 16:33:18.000000000","uploader":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"ref":"refs/changes/39/828139/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/39/828139/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/39/828139/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/39/828139/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/39/828139/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/39/828139/6"}}},"commit":{"parents":[{"commit":"79353a34ec6b1b8ab1cf232c0bd477f4e11fb435","subject":"Add missing packages to test dependencies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/79353a34ec6b1b8ab1cf232c0bd477f4e11fb435"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-05-11 00:50:29.000000000","tz":-240},"committer":{"name":"Andre Aranha","email":"afariasa@redhat.com","date":"2022-02-10 16:33:18.000000000","tz":0},"subject":"Add fips check jobs","message":"Add fips check jobs\n\nSome jobs have been modified to run to confirm that functionality is\nstill working when FIPS is enabled on the nodes.\n\nAs the FIPS tests currently run on centos nodes, code is added\nto the test-setup script to set up the databases correctly.  Also had\nto increase the swap space on the nodes; see [0] for an explanation.\n\nThe current FIPS jobs run using python 3.6 on centos-8-stream.  We will\nmodify these to run on centos-9-stream and python 3.9 in a\nsubsequent patch.\n\n[0] https://review.opendev.org/c/openstack/devstack/+/803706\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/828752\n\nChange-Id: I060d8247c7b09f63990ea411e6c6a056bb50410d\n(cherry picked from commit 0f13754f84919ce095cd6962e4b369c926738b77)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e94c9b9d4f0a4c80211e5029271e1ae42dcaac07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e94c9b9d4f0a4c80211e5029271e1ae42dcaac07"}]},"branch":"refs/heads/stable/wallaby","description":"Rebase"},"139b5de076c15a9a0eb0b9fa4e1e49030ed0fb65":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2022-07-21 11:21:21.000000000","uploader":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"ref":"refs/changes/39/828139/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/39/828139/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/39/828139/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/39/828139/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/39/828139/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/39/828139/7"}}},"commit":{"parents":[{"commit":"79353a34ec6b1b8ab1cf232c0bd477f4e11fb435","subject":"Add missing packages to test dependencies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/79353a34ec6b1b8ab1cf232c0bd477f4e11fb435"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-05-11 00:50:29.000000000","tz":-240},"committer":{"name":"Andre Aranha","email":"afariasa@redhat.com","date":"2022-07-21 11:21:21.000000000","tz":0},"subject":"Add fips check jobs","message":"Add fips check jobs\n\nSome jobs have been modified to run to confirm that functionality is\nstill working when FIPS is enabled on the nodes.\n\nAs the FIPS tests currently run on centos nodes, code is added\nto the test-setup script to set up the databases correctly.  Also had\nto increase the swap space on the nodes; see [0] for an explanation.\n\nThe current FIPS jobs run using python 3.6 on centos-8-stream.  We will\nmodify these to run on centos-9-stream and python 3.9 in a\nsubsequent patch.\n\n[0] https://review.opendev.org/c/openstack/devstack/+/803706\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/848284\n\nChange-Id: I060d8247c7b09f63990ea411e6c6a056bb50410d\n(cherry picked from commit 0f13754f84919ce095cd6962e4b369c926738b77)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/139b5de076c15a9a0eb0b9fa4e1e49030ed0fb65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/139b5de076c15a9a0eb0b9fa4e1e49030ed0fb65"}]},"branch":"refs/heads/stable/wallaby","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
