)]}'
{"id":"openstack%2Fglance~828179","triplet_id":"openstack%2Fglance~stable%2Fxena~I060d8247c7b09f63990ea411e6c6a056bb50410d","project":"openstack/glance","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"5202":{"account":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"last_update":"2023-02-09 15:02:56.000000000","reason":"Change was abandoned"},"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2023-02-09 15:02:56.000000000","reason":"Change was abandoned"},"7973":{"account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"last_update":"2023-02-09 15:02:56.000000000","reason":"Change was abandoned"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2022-05-05 14:52:52.000000000","reason":"Abhishek Kekane replied on the change"},"34120":{"account":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"last_update":"2023-02-09 15:02:56.000000000","reason":"Change was abandoned"},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2023-02-09 15:02:56.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I060d8247c7b09f63990ea411e6c6a056bb50410d","subject":"Add fips check jobs","status":"ABANDONED","created":"2022-02-07 18:27:17.000000000","updated":"2023-02-09 15:02:56.000000000","total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"483733f0ceb990acc33205a748a5326b92b9528c","_number":828179,"virtual_id_number":828179,"owner":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-02-10 18:01:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":2,"date":"2022-05-05 14:52:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":2,"date":"2022-07-18 13:39:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":1,"date":"2022-07-18 13:39:25.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"}],"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-07 18:27:17.000000000","updated_by":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"reviewer":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"state":"CC"},{"updated":"2022-02-07 18:31:02.000000000","updated_by":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2022-02-07 18:31:02.000000000","updated_by":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"},{"updated":"2022-02-07 18:31:02.000000000","updated_by":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2022-02-07 18:31:02.000000000","updated_by":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2022-02-07 18:31:02.000000000","updated_by":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2022-02-07 19:40:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-18 13:39:25.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"922ca8efd3cf9e9bcbd7f9a8b4563fdcfbd116ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2022-02-07 18:27:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"59804ebe3bfe70bf3ee60868976dc6ec0db5cb88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-07 19:40:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/d8bdfec9228847708ef703e014d6bfeb : SUCCESS in 57m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f01adf3fa72b44d3a588dd8f6aa6e73e : SUCCESS in 55m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9f836d46fa154cb6a04a598c822e9971 : SUCCESS in 42m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55502e5bf6a541be83587682e787cb97 : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a8c021be5de1482d8741de86168a51db : SUCCESS in 6m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c4fb593df26242f78fb438ab4fbf7573 : SUCCESS in 6m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bfb59cd415da48ecb1f3f3f0cd8c1a47 : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd82e48e2baf4b159573b1a69a2c222b : SUCCESS in 6m 53s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/469849d13649434285b03e94fb3e8955 : FAILURE in 10m 58s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/63a8467d7aef477294693f8ea785fd79 : SUCCESS in 10m 01s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c788142ebce140cf84bda5f781cf7fd3 : SUCCESS in 8m 30s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/ca2b164e1b754111a1c73cb6861f203f : SUCCESS in 8m 52s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/2cc4d87e2b094965bed22cbc9548482f : SUCCESS in 5m 38s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/fc150f5fa9f54f42bb89a59aabd121d9 : SUCCESS in 1h 10m 35s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/41360fe1b9ac4244b5c25ffd92c80f8d : SUCCESS in 1h 07m 19s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/ce089b73014e49c8ab682aadaf322d84 : SUCCESS in 1h 02m 55s\n- glance-multistore-cinder-import-fips https://zuul.opendev.org/t/openstack/build/43148175b8a5419192e29eca91a0a748 : FAILURE in 31m 12s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/347f5dfa2fa245fd8619e3af804f390d : SUCCESS in 43m 13s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/52b735ca9396498ca7acd531c833dd09 : SUCCESS in 1h 10m 52s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b4f2ced266f048b192f8c61b954b8037 : SUCCESS in 33m 22s","accounts_in_message":[],"_revision_number":1},{"id":"b317c009bb9456b7775cc8ce0e5eeefed03e8829","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2022-02-08 10:10:00.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"09d2cc203783b0530e065768045cd15adf86aec6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2022-02-08 10:11:43.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"d2bfce61d4b329a3d6d51a616914aca3fc007b69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 11:28:51.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/b98f0feaa70a4bb791f37faa1d8d96a1 : SUCCESS in 1h 00m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a115c4ab2b43455986b738fae19b0ec1 : SUCCESS in 1h 13m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8011e2f99818413ca831a8803a29d213 : SUCCESS in 54m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62697d027e5042a592c20a26e67381ee : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7fd8409da4464bcd8cc3a34ddc47f294 : SUCCESS in 7m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/053f729093b94caabd8b45bd965e8015 : SUCCESS in 8m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d44e964094604614bc0552d387165607 : SUCCESS in 7m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab4076737ae24163a3144f0a92ec15ce : SUCCESS in 8m 33s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/6eae0b61804b47c7875c56b7480a06cb : FAILURE in 20m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2b7e94175d7143e49c0ecb64949191c3 : SUCCESS in 9m 12s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/33018638fc9a44e38412bfaff9d9eab9 : SUCCESS in 8m 45s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/d184a0f542b14d63afc52ed7c78f01f6 : SUCCESS in 8m 21s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/75356907134a436b9296a794f4ff76f0 : SUCCESS in 7m 13s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/183f441d6c404fcfb0763c70f04ab248 : SUCCESS in 1h 02m 59s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/9b6df778fe2e4507a54ec224500f2b33 : SUCCESS in 56m 01s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/203d754da9be41d4a3d751f225c40e7d : SUCCESS in 1h 05m 12s\n- glance-multistore-cinder-import-fips https://zuul.opendev.org/t/openstack/build/1ae2e29f2cd7485a92bfcce4256bf508 : POST_FAILURE in 22m 10s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b201e5b6d28a467c9204dec016ff8184 : SUCCESS in 42m 18s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d37e57548cbd46e48af7aec3bb86f6ec : SUCCESS in 1h 03m 55s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b04fe1fe6b574c0b9bf8f76214b153b0 : SUCCESS in 26m 20s","accounts_in_message":[],"_revision_number":3},{"id":"915c5622370f51a72c9f5cfa26d9484959298ea2","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2022-02-09 11:36:11.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4f71f44622a670c56d87cdd39462ff44f28da844","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 13:33:06.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/77452a58237e45cb8913927c398fa4d6 : SUCCESS in 1h 11m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b94ccba909ea4eada533233c3d72156d : SUCCESS in 1h 01m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c38b7e3f7b3448a2ae91fa5d83e1db0e : SUCCESS in 34m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0100dca977364c8d84062a75938d0039 : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/59a5679c0f594f1e98992efab3fd2fcd : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/556e9916f0154c14a1706e7f420cac5a : SUCCESS in 5m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f1afb28d8f894198acd27cef6299cedd : SUCCESS in 6m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c3c66df02604ccdbe1423aa26674126 : SUCCESS in 5m 44s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/8daf1e700c834bb1bd538230656d581e : SUCCESS in 12m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ca0c4b92920d445e9b0099cd54c75033 : SUCCESS in 8m 36s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2258c3bad78b4fefb574e62b6be45c66 : SUCCESS in 9m 46s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/096c6ba8c6254a5f871215b6a1ad2d69 : SUCCESS in 7m 52s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/cc4a43dfd14d41b3b035df07538dbbb9 : SUCCESS in 5m 53s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d7f87a8688bc4267977d0ed8c06e766f : SUCCESS in 58m 43s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6f888887be2c49d28410e18e2e592a3b : SUCCESS in 1h 06m 36s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b824ad3143a7466fa02fdf45bbf0bc1f : SUCCESS in 50m 57s\n- glance-multistore-cinder-import-fips https://zuul.opendev.org/t/openstack/build/fdb375b0b0724cefb236eb9bd9587d4c : FAILURE in 35m 03s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5e37ce30cb654139a78829fe59456317 : SUCCESS in 41m 35s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e6a75a80110d4556904bd9d80c49be7b : SUCCESS in 1h 07m 31s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f93540abf59e4a96a233f7b983cdff60 : SUCCESS in 35m 35s","accounts_in_message":[],"_revision_number":3},{"id":"b4cb59e65e175e32797ac1c7a6150764b85ab753","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2022-02-10 15:28:45.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"a6130f2d8e263397ef246117ccb942c930916694","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2022-02-10 15:28:53.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"c1f253fdcb7dc2a12bf9e85329fd2a6b39566d73","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2022-02-10 15:29:35.000000000","message":"Patch Set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"1a6efd13321fee161a8e211df9f5608b749d4310","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 15:30:47.000000000","message":"Patch Set 6: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":6},{"id":"ab6aedd6f5a92486db6b40d1d8cf3eb5134c6246","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2022-02-10 16:27:53.000000000","message":"Patch Set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"0491b5453928fc5ec2061ef354a9d08972e282ce","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2022-02-10 16:28:02.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"5b7cefea3024f6052fcba26742faf989098dd3d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 18:01:52.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/85f318d7c4004fac8713c40c3a88fd80 : SUCCESS in 1h 03m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/100a952b0d2e4e8fb7baa1f11117f996 : SUCCESS in 54m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c1c55b96f00b4aa798d74062726efbbe : SUCCESS in 53m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/422af9884bad43e5b1c1178e27c37de7 : SUCCESS in 6m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1315b78ea0084a23ab0c74ef6f1128b4 : SUCCESS in 6m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3b5a5776ac7949199d7e37f1fc85cc69 : SUCCESS in 8m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c2341e9500c24a25943ef41756909a4b : SUCCESS in 7m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aacbe7a705e14f2bbdbfdc83dd031f92 : SUCCESS in 7m 51s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/170010bc54eb44119bb34785ec7308bc : SUCCESS in 12m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5ecf1efb1cc34bf88754d70fed8d0007 : SUCCESS in 11m 34s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/61ba2209c65d4729a641a2e6460c324f : SUCCESS in 11m 28s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/72d1865caecc4d8eb744dff65d2c011d : SUCCESS in 9m 53s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/884fe77df6264f04b166e8e54c56ce39 : SUCCESS in 6m 44s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/a89d8199c9534ae3a02e18478089b8f8 : SUCCESS in 1h 11m 10s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0358f381aa5841969ff6f3049d089ccb : SUCCESS in 1h 14m 48s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/5a1df18589d545c491dd4d9d232a000b : SUCCESS in 1h 07m 49s\n- glance-multistore-cinder-import-fips https://zuul.opendev.org/t/openstack/build/1fa008de329f416989aee79a2657bbda : FAILURE in 1h 27m 00s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/22dabef6bb7d4fb5bbecc233aaf04cb7 : SUCCESS in 44m 36s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/646c554307eb4960bff0c8188937c4ce : SUCCESS in 1h 09m 34s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f7b03ac8fbfa4de5adf1987764f7a2c5 : SUCCESS in 36m 09s","accounts_in_message":[],"_revision_number":7},{"id":"620ced69744a046488d79d0109eca37d7a2c2284","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2022-02-16 10:33:39.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"e22c57a3803bfd99afcb1436bbbfdf449e041727","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 12:18:11.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/b25673bb8ce64511b26e0e20c024f6e5 : SUCCESS in 1h 25m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6d4c0f512406465ea5278a2f6ad4e708 : SUCCESS in 1h 04m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/83945ccce70a42709da519676f6f6f13 : SUCCESS in 49m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8cf979575e64e50b46cb79d44bf7722 : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1ca1dd0b2df54ffaa458464ef7aaeef3 : SUCCESS in 5m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e662ce19b3fc44ef98a532c6b7562fcc : SUCCESS in 5m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6e27afb34e6c4bbbaedb2555ad3960c9 : SUCCESS in 6m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be83d9ccda1d4fa39bd5e65db873f917 : SUCCESS in 6m 46s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/ef5aca30beef4951a9eba06c94852cc5 : SUCCESS in 10m 23s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8a18593f045c492798c74fa398a65371 : SUCCESS in 9m 09s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/87d4ac3fa15242d0a0fe5fea7779b989 : SUCCESS in 8m 44s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/8939ffbb3de942518514e2563b6f2ce7 : SUCCESS in 8m 47s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/6a52d58c42094ba688b2a3c75536fa7c : SUCCESS in 5m 37s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/576dc5fd97d9462aaf8a05e810bcdd88 : SUCCESS in 1h 09m 49s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f41fc42ddf0d4525b37e0435cde4d73a : SUCCESS in 1h 19m 16s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/445cadbb8834461e94ed716e2a83fbab : SUCCESS in 1h 04m 50s\n- glance-multistore-cinder-import-fips https://zuul.opendev.org/t/openstack/build/4459a7c9261042af81ffd5a7ef3a680d : FAILURE in 1h 36m 09s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e1b70e04bb484fe38ef00a73d1294c05 : SUCCESS in 1h 05m 12s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a20c13bfc0d44dd0b6bbe8ea772b0dbe : SUCCESS in 1h 21m 38s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b6bb88d926f54dc987b7493376dd7ea2 : SUCCESS in 44m 30s","accounts_in_message":[],"_revision_number":7},{"id":"1dc82dcccdb5e4ab32b39994250fd22ddbdc7de5","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2022-03-02 09:55:43.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"6a56d219e11eb3e00de3bcd1463c9463d839c8f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 11:38:38.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/50a47b457fd04c7db2599fbb06f5bf82 : SUCCESS in 1h 08m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8c3f96b42f1d428ea5a7bf7bc62c4758 : SUCCESS in 1h 03m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/69ee2149d2514a08854750e06b4e956e : SUCCESS in 47m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f2367ac53a54ca2bfb9ad660d08198f : SUCCESS in 7m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ea6c0d2e5db54faba7fe3c885e48695e : SUCCESS in 7m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d33369d842c54d448c2d9b31b300ec6c : SUCCESS in 7m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0bcd8a6fb9204b9eb6b9d3b453b29880 : SUCCESS in 8m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/371fb70e510347bdaeedf81dba603563 : SUCCESS in 8m 12s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/d97e479bc98f4b1d8615561a3892bd54 : SUCCESS in 13m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9e50303edd144a5898dca44ecf0ffc46 : SUCCESS in 10m 29s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7ace693482c04a8898c3dc1899d4ac66 : SUCCESS in 10m 21s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/02a8274216414e249bd1f48546b70653 : SUCCESS in 10m 33s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/88f3c1c9fb2c4ee78bcb4d9cbc0302aa : SUCCESS in 7m 00s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/5cfa42312f1b439589e7e371e31f6f9a : SUCCESS in 1h 26m 18s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e6a25c43476a4c3588a39ea44a5bf94e : SUCCESS in 1h 17m 51s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/fbc43ac4c56c45238c72aafcd6940524 : SUCCESS in 1h 19m 14s\n- glance-multistore-cinder-import-fips https://zuul.opendev.org/t/openstack/build/48865969a89c48668c3356cdeff402d9 : FAILURE in 1h 36m 01s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/99cd4e433cd74e6eb79a9f109711043a : SUCCESS in 51m 17s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5d62de045f0640b785a049e8b047e8a1 : SUCCESS in 1h 21m 53s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7ff5895ec0f545508734418a0fe35e1a : SUCCESS in 35m 34s","accounts_in_message":[],"_revision_number":7},{"id":"b36c99f3f0e8f0416c0d85be2d8720c4ed47d9d7","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-05-05 14:52:52.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"f7856d17141293f7d3fa845d119cba8ec3674bfb","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-07-18 13:39:25.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"483733f0ceb990acc33205a748a5326b92b9528c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-02-09 15:02:56.000000000","message":"Abandoned\n\nThis branch is too old, we will not merge this.","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"ab4554b8866c1583bad43eee9e6ab74665cb6dc3","revisions":{"dd6aad23dd7392670388f8b07ba02ea34c168f03":{"kind":"REWORK","_number":1,"created":"2022-02-07 18:27:17.000000000","uploader":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"ref":"refs/changes/79/828179/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/79/828179/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/79/828179/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/79/828179/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/79/828179/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/79/828179/1"}}},"commit":{"parents":[{"commit":"b76dd7880cf40f8f2cc9425c54941c5d637401a2","subject":"[stable-only] Specify openstacksdk stable branch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b76dd7880cf40f8f2cc9425c54941c5d637401a2"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-05-11 00:50:29.000000000","tz":-240},"committer":{"name":"afariasa","email":"afariasa@redhat.com","date":"2022-02-07 18:27:36.000000000","tz":0},"subject":"Add fips check jobs","message":"Add fips check jobs\n\nSome jobs have been modified to run to confirm that functionality is\nstill working when FIPS is enabled on the nodes.\n\nAs the FIPS tests currently run on centos nodes, code is added\nto the test-setup script to set up the databases correctly.  Also had\nto increase the swap space on the nodes; see [0] for an explanation.\n\nThe current FIPS jobs run using python 3.6 on centos-8-stream.  We will\nmodify these to run on centos-9-stream and python 3.9 in a\nsubsequent patch.\n\n[0] https://review.opendev.org/c/openstack/devstack/+/803706\n\nChange-Id: I060d8247c7b09f63990ea411e6c6a056bb50410d\n(cherry picked from commit 0f13754f84919ce095cd6962e4b369c926738b77)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dd6aad23dd7392670388f8b07ba02ea34c168f03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dd6aad23dd7392670388f8b07ba02ea34c168f03"}]},"branch":"refs/heads/stable/xena"},"d4446cb96b661dbc641d3aa6975238c09c24148a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-02-08 10:10:00.000000000","uploader":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"ref":"refs/changes/79/828179/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/79/828179/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/79/828179/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/79/828179/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/79/828179/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/79/828179/2"}}},"commit":{"parents":[{"commit":"f57da038e14b9b4dff8a75547dbc1b577f45c7f8","subject":"Add missing packages to test dependencies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f57da038e14b9b4dff8a75547dbc1b577f45c7f8"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-05-11 00:50:29.000000000","tz":-240},"committer":{"name":"Andre Aranha","email":"afariasa@redhat.com","date":"2022-02-08 10:10:00.000000000","tz":0},"subject":"Add fips check jobs","message":"Add fips check jobs\n\nSome jobs have been modified to run to confirm that functionality is\nstill working when FIPS is enabled on the nodes.\n\nAs the FIPS tests currently run on centos nodes, code is added\nto the test-setup script to set up the databases correctly.  Also had\nto increase the swap space on the nodes; see [0] for an explanation.\n\nThe current FIPS jobs run using python 3.6 on centos-8-stream.  We will\nmodify these to run on centos-9-stream and python 3.9 in a\nsubsequent patch.\n\n[0] https://review.opendev.org/c/openstack/devstack/+/803706\n\nChange-Id: I060d8247c7b09f63990ea411e6c6a056bb50410d\n(cherry picked from commit 0f13754f84919ce095cd6962e4b369c926738b77)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d4446cb96b661dbc641d3aa6975238c09c24148a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d4446cb96b661dbc641d3aa6975238c09c24148a"}]},"branch":"refs/heads/stable/xena","description":"Rebase"},"0bf43ee77d4969865a76561efd23d609e673b25a":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-02-08 10:11:43.000000000","uploader":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"ref":"refs/changes/79/828179/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/79/828179/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/79/828179/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/79/828179/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/79/828179/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/79/828179/3"}}},"commit":{"parents":[{"commit":"f57da038e14b9b4dff8a75547dbc1b577f45c7f8","subject":"Add missing packages to test dependencies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f57da038e14b9b4dff8a75547dbc1b577f45c7f8"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-05-11 00:50:29.000000000","tz":-240},"committer":{"name":"Andre Aranha","email":"afariasa@redhat.com","date":"2022-02-08 10:11:43.000000000","tz":0},"subject":"Add fips check jobs","message":"Add fips check jobs\n\nSome jobs have been modified to run to confirm that functionality is\nstill working when FIPS is enabled on the nodes.\n\nAs the FIPS tests currently run on centos nodes, code is added\nto the test-setup script to set up the databases correctly.  Also had\nto increase the swap space on the nodes; see [0] for an explanation.\n\nThe current FIPS jobs run using python 3.6 on centos-8-stream.  We will\nmodify these to run on centos-9-stream and python 3.9 in a\nsubsequent patch.\n\n[0] https://review.opendev.org/c/openstack/devstack/+/803706\n\nDepends-On: https://review.opendev.org/c/openstack/glance/+/828233\n\nChange-Id: I060d8247c7b09f63990ea411e6c6a056bb50410d\n(cherry picked from commit 0f13754f84919ce095cd6962e4b369c926738b77)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0bf43ee77d4969865a76561efd23d609e673b25a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0bf43ee77d4969865a76561efd23d609e673b25a"}]},"branch":"refs/heads/stable/xena","description":"Edit commit message"},"63c4501508dcbe77d08a4d1ebf01d91ef8a2492d":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2022-02-10 15:28:45.000000000","uploader":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"ref":"refs/changes/79/828179/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/79/828179/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/79/828179/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/79/828179/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/79/828179/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/79/828179/4"}}},"commit":{"parents":[{"commit":"f57da038e14b9b4dff8a75547dbc1b577f45c7f8","subject":"Add missing packages to test dependencies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f57da038e14b9b4dff8a75547dbc1b577f45c7f8"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-05-11 00:50:29.000000000","tz":-240},"committer":{"name":"Andre Aranha","email":"afariasa@redhat.com","date":"2022-02-10 15:28:45.000000000","tz":0},"subject":"Add fips check jobs","message":"Add fips check jobs\n\nSome jobs have been modified to run to confirm that functionality is\nstill working when FIPS is enabled on the nodes.\n\nAs the FIPS tests currently run on centos nodes, code is added\nto the test-setup script to set up the databases correctly.  Also had\nto increase the swap space on the nodes; see [0] for an explanation.\n\nThe current FIPS jobs run using python 3.6 on centos-8-stream.  We will\nmodify these to run on centos-9-stream and python 3.9 in a\nsubsequent patch.\n\n[0] https://review.opendev.org/c/openstack/devstack/+/803706\n\nChange-Id: I060d8247c7b09f63990ea411e6c6a056bb50410d\n(cherry picked from commit 0f13754f84919ce095cd6962e4b369c926738b77)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/63c4501508dcbe77d08a4d1ebf01d91ef8a2492d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/63c4501508dcbe77d08a4d1ebf01d91ef8a2492d"}]},"branch":"refs/heads/stable/xena","description":"Edit commit message"},"8c56895b924f587a52b831b0bf361031b5024b3c":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2022-02-10 15:28:53.000000000","uploader":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"ref":"refs/changes/79/828179/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/79/828179/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/79/828179/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/79/828179/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/79/828179/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/79/828179/5"}}},"commit":{"parents":[{"commit":"de3c61b37bd4a0f53f9866704a14de9fc8f32972","subject":"Add missing packages to test dependencies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/de3c61b37bd4a0f53f9866704a14de9fc8f32972"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-05-11 00:50:29.000000000","tz":-240},"committer":{"name":"Andre Aranha","email":"afariasa@redhat.com","date":"2022-02-10 15:28:53.000000000","tz":0},"subject":"Add fips check jobs","message":"Add fips check jobs\n\nSome jobs have been modified to run to confirm that functionality is\nstill working when FIPS is enabled on the nodes.\n\nAs the FIPS tests currently run on centos nodes, code is added\nto the test-setup script to set up the databases correctly.  Also had\nto increase the swap space on the nodes; see [0] for an explanation.\n\nThe current FIPS jobs run using python 3.6 on centos-8-stream.  We will\nmodify these to run on centos-9-stream and python 3.9 in a\nsubsequent patch.\n\n[0] https://review.opendev.org/c/openstack/devstack/+/803706\n\nChange-Id: I060d8247c7b09f63990ea411e6c6a056bb50410d\n(cherry picked from commit 0f13754f84919ce095cd6962e4b369c926738b77)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8c56895b924f587a52b831b0bf361031b5024b3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8c56895b924f587a52b831b0bf361031b5024b3c"}]},"branch":"refs/heads/stable/xena","description":"Rebase"},"91e2ee6354e4339a63ce0c7b17fe7bf7a271f428":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2022-02-10 15:29:35.000000000","uploader":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"ref":"refs/changes/79/828179/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/79/828179/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/79/828179/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/79/828179/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/79/828179/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/79/828179/6"}}},"commit":{"parents":[{"commit":"de3c61b37bd4a0f53f9866704a14de9fc8f32972","subject":"Add missing packages to test dependencies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/de3c61b37bd4a0f53f9866704a14de9fc8f32972"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-05-11 00:50:29.000000000","tz":-240},"committer":{"name":"Andre Aranha","email":"afariasa@redhat.com","date":"2022-02-10 15:29:35.000000000","tz":0},"subject":"Add fips check jobs","message":"Add fips check jobs\n\nSome jobs have been modified to run to confirm that functionality is\nstill working when FIPS is enabled on the nodes.\n\nAs the FIPS tests currently run on centos nodes, code is added\nto the test-setup script to set up the databases correctly.  Also had\nto increase the swap space on the nodes; see [0] for an explanation.\n\nThe current FIPS jobs run using python 3.6 on centos-8-stream.  We will\nmodify these to run on centos-9-stream and python 3.9 in a\nsubsequent patch.\n\n[0] https://review.opendev.org/c/openstack/devstack/+/803706\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/828661\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/828723\n\nChange-Id: I060d8247c7b09f63990ea411e6c6a056bb50410d\n(cherry picked from commit 0f13754f84919ce095cd6962e4b369c926738b77)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/91e2ee6354e4339a63ce0c7b17fe7bf7a271f428"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/91e2ee6354e4339a63ce0c7b17fe7bf7a271f428"}]},"branch":"refs/heads/stable/xena","description":"Edit commit message"},"ab4554b8866c1583bad43eee9e6ab74665cb6dc3":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2022-02-10 16:27:53.000000000","uploader":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"ref":"refs/changes/79/828179/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/79/828179/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/79/828179/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/79/828179/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/79/828179/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/79/828179/7"}}},"commit":{"parents":[{"commit":"de3c61b37bd4a0f53f9866704a14de9fc8f32972","subject":"Add missing packages to test dependencies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/de3c61b37bd4a0f53f9866704a14de9fc8f32972"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-05-11 00:50:29.000000000","tz":-240},"committer":{"name":"Andre Aranha","email":"afariasa@redhat.com","date":"2022-02-10 16:27:53.000000000","tz":0},"subject":"Add fips check jobs","message":"Add fips check jobs\n\nSome jobs have been modified to run to confirm that functionality is\nstill working when FIPS is enabled on the nodes.\n\nAs the FIPS tests currently run on centos nodes, code is added\nto the test-setup script to set up the databases correctly.  Also had\nto increase the swap space on the nodes; see [0] for an explanation.\n\nThe current FIPS jobs run using python 3.6 on centos-8-stream.  We will\nmodify these to run on centos-9-stream and python 3.9 in a\nsubsequent patch.\n\n[0] https://review.opendev.org/c/openstack/devstack/+/803706\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/828723\n\nChange-Id: I060d8247c7b09f63990ea411e6c6a056bb50410d\n(cherry picked from commit 0f13754f84919ce095cd6962e4b369c926738b77)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ab4554b8866c1583bad43eee9e6ab74665cb6dc3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ab4554b8866c1583bad43eee9e6ab74665cb6dc3"}]},"branch":"refs/heads/stable/xena","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
