)]}'
{"id":"openstack%2Fglance~828240","triplet_id":"openstack%2Fglance~stable%2Fwallaby~If488ae4299a6cab521eec9c5e9ce4f29f94057e1","project":"openstack/glance","branch":"stable/wallaby","topic":"bug/1946457-stable/xena-stable/wallaby","attention_set":{},"removed_from_attention_set":{"5202":{"account":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"last_update":"2023-02-09 14:58:15.000000000","reason":"Change was abandoned"},"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2022-05-05 14:39:42.000000000","reason":"Brian Rosmaita replied on the change"},"7973":{"account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"last_update":"2023-02-09 14:58:15.000000000","reason":"Change was abandoned"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2022-07-13 05:19:31.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},"34120":{"account":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"last_update":"2023-02-09 14:58:15.000000000","reason":"Change was abandoned"},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2023-02-09 14:58:15.000000000","reason":"Change was abandoned"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2023-02-09 14:58:15.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"If488ae4299a6cab521eec9c5e9ce4f29f94057e1","subject":"Update qcow test to use qcow2 instead","status":"ABANDONED","created":"2022-02-08 10:14:08.000000000","updated":"2023-02-09 14:58:15.000000000","total_comment_count":7,"unresolved_comment_count":1,"has_review_started":true,"cherry_pick_of_change":828233,"cherry_pick_of_patch_set":1,"meta_rev_id":"4119d4bfac27a9e9e4dfa33998b847d76dc0576c","_number":828240,"virtual_id_number":828240,"owner":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-02-06 19:56:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2022-11-21 19:49:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":2,"date":"2022-07-13 05:19:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2022-11-21 19:49:37.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-08 10:49:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-02 10:00:16.000000000","updated_by":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2022-03-02 10:00:16.000000000","updated_by":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2022-03-02 10:00:16.000000000","updated_by":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"},{"updated":"2022-03-02 10:00:16.000000000","updated_by":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2022-03-02 10:00:16.000000000","updated_by":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2022-07-12 18:05:35.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"CC"}],"messages":[{"id":"5abe6a1759a942b44edae9d02710dd48756e69f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2022-02-08 10:14:08.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.","accounts_in_message":[],"_revision_number":1},{"id":"04343d19da873b89196bcb7686027dc27481eb8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 10:49:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/556e4158c7504ffe9f8f0cd57ea08c7b : SUCCESS in 6m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83ac9c0a5e3a4d6b9b3eb653ec07d8af : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8c06aa0ef29d48679146134d21f74479 : SUCCESS in 6m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ab9d4bc1e3d5466a89c969673fde16b0 : SUCCESS in 5m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3b66aee15b394dfe9cecbc2d27e33c02 : SUCCESS in 6m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bf3039bcf3342e384b8a784edff03be : SUCCESS in 7m 38s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6fe23b4b43de442e81223a2ddf4f7b6b : SUCCESS in 12m 15s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e3358e12db0745d3b5f589e35aa2958c : SUCCESS in 11m 32s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/2969dbf153ef44a085a791c32106d9f3 : SUCCESS in 6m 28s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b38a3f3f24d844e1a97cea24cb52b283 : SUCCESS in 32m 40s","accounts_in_message":[],"_revision_number":1},{"id":"acdef8b21a63f41f0212f47c529d0ed848df93c2","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-05-05 14:39:42.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e3d6b94aafec25bd5c19ac3576be5237bf80bfb0","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-05-05 14:40:40.000000000","message":"Patch Set 1: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f0779ed0dbaf77f67448d62a2d5b425c6f333068","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-05-05 14:43:31.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"05bf153e7c328335d262f16d7e3b17284ae3bacd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-07-12 18:05:06.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"8d9fe389cab21e882e67f6004835f46e12aa8eea","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-07-12 18:05:35.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"319b48600f8bf4c716a63ac1677b47aa8eecbe23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 18:51:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c932c7834cee403692033a73f9b75061\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b339d78504234ebcbed6bf2e2f213015 : SUCCESS in 11m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58800d9d83dd4e7d9a36a5e4d0e60905 : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1a61f572048743fe8230d6809711ead2 : SUCCESS in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f8c22ce98c2d42acbea6a6a2542598ab : SUCCESS in 5m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d5cb155bbe2540e7a04ec9bc0f5eb8e3 : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b99ec607e84740eaa10cb79e632ead2b : SUCCESS in 5m 36s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b7219e801d8e4bacb2bd0b0905e1d53a : SUCCESS in 7m 57s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9bd4603cea3f435e80cee3b19b7f92c9 : SUCCESS in 7m 48s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/3ad8170fb5a849119b2df2bbb1d69b1d : SUCCESS in 4m 35s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/00ab7053fb3747b7a4676cc305769e7c : SUCCESS in 44m 27s","accounts_in_message":[],"_revision_number":2},{"id":"c68bd97ce03238677c9d8defe6aa3f85d426c871","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-07-13 05:19:31.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"953a05eb5ff276ecb162c5559210adee44a5b357","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-11-21 19:49:37.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"14140d950f5145271907e3ed0fc20cd1e8bcb6ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-21 19:50:21.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"786ed49ee1b4724ce28bc231a7a122f1152fd234","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-21 21:37:39.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d11fa9a3454e40e5a8c249d30339b042\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1406ffbe337a41f3a7d610548f375831 : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0de33269496f4e5ab9eae5ab1bc2e4c2 : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4d96e8352da04fcfbe84c6f44ec76fe2 : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/111de1e9d97045e291633fbd3a10da56 : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef9be13edd3b42e8bd9d0e37efa8efe7 : SUCCESS in 5m 35s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5a28c0d8fe104a63a5dcb3ac839f7c17 : SUCCESS in 7m 48s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e91f9813a8434655a920da24764261cf : SUCCESS in 8m 13s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/859c4d5fd88849ac8cf6b4c8a021847c : FAILURE in 36m 45s","accounts_in_message":[],"_revision_number":2},{"id":"e8f3758c724452eb76fd77de5bfa56c45d92ec54","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-02-06 19:11:41.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9be34ede12a4e6e051460c7e44eda7a2da4b5d63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-06 19:56:51.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/987109474110429f8a9afb9f40ca50fb\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8e35becee1bd41ac948d64e28cfa26bd : SUCCESS in 6m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fcb0725114a4472ab3bc907adf8ecbfd : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ba07ebb59a2d42e79b47c6f75e133e21 : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0ce5a5033a8b4e39907a3d686ecc8ee6 : SUCCESS in 5m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e924d570dedf41b08a11c9809a806083 : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89cb34fa947e46dea0aa3f2c78a39c1e : FAILURE in 5m 12s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f5894b4f107a44d7a0da447a5b8b178d : SUCCESS in 8m 13s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c897f68fc052420ca4cee656bf0a25ee : SUCCESS in 8m 21s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/eb04bf70c589483dbf28378c8bc92d28 : SUCCESS in 38m 19s","accounts_in_message":[],"_revision_number":2},{"id":"4119d4bfac27a9e9e4dfa33998b847d76dc0576c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-02-09 14:58:15.000000000","message":"Abandoned\n\nThis branch is too old, we will not merge this.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c97dacb8d0a1260c0aaf19e20ee2bb6547f77584","revisions":{"445bbb272463a7abf3bbedd2f8c3129713f2a758":{"kind":"REWORK","_number":1,"created":"2022-02-08 10:14:08.000000000","uploader":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"ref":"refs/changes/40/828240/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/40/828240/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/40/828240/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/40/828240/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/40/828240/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/40/828240/1"}}},"commit":{"parents":[{"commit":"b1d635c287f950b15df207455d64d97228963e5f","subject":"Fix failing copy_image flow init","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b1d635c287f950b15df207455d64d97228963e5f"}]}],"author":{"name":"Andre Aranha","email":"afariasa@redhat.com","date":"2021-10-11 10:20:38.000000000","tz":0},"committer":{"name":"Andre Aranha","email":"afariasa@redhat.com","date":"2022-02-08 10:14:08.000000000","tz":0},"subject":"Update qcow test to use qcow2 instead","message":"Update qcow test to use qcow2 instead\n\nqemu-img package for Centos 8 doesn\u0027t support qcow disk format anymore,\ndue to this, tests using qcow are failing with\n`Unknown file format \u0027qcow\u0027`.\nThis patch updates the qcow tests for qcow2 instead.\n\nCloses-Bug: #1946457\n\nChange-Id: If488ae4299a6cab521eec9c5e9ce4f29f94057e1\n(cherry picked from commit 947e541fc0f7c0fc2b759898ab8b9401d152f3d0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/445bbb272463a7abf3bbedd2f8c3129713f2a758"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/445bbb272463a7abf3bbedd2f8c3129713f2a758"}]},"branch":"refs/heads/stable/wallaby"},"c97dacb8d0a1260c0aaf19e20ee2bb6547f77584":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-07-12 18:05:06.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/40/828240/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/40/828240/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/40/828240/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/40/828240/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/40/828240/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/40/828240/2"}}},"commit":{"parents":[{"commit":"b1d635c287f950b15df207455d64d97228963e5f","subject":"Fix failing copy_image flow init","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b1d635c287f950b15df207455d64d97228963e5f"}]}],"author":{"name":"Andre Aranha","email":"afariasa@redhat.com","date":"2021-10-11 10:20:38.000000000","tz":0},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2022-07-12 18:04:25.000000000","tz":120},"subject":"Update qcow test to use qcow2 instead","message":"Update qcow test to use qcow2 instead\n\nqemu-img package for Centos 8 doesn\u0027t support qcow disk format anymore,\ndue to this, tests using qcow are failing with\n`Unknown file format \u0027qcow\u0027`.\nThis patch updates the qcow tests for qcow2 instead.\n\nCloses-Bug: #1946457\n\nChange-Id: If488ae4299a6cab521eec9c5e9ce4f29f94057e1\n(cherry picked from commit 947e541fc0f7c0fc2b759898ab8b9401d152f3d0)\n(cherry picked from commit 881b24a2ad882ba06e16acd8e5a54eeb9d30e1a5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c97dacb8d0a1260c0aaf19e20ee2bb6547f77584"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c97dacb8d0a1260c0aaf19e20ee2bb6547f77584"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
