)]}'
{"id":"openstack%2Fglance~828335","triplet_id":"openstack%2Fglance~master~Ib244469e09e396ea897d1778bb1bad15cebf1f10","project":"openstack/glance","branch":"master","topic":"py3-only","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-02-15 06:52:45.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib244469e09e396ea897d1778bb1bad15cebf1f10","subject":"Add Python 3 only classifier","status":"MERGED","created":"2022-02-08 14:02:26.000000000","updated":"2022-02-15 06:56:03.000000000","submitted":"2022-02-15 06:52:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"828335-py3-only","meta_rev_id":"081b753fba5e1cba8309551d927341c9f1284862","_number":828335,"virtual_id_number":828335,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-15 06:52:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2022-02-15 05:31:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2022-02-14 19:27:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2022-02-15 05:31:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-08 16:05:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-14 19:27:15.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2022-02-15 05:31:41.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"bba3534455a8315bec7e491b41e8a4ab0569b006","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-08 14:02:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"16e879d238cdf2899fcc1421fbac68797743c259","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-08 14:05:59.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"c0d9d96bee5d5b06481f03a637c55e8d1b11072a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-08 14:06:11.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"66b3bdceca1d92ebf3908ea9feebcdb143d48cb2","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-08 14:06:28.000000000","message":"Topic set to python3-only-generate-constraints","accounts_in_message":[],"_revision_number":3},{"id":"6349903817e3a715d943082b7bbc1b0b4c8f9eb5","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-08 14:06:29.000000000","message":"Topic python3-only-generate-constraints removed","accounts_in_message":[],"_revision_number":3},{"id":"862fcf703b52302608d927f6080ec5cffb29aa7e","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-08 14:06:45.000000000","message":"Topic set to py3-only","accounts_in_message":[],"_revision_number":3},{"id":"ae2b90b5163c97bb4c83b123bca387b4af1232ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 16:05:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/57d27523ab4843c8bd4a20da7968bb03 : SUCCESS in 4m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/32dcec20604b4a5a8e12276df743b386 : SUCCESS in 57m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6e4661c936dd49b39904fd8e959af2d6 : SUCCESS in 1h 13m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8e7a6eafade34938ac17065983679b9e : SUCCESS in 48m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e59384d904e2423aa7f68cda1149482d : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0bca3f6c48cc491e8249b0179ba34d2e : SUCCESS in 7m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5f293129b9da402abe1bab9363155d3e : SUCCESS in 7m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f64a30bb02584e28b75c5241dde86701 : SUCCESS in 8m 34s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/6fb074658b2a4a88beaf50ca5d88cbfd : SUCCESS in 13m 44s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ce81dd086d9c4f73826b0241f7b45a26 : SUCCESS in 11m 07s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/d719efcce9b74faeba179a0efcd22af6 : SUCCESS in 10m 26s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/00e6a45f48304b18a8e1a184df9fc123 : SUCCESS in 1h 03m 55s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/cd12636d5be143b1ab6551ebd7bc9a5d : SUCCESS in 1h 12m 38s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f18aba1cc15446fa9035e97887716490 : SUCCESS in 1h 10m 36s\n- glance-multistore-cinder-import-fips https://zuul.opendev.org/t/openstack/build/1dddd1e17fdf4b9eaf848cd181a38a2e : SUCCESS in 1h 44m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/849d1cea66bb439088d9f9bd6c36b3bd : SUCCESS in 51m 33s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f27f017a4dd44c3aa1ae789d50b91932 : SUCCESS in 1h 21m 47s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1688a8d029c34de09934db00de989099 : SUCCESS in 34m 53s","accounts_in_message":[],"_revision_number":3},{"id":"3068296f7b616922c3d6227762c6593076822ff0","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-02-14 19:27:15.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"edc5c8af13a93b0e00f627d58ab8fe53ac75c486","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-02-15 05:31:41.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c1ca0999baa1e5e8467d516913df473cda2fe7a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 05:32:35.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ec7150625c1a9f7031e1ca956850c4442a803f1d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 06:52:45.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/fbe8edf9288b4ba092dbe0659d68ba4e : SUCCESS in 2m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/49eb1d7044c44382a07f9ef488784900 : SUCCESS in 57m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/35a5eaf4c95d41439e42cb97d92f6de0 : SUCCESS in 1h 09m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2f9d8c157ff04c80963669c0d460ae74 : SUCCESS in 43m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8b51be744654a1586961ade667ef86f : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/184f4278b83a4a7aa73f2a310f56f169 : SUCCESS in 6m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d3d76aaa7efa4e9dac78f70299891369 : SUCCESS in 6m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ab7fa1695e34af8bd605dc093630b28 : SUCCESS in 6m 12s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9249fd40313c4c2c93128dade32ec5db : SUCCESS in 9m 41s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/31b5382a37244bdaa440a84bc91cdfb4 : SUCCESS in 9m 38s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4e834d2c61124f23b8ea32d1a5e60212 : SUCCESS in 1h 08m 59s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/bc9f1bce13f84608bdf458ab9a9ee83e : SUCCESS in 1h 04m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f2ec2e431b8147419b636391237de4d2 : SUCCESS in 53m 01s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7a58e94023114bc3a01800201a9b5d67 : SUCCESS in 1h 11m 40s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/521a8110acc446b1842145f31e17fcf3 : SUCCESS in 31m 48s","accounts_in_message":[],"_revision_number":3},{"id":"4fbbd7f207db0a4bbb1c5208b623eff9d4b9bcc3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 06:52:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"081b753fba5e1cba8309551d927341c9f1284862","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 06:56:03.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f649c58db044e0492a130b61375020c : SUCCESS in 58s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"bf2266159626aa3ba1473ba1db06914b37aa67ec","revisions":{"bd43fae30e7f91a2cc8df0be11f5543f503a7d4b":{"kind":"REWORK","_number":1,"created":"2022-02-08 14:02:26.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/35/828335/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/35/828335/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/828335/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/828335/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/828335/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/35/828335/1"}}},"commit":{"parents":[{"commit":"502fa0ffc8970c087c5924742231812c0da6a114","subject":"Remove six.iteritems usage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/502fa0ffc8970c087c5924742231812c0da6a114"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-08 14:01:31.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-08 14:01:31.000000000","tz":540},"subject":"Add Python 3 only classifier","message":"Add Python 3 only classifier\n\nCurrent Glance no longer supports Python 2. This adds the classifier\nto clearly states that, following the other components.\n\nChange-Id: Ib244469e09e396ea897d1778bb1bad15cebf1f10\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bd43fae30e7f91a2cc8df0be11f5543f503a7d4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bd43fae30e7f91a2cc8df0be11f5543f503a7d4b"}]},"branch":"refs/heads/master"},"cd4e02296fc33439e6b3e640205ef30d3f7aec24":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-02-08 14:05:59.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/35/828335/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/35/828335/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/828335/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/828335/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/828335/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/35/828335/2"}}},"commit":{"parents":[{"commit":"502fa0ffc8970c087c5924742231812c0da6a114","subject":"Remove six.iteritems usage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/502fa0ffc8970c087c5924742231812c0da6a114"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-08 14:01:31.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-08 14:05:59.000000000","tz":0},"subject":"Add Python 3 only classifier","message":"Add Python 3 only classifier\n\nPython 2 support was removed during Ussuri cycle. This change adds\nthe classifier to clearly states that only Python 3 is supported.\n\nChange-Id: Ib244469e09e396ea897d1778bb1bad15cebf1f10\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cd4e02296fc33439e6b3e640205ef30d3f7aec24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cd4e02296fc33439e6b3e640205ef30d3f7aec24"}]},"branch":"refs/heads/master","description":"Edit commit message"},"bf2266159626aa3ba1473ba1db06914b37aa67ec":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-02-08 14:06:11.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/35/828335/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/35/828335/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/828335/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/828335/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/828335/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/35/828335/3"}}},"commit":{"parents":[{"commit":"502fa0ffc8970c087c5924742231812c0da6a114","subject":"Remove six.iteritems usage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/502fa0ffc8970c087c5924742231812c0da6a114"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-08 14:01:31.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-08 14:06:10.000000000","tz":0},"subject":"Add Python 3 only classifier","message":"Add Python 3 only classifier\n\nPython 2 support was removed during Ussuri cycle. This change adds\nthe classifier to clearly state that only Python 3 is supported.\n\nChange-Id: Ib244469e09e396ea897d1778bb1bad15cebf1f10\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bf2266159626aa3ba1473ba1db06914b37aa67ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bf2266159626aa3ba1473ba1db06914b37aa67ec"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
