)]}'
{"id":"openstack%2Fglance~829025","triplet_id":"openstack%2Fglance~master~I612121ff4edd00861baed546d8f18775f02860af","project":"openstack/glance","branch":"master","topic":"delete-rs-type","attention_set":{"19138":{"account":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"last_update":"2022-02-14 15:23:03.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I612121ff4edd00861baed546d8f18775f02860af","subject":"Add new API to delete metadef resource type","status":"NEW","created":"2022-02-14 13:30:47.000000000","updated":"2022-02-15 05:42:40.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c5dc13525f5b0c9561e7b9f8af83ad3a3de7b832","_number":829025,"virtual_id_number":829025,"owner":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-02-14 15:23:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":-1,"date":"2022-02-15 05:42:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-14 15:23:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-15 05:42:40.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"0708dfe00d6fd5d28722b312bd955c3386129b28","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2022-02-14 13:30:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dda39fdae25716222e8c7a66cf5d2fd52210bc89","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2022-02-14 13:33:58.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"462bb39189aff6415c957a7c92f45668d30f088d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2022-02-14 13:36:10.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"93a77ca6b2d0757656438fc61a6a9670865977e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 15:23:03.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/0c0b3406a70c47fe89fc3662a3a3c016 : SUCCESS in 5m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/21308aac679a49c89a19aed351b5c9d1 : SUCCESS in 49m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ad5a537431634b60adc40271423222c5 : FAILURE in 58m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/84718959659649afa768b73cb2639496 : SUCCESS in 38m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4649979350c4a6cbb39e94bd459e9d0 : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3eac3bf77654428d94df19e50bfa4696 : SUCCESS in 5m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ab971eb270fe4c588cc1bd08e9910d7b : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/630fcfe089b245af97feeb32544d87ff : SUCCESS in 5m 54s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/ff16bd32ad934f99b8ebed2deb573128 : FAILURE in 11m 24s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/11809a6d2778426d9292bbf56691ca7e : FAILURE in 8m 09s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/6388e9eca3d14203b6e8957d19a17b55 : FAILURE in 8m 22s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/f1bf815796444e9da57884abab6e61fc : SUCCESS in 4m 25s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/fb2d2bc059ea436a8b2c72b5b563701d : FAILURE in 56m 15s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d76b494e5e5c45c68dc93dbe3d2e6d0c : FAILURE in 56m 32s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/4215c24f238b4f9e80cd0ec65d37df8f : FAILURE in 57m 01s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/454cf3c09af24a36b33ac591822797fe : FAILURE in 54m 34s\n- glance-multistore-cinder-import-fips https://zuul.opendev.org/t/openstack/build/751fde622d4d4c4c92c94690ac37db75 : FAILURE in 1h 44m 57s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d283bdccd65d4d8ea1db03e4d8deaf87 : SUCCESS in 53m 09s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c32267b946b34bc7aaeb78611fae816b : FAILURE in 1h 00m 49s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e6de1a773f13461ea4879ae78a158063 : SUCCESS in 23m 39s","accounts_in_message":[],"_revision_number":3},{"id":"c5dc13525f5b0c9561e7b9f8af83ad3a3de7b832","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-02-15 05:42:40.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ecbf8bece799ddc6df45cda8451869b116dc91b7","revisions":{"5d04fa09eaa5f09f2d5a060de8a2cc51e2a7310b":{"kind":"REWORK","_number":1,"created":"2022-02-14 13:30:47.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/25/829025/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/25/829025/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/25/829025/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/25/829025/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/25/829025/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/25/829025/1"}}},"commit":{"parents":[{"commit":"4f4fc9b15d7c32eaa8798bde9372db587c6e79a3","subject":"Merge \"Remove six.assertRaisesRegex usage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4f4fc9b15d7c32eaa8798bde9372db587c6e79a3"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2022-02-10 13:33:06.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2022-02-14 13:20:49.000000000","tz":0},"subject":"Add new API to delete metadef resource type","message":"Add new API to delete metadef resource type\n\nThis change add new Delete API for metadef resource type with\nthe new policy rule ``delete_metadef_resource_type`` which\ndefaults to admin only.\n\nNOTE: Renamed all the resource type deassociation methods from\n``delete/remove`` to ``deassociate`` to avoid confusion.\n\nImplements: bluprint add-create-and-delete-api-for-metadef-resource-type\nChange-Id: I612121ff4edd00861baed546d8f18775f02860af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5d04fa09eaa5f09f2d5a060de8a2cc51e2a7310b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5d04fa09eaa5f09f2d5a060de8a2cc51e2a7310b"}]},"branch":"refs/heads/master"},"51cf13592e515c61b972cdbfce6cf404f26c6b04":{"kind":"REWORK","_number":2,"created":"2022-02-14 13:33:58.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/25/829025/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/25/829025/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/25/829025/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/25/829025/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/25/829025/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/25/829025/2"}}},"commit":{"parents":[{"commit":"4f4fc9b15d7c32eaa8798bde9372db587c6e79a3","subject":"Merge \"Remove six.assertRaisesRegex usage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4f4fc9b15d7c32eaa8798bde9372db587c6e79a3"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2022-02-10 13:33:06.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2022-02-14 13:33:58.000000000","tz":0},"subject":"Add new API to delete metadef resource type","message":"Add new API to delete metadef resource type\n\nThis change add new Delete API for metadef resource type with\nthe new policy rule ``delete_metadef_resource_type`` which\ndefaults to admin only.\n\nNOTE: Renamed all the resource type deassociation methods from\n``delete/remove`` to ``deassociate`` to avoid confusion.\n\nImplements: bluprint add-create-and-delete-api-for-metadef-resource-type\nChange-Id: I612121ff4edd00861baed546d8f18775f02860af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/51cf13592e515c61b972cdbfce6cf404f26c6b04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/51cf13592e515c61b972cdbfce6cf404f26c6b04"}]},"branch":"refs/heads/master"},"ecbf8bece799ddc6df45cda8451869b116dc91b7":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-02-14 13:36:10.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/25/829025/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/25/829025/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/25/829025/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/25/829025/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/25/829025/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/25/829025/3"}}},"commit":{"parents":[{"commit":"4f4fc9b15d7c32eaa8798bde9372db587c6e79a3","subject":"Merge \"Remove six.assertRaisesRegex usage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4f4fc9b15d7c32eaa8798bde9372db587c6e79a3"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2022-02-10 13:33:06.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2022-02-14 13:36:10.000000000","tz":0},"subject":"Add new API to delete metadef resource type","message":"Add new API to delete metadef resource type\n\nThis change add new Delete API for metadef resource type with\nthe new policy rule ``delete_metadef_resource_type`` which\ndefaults to admin only.\n\nNOTE: Renamed all the resource type deassociation methods from\n``delete/remove`` to ``deassociate`` since we have proper delete\nAPI instead of just deassociation.\n\nImplements: https://blueprints.launchpad.net/glance/+spec/add-create-and-delete-api-for-metadef-resource-type\nChange-Id: I612121ff4edd00861baed546d8f18775f02860af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ecbf8bece799ddc6df45cda8451869b116dc91b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ecbf8bece799ddc6df45cda8451869b116dc91b7"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
