)]}'
{"id":"openstack%2Fglance~831381","triplet_id":"openstack%2Fglance~master~Ic40f582f83e04c6915a3fcb231d6d95ca071c100","project":"openstack/glance","branch":"master","topic":"bug/1962581","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2022-03-02 01:03:52.000000000","reason":"Change was submitted"},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2022-03-02 01:03:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic40f582f83e04c6915a3fcb231d6d95ca071c100","subject":"Change default value for [wsgi]/python_interpreter","status":"MERGED","created":"2022-03-01 14:17:43.000000000","updated":"2022-03-02 10:09:58.000000000","submitted":"2022-03-02 01:03:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"831381-bug/1962581","meta_rev_id":"208ffe2900a33d1eef7e3aea565e7c5b738462ff","_number":831381,"virtual_id_number":831381,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-02 01:03:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":2,"date":"2022-03-01 16:54:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":1,"date":"2022-03-01 19:49:06.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":1,"date":"2022-03-01 22:19:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":1,"date":"2022-03-02 10:09:58.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-01 14:53:59.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2022-03-01 15:54:16.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2022-03-01 16:01:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-02 10:09:58.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"cd51749115afc37b1ee81efa0f07c4e350eb75f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-03-01 14:17:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"31973979ae77545e37521944dda298255eb51149","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-03-01 14:53:59.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3ecafce3342bbad2180c2f7f57c5849b17b532ea","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-03-01 15:41:39.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"a210a308769c4214d7ee14850f97df98252266ef","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-03-01 15:53:37.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f5e119b0340d90d000590fe7afef7e6e1eb29d48","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2022-03-01 15:54:16.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"065da6d357a5c41b72b01670640ef0cd923d6ace","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2022-03-01 15:58:34.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9d533813e30c77ee9940be7b60af72afcc67a94e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 16:01:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/613d903e11364f22826109ad58a8c7c3 : SUCCESS in 55m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/94e0c81dd46b4abdbfd98251befad190 : SUCCESS in 1h 16m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d230282a3fd241e7a32eb0364ea8b60c : SUCCESS in 48m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9f852422a5649919fad8e30bcca4aa7 : SUCCESS in 6m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/70c6ecc1daa7452e86c8d1c252d3ad57 : SUCCESS in 7m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f6f481bebde543688242747389c93ffc : SUCCESS in 7m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb4fdaa4e9e94ad0a4516ea6999d3fa2 : SUCCESS in 7m 39s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/40e2faf6765c4a709537232a4c827c66 : SUCCESS in 12m 22s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e82632d61fc942eebb1eb28787c26905 : SUCCESS in 10m 51s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/bf76912432c841bfb4b571c299bfdc0b : SUCCESS in 11m 30s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/71b475ef50294e73bb1c097bf726252b : SUCCESS in 6m 16s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/78cd1613c6774effaf1fa067d915ec57 : FAILURE in 26m 50s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/185a880e0d3d4e5e9b27b4d6ff538346 : SUCCESS in 1h 14m 18s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/541861f30c4d4d64aad4f679d1b7e1ba : SUCCESS in 1h 08m 25s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8084e9f303d6410baac294baf1b9dc12 : SUCCESS in 1h 35m 30s\n- glance-multistore-cinder-import-fips https://zuul.opendev.org/t/openstack/build/c23fe2d5a5284df99e9d3462e43cf657 : FAILURE in 1h 38m 20s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1019e8eb8e274820ae8250cb8d6fa704 : SUCCESS in 53m 00s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9c4fbdb6743c4af4a877f56dddd9696e : FAILURE in 32m 57s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/be102b944dc449259ea320440e8c9253 : SUCCESS in 34m 32s","accounts_in_message":[],"_revision_number":1},{"id":"db372ed3960069dc2026bed862153577dc5a8f4c","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-03-01 16:05:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9883786736cfe81e48822fad1a677621863b1dc0","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2022-03-01 16:05:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a2c5b2a7c4c15741eda0a7fe7c8ee1a020767650","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-03-01 16:28:20.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5c117f2dbac40b185ed00e165ddeccbb1aab1053","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-03-01 16:37:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"082061f1bf0a4f858a6b05ef97744b80e7d9a0fb","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2022-03-01 16:54:16.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"be3f06ab7e89d6e2aac3e0cd024d161e51cf5ba0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 18:35:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/d6d555efda7249bfa5247a654a703d1f : SUCCESS in 46m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/01ef53b6ca6b4b4193101a63c6631495 : SUCCESS in 1h 01m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5239a1ad3b50442dbe207d3723bcc076 : SUCCESS in 37m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39f7b9276dc0428a8c8caacbf516421c : SUCCESS in 6m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/12013d971bf0496699b6b1eb83dae901 : SUCCESS in 7m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/67b443377d4b49b5937cbef8fdba920a : SUCCESS in 8m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7569e1a0d2742e58ef5ea8361670e9f : SUCCESS in 7m 50s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/d554e78e0d4445939e72c560191746c5 : SUCCESS in 12m 08s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/58eb3506fa3c4f5f9718407cebeaaaee : SUCCESS in 10m 29s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/409369009a514b5c9971a1b5b17ed19b : SUCCESS in 10m 57s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/f2346001e58240e3acccf2ca03d3d453 : SUCCESS in 6m 47s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/65d3f04438194c7ab9bcce39d01fb06c : SUCCESS in 59m 16s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/28b7afa387204ddb985ba36db6e8c1cb : SUCCESS in 1h 03m 11s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f800d75f25cd4734a7362419c83123bf : SUCCESS in 1h 04m 39s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b7b807a438d240bcba45485b87ad1799 : SUCCESS in 1h 16m 41s\n- glance-multistore-cinder-import-fips https://zuul.opendev.org/t/openstack/build/6c0b6ca9c65f46e7b7c301642769bee1 : FAILURE in 1h 33m 25s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9fb348af204a4b4b8ebed5b9e02340a7 : SUCCESS in 45m 31s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ff0489a403cf4efea94f7391e05bba71 : SUCCESS in 1h 01m 50s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/79498278b8134b1c83157ec804c1cb96 : SUCCESS in 25m 17s","accounts_in_message":[],"_revision_number":2},{"id":"eba12412cb061f0df98ed24a14dd087a53a7d761","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-03-01 19:49:06.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0454c72a2e8f4b23fabf76a1cb6c49d320299e6b","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2022-03-01 22:19:52.000000000","message":"Patch Set 2: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c24c2c81d8ca0fce5769f8ef35ae5b7a1efc604c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 22:21:02.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"10b5b232b5dbe49bf60183fd9e48c13d87ad88e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 01:03:51.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/e11e2e57cff94831a83029bb2f45a9f6 : SUCCESS in 54m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0ec8fe36d6aa473caeae522781e3cbf9 : SUCCESS in 1h 10m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2a636259177a455eb60824e70761faab : SUCCESS in 48m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c29b0ff5f57843dc8bb19f86c8cf0044 : SUCCESS in 6m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/51c4c6cf334a4bc395252e723791d55a : SUCCESS in 6m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ea9a8599517343049fb77b2a7d00f7c0 : SUCCESS in 8m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c3160e2f5fa427590a1979839657124 : SUCCESS in 7m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bc9cfed7635446c8a5e74faf6b7635d8 : SUCCESS in 10m 56s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0862012ae46e4525a6a24c993b7d96c5 : SUCCESS in 11m 53s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/95da48cabf49459f959233fc45d8d35e : SUCCESS in 1h 07m 18s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/5d836331487f490285689106e7519eed : SUCCESS in 1h 07m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0e76fba29e1d484191e38d9b58c9a313 : SUCCESS in 49m 31s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ca31f6a3597d48bebd6db013cd994d38 : SUCCESS in 1h 23m 02s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5803a83e9f90422fa1966376c2099392 : SUCCESS in 27m 30s","accounts_in_message":[],"_revision_number":2},{"id":"4824df7880e1796673b2febaa3efd27254d74ec5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 01:03:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"0063aeff1959ed913e514635a204aa72f6c20a57","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 01:05:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa3d60004821405aa3c1693ac39f2a3d : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":2},{"id":"2cb275c12603b67877c8a670c83d3384feef26cb","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-03-02 10:09:58.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/831381/831381/2/check-not-vote/ext-glance-zuul/2e664c4 : FAILURE in 24s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"daa602baa900530fbd0a49ad67e7e1b5e0ae9343","revisions":{"20ddece5b64a8d2f091a8ca737f2f2efda5dfab6":{"kind":"REWORK","_number":1,"created":"2022-03-01 14:17:43.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/81/831381/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/81/831381/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/81/831381/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/81/831381/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/81/831381/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/81/831381/1"}}},"commit":{"parents":[{"commit":"f5776e6c1cee15e05a8618bda67ae5c7f85b32e4","subject":"Merge \"Fix overwriting of existing tags while creating new tags\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f5776e6c1cee15e05a8618bda67ae5c7f85b32e4"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-03-01 14:13:09.000000000","tz":-300},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-03-01 14:13:09.000000000","tz":-300},"subject":"Change default value for [wsgi]/python_interpreter","message":"Change default value for [wsgi]/python_interpreter\n\nThe default value is being set too early.  We don\u0027t want to use\nthe value of sys.executable at the time the config is generated,\nbecause that\u0027s unlikely to map to an existing interpreter in an\nactual deployment.\n\nChange-Id: Ic40f582f83e04c6915a3fcb231d6d95ca071c100\nCloses-bug: #1962581\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/20ddece5b64a8d2f091a8ca737f2f2efda5dfab6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/20ddece5b64a8d2f091a8ca737f2f2efda5dfab6"}]},"branch":"refs/heads/master"},"daa602baa900530fbd0a49ad67e7e1b5e0ae9343":{"kind":"REWORK","_number":2,"created":"2022-03-01 16:37:16.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/81/831381/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/81/831381/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/81/831381/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/81/831381/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/81/831381/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/81/831381/2"}}},"commit":{"parents":[{"commit":"f5776e6c1cee15e05a8618bda67ae5c7f85b32e4","subject":"Merge \"Fix overwriting of existing tags while creating new tags\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f5776e6c1cee15e05a8618bda67ae5c7f85b32e4"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-03-01 14:13:09.000000000","tz":-300},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2022-03-01 17:44:13.000000000","tz":-480},"subject":"Change default value for [wsgi]/python_interpreter","message":"Change default value for [wsgi]/python_interpreter\n\nThe default value is being set too early.  We don\u0027t want to use\nthe value of sys.executable at the time the config is generated,\nbecause that\u0027s unlikely to map to an existing interpreter in an\nactual deployment.\n\nChange-Id: Ic40f582f83e04c6915a3fcb231d6d95ca071c100\nCloses-bug: #1962581\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/daa602baa900530fbd0a49ad67e7e1b5e0ae9343"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/daa602baa900530fbd0a49ad67e7e1b5e0ae9343"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
