)]}'
{"id":"openstack%2Fglance~831892","triplet_id":"openstack%2Fglance~master~Id8be1c2c3ce022e068b37729af072694737d8df0","project":"openstack/glance","branch":"master","topic":"zanata/translations","hashtags":[],"change_id":"Id8be1c2c3ce022e068b37729af072694737d8df0","subject":"Imported Translations from Zanata","status":"MERGED","created":"2022-03-04 02:35:44.000000000","updated":"2022-04-20 05:44:23.000000000","submitted":"2022-04-20 05:42:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"831892-zanata/translations","meta_rev_id":"f4e0f804669f08af4c58f060387ed55169773bd9","_number":831892,"virtual_id_number":831892,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-20 05:42:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2022-04-20 05:29:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2022-04-20 05:29:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-04 02:36:32.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-04 03:22:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-20 05:29:52.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"82fd674d5fa6d2486c4cbb9d66cde0df7c34ef98","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-03-04 02:35:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"09da9940e78a8d46371321a2f1b9c18e060b4a6f","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-03-04 02:36:32.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/831892/831892/1/check-not-vote/ext-glance-zuul/bbdccf3 : FAILURE in 23s","accounts_in_message":[],"_revision_number":1},{"id":"2ffe472315dd3fa0eadcc29d56391528f5b28812","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-04 03:22:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb8bf314b2b5482caeac9c82cac8753d : SUCCESS in 8m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b2ca593b2c9c4704bdc2596cdb2dd6a2 : SUCCESS in 8m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/07a5bb9740ce460387f22dd904ce805a : SUCCESS in 9m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a848a012c9748d38860bc71ff159959 : SUCCESS in 9m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/89a5ede1fc474a7199575780aa86e11c : SUCCESS in 10m 15s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/86fd32670af5479cb9d3d82ad6fa0d82 : SUCCESS in 12m 31s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/946ea7ce6e5141d8b162e4ac98d2abff : SUCCESS in 14m 08s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/952e0abc667c4394b4d72677fbe7e55b : SUCCESS in 12m 04s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/f1869f172f6d4d2294ff2fcc0074b253 : SUCCESS in 10m 05s","accounts_in_message":[],"_revision_number":1},{"id":"669dbfb011eb050091ea6205f19da4517fa4781d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-03-10 02:44:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f9115d9382e080a47e87dc594598ac34a9c117a5","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-03-10 02:45:12.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/831892/831892/2/check-not-vote/ext-glance-zuul/56f65f5 : FAILURE in 20s","accounts_in_message":[],"_revision_number":2},{"id":"e60e791b79c155dbd5c65b2cbcdf681d428319c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 03:08:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4c4f7452d684780adabe44dca6aa4d2 : SUCCESS in 8m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7b1b9a0cf7574f9a88bd85639f4d00f6 : SUCCESS in 9m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e90f3a52ecec4db989aa004debff0308 : SUCCESS in 11m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a32324a668c4aaaadd65f977acb9b6b : SUCCESS in 9m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/52c917efcb314ed7a06be5d8ae21beb6 : SUCCESS in 9m 30s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/a4c24e3e3617414ba7e23b292f4c739a : SUCCESS in 15m 03s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5c27676567db4f1f80dbe7126e7733a8 : SUCCESS in 12m 58s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/c28acc0ed5fe4e6aadefdafb3d82ff14 : SUCCESS in 12m 58s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/d97c59675c454d449165d94d53ae1b15 : SUCCESS in 8m 40s","accounts_in_message":[],"_revision_number":2},{"id":"03c0d8342c2b80b8e684acd0920386d8917cd604","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-03-15 02:38:09.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0b199087a93ba5e321086feb0285ddd38c010ecf","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-03-15 02:38:56.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/831892/831892/3/check-not-vote/ext-glance-zuul/de98656 : FAILURE in 22s","accounts_in_message":[],"_revision_number":3},{"id":"e6da90b672404e6de904515088b4e34bcda8d48e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 03:21:35.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0fc9b155ab24fad9fc7c056cad9d9a1 : SUCCESS in 7m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/06cf43e34e9746639c7b7fca3c4fc81f : SUCCESS in 7m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cdf870965bfc4d8e94ddfaf8ba2c0ff4 : SUCCESS in 7m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b4ec1de28c84a11a7b793c87b47e5c7 : SUCCESS in 8m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a3cdec423b10463792dc905a68ae90b0 : SUCCESS in 7m 51s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/cbcd03479257412c9b54f1b08f1588be : SUCCESS in 13m 48s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3929932111f54618ad160ea2ec2a79b5 : SUCCESS in 11m 13s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/6951607e0edd492688d18e82c60ab5ae : SUCCESS in 14m 09s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/11c63a79106b44bf8cbe91e8e2cae370 : SUCCESS in 7m 43s","accounts_in_message":[],"_revision_number":3},{"id":"e899dadd3be9265642185e96777de0df9f882564","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-04-13 02:38:06.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"aa9412ec4806fca9fdb662bbdb08663c66b17d82","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-04-13 02:38:57.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/831892/831892/4/check-not-vote/ext-glance-zuul/61f7e4b : FAILURE in 17s","accounts_in_message":[],"_revision_number":4},{"id":"15674e53edb365219c81b099dff342b89004fb25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-13 03:21:02.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91b55585e94a4ab1a53fa01e303cc621 : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aad1e1252dcf4371b9b0f4c127755299 : SUCCESS in 7m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/68249fb453294259b045dd378d6d3eb7 : SUCCESS in 7m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42de2e7fbca8481aae5c0146e92ef9b9 : SUCCESS in 7m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7dffbfb9f9e24ed28eef54cf8190891d : SUCCESS in 7m 09s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/ccf4d55552b649fb8557c11ac5dcc8b7 : SUCCESS in 12m 00s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/54b832177af8464490c7f3b447850381 : SUCCESS in 11m 17s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/cb5a89f624034bb383a6e3ec77576f5f : SUCCESS in 10m 55s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/26a2f1ed555e4de6b287e06526e0731c : SUCCESS in 6m 11s","accounts_in_message":[],"_revision_number":4},{"id":"729d8845b7216a0779a64dda1cf58cdc439b502c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-04-15 02:31:16.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"a6018eea7dfbd1ed02322635d6c5c5eebd00de27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-15 03:25:58.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/579cef53ee5e40b993a8f12954f285f5 : SUCCESS in 6m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/71e67aa81a1e47b2b41de2d4fe263a3b : SUCCESS in 6m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aa73d2b1f2e74c00a43edb3b044eced9 : SUCCESS in 7m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53c9aca108e04845ba7be6f02876d28f : SUCCESS in 7m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5b0baf1177994abf998d53e341a083c7 : SUCCESS in 7m 05s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/edec638e2ab242f1b26fe19285d51768 : SUCCESS in 10m 49s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/512c0d2573174e9c8efc4258aaa7568b : FAILURE in 14m 06s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/9d850e0fddae4f6c8b6cfd911ffd7620 : SUCCESS in 10m 39s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/8d83cf5ffe9040e581d36fd0dbf7a5fb : SUCCESS in 6m 37s","accounts_in_message":[],"_revision_number":5},{"id":"71a383b35b2638b2398fd99990a229ff9c006649","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-04-18 04:47:37.000000000","message":"Patch Set 5:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/831892/831892/5/check-not-vote/ext-glance-zuul/062c612 : FAILURE in 17s","accounts_in_message":[],"_revision_number":5},{"id":"621311179e2849f80cca0481bf21b81fbd99cd3c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-04-20 02:36:02.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"1b7e2e85355d26643e996fdae4c1c1f069b67d2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 03:32:58.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9b5d9366dbf473ba1e994b439151206 : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/13882bb804754fa19a2cd2d14542e68c : SUCCESS in 7m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/816a40c5a9e04bfa9874246bdee018cc : SUCCESS in 7m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecb177c90f93467eb4d50483d8bda566 : SUCCESS in 8m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d2e262dfffe34609ad8621cebbb2af19 : SUCCESS in 6m 22s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/6ef86b386d264eeaa397f4e4b92f9188 : SUCCESS in 12m 29s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ea9872a94d39422b91e6c011ebd81225 : SUCCESS in 10m 39s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/4c23c20078884c68a916ab6ded25c514 : SUCCESS in 10m 04s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/bc80f46bab1c4506a180005294a3cf78 : SUCCESS in 5m 40s","accounts_in_message":[],"_revision_number":6},{"id":"a3a23c4610a516b2bf445810f994db04d2becec7","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-04-20 05:29:52.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"1e38e6ffbd6dfabe156cac955d1c618a28dad04e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 05:30:29.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"2b0873788719215762391a9d9990c0448b8388cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 05:42:42.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0c8a9f81d0c440fb1a2d9908681c8c6 : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eaa4a6b928b24553886727e1b74ee45e : SUCCESS in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b697e68c4f024f46995548f140475c9e : SUCCESS in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a66cb925d9014ddda441f902369f1fd8 : SUCCESS in 6m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c5db83ede0a84ddda4ca80f9933ca17b : SUCCESS in 5m 40s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0860157c584c4cd8a43ace5072691724 : SUCCESS in 9m 43s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e1f094f3d6cb4526ab0719f740c5d922 : SUCCESS in 9m 31s","accounts_in_message":[],"_revision_number":6},{"id":"3bc16f08e59617cd7763b31a12d6d03549886454","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 05:42:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"f4e0f804669f08af4c58f060387ed55169773bd9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 05:44:23.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7339fc073c8d4a41b1968008a919054a : SUCCESS in 1m 15s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c4e81df625c248c3a54d2ba2ac46c847 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"4abfa2573c5694dbc8a5bd7dab093127959bca78","revisions":{"605d210dff826e07d3338e0d69a3b7c8951f5b5f":{"kind":"REWORK","_number":1,"created":"2022-03-04 02:35:44.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/92/831892/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/92/831892/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/831892/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/831892/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/831892/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/92/831892/1"}}},"commit":{"parents":[{"commit":"7af4d971fc87b9d8ce31f8a79bae5bffa65e64a0","subject":"Merge \"Fix warning in the release notes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7af4d971fc87b9d8ce31f8a79bae5bffa65e64a0"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-03-04 02:35:42.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-03-04 02:35:42.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: Id8be1c2c3ce022e068b37729af072694737d8df0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/605d210dff826e07d3338e0d69a3b7c8951f5b5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/605d210dff826e07d3338e0d69a3b7c8951f5b5f"}]},"branch":"refs/heads/master"},"c24c69dc7ab119ee94cfb91fa74a1cd84fe48622":{"kind":"REWORK","_number":2,"created":"2022-03-10 02:44:26.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/92/831892/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/92/831892/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/831892/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/831892/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/831892/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/92/831892/2"}}},"commit":{"parents":[{"commit":"c99c3af4ff6a642417846852d5b7d89f9607782b","subject":"Merge \"Document new properties used by Ironic\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c99c3af4ff6a642417846852d5b7d89f9607782b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-03-10 02:44:24.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-03-10 02:44:24.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: Id8be1c2c3ce022e068b37729af072694737d8df0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c24c69dc7ab119ee94cfb91fa74a1cd84fe48622"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c24c69dc7ab119ee94cfb91fa74a1cd84fe48622"}]},"branch":"refs/heads/master"},"43b737c8317e6ac56924db7a7671eff3c150e8fc":{"kind":"REWORK","_number":3,"created":"2022-03-15 02:38:09.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/92/831892/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/92/831892/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/831892/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/831892/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/831892/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/92/831892/3"}}},"commit":{"parents":[{"commit":"c99c3af4ff6a642417846852d5b7d89f9607782b","subject":"Merge \"Document new properties used by Ironic\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c99c3af4ff6a642417846852d5b7d89f9607782b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-03-15 02:38:07.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-03-15 02:38:07.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: Id8be1c2c3ce022e068b37729af072694737d8df0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/43b737c8317e6ac56924db7a7671eff3c150e8fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/43b737c8317e6ac56924db7a7671eff3c150e8fc"}]},"branch":"refs/heads/master"},"d9258c7d8c933d59e17ec7b3e4e49917c2fd7697":{"kind":"REWORK","_number":4,"created":"2022-04-13 02:38:06.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/92/831892/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/92/831892/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/831892/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/831892/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/831892/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/92/831892/4"}}},"commit":{"parents":[{"commit":"3949fc467cccc8396bf92c35dcebe5be84e11281","subject":"api-ref needs update about \u0027checksum\u0027 image prop","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3949fc467cccc8396bf92c35dcebe5be84e11281"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-04-13 02:38:04.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-04-13 02:38:04.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: Id8be1c2c3ce022e068b37729af072694737d8df0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d9258c7d8c933d59e17ec7b3e4e49917c2fd7697"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d9258c7d8c933d59e17ec7b3e4e49917c2fd7697"}]},"branch":"refs/heads/master"},"fc242cc9b81340b2c16770741fe9e618557141cd":{"kind":"REWORK","_number":5,"created":"2022-04-15 02:31:16.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/92/831892/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/92/831892/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/831892/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/831892/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/831892/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/92/831892/5"}}},"commit":{"parents":[{"commit":"e8e7d66cf0ac6302fc9a9833c88e15c01d1c9ff5","subject":"Merge \"Update http with https\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e8e7d66cf0ac6302fc9a9833c88e15c01d1c9ff5"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-04-15 02:31:15.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-04-15 02:31:15.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: Id8be1c2c3ce022e068b37729af072694737d8df0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fc242cc9b81340b2c16770741fe9e618557141cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fc242cc9b81340b2c16770741fe9e618557141cd"}]},"branch":"refs/heads/master"},"4abfa2573c5694dbc8a5bd7dab093127959bca78":{"kind":"REWORK","_number":6,"created":"2022-04-20 02:36:02.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/92/831892/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/92/831892/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/831892/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/831892/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/831892/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/92/831892/6"}}},"commit":{"parents":[{"commit":"4fb6fbea7996c5852220f1f71bc3d1df3cfc6e24","subject":"Merge \"Move FIPS jobs to experimental and periodic queue\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4fb6fbea7996c5852220f1f71bc3d1df3cfc6e24"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-04-20 02:36:00.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-04-20 02:36:00.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: Id8be1c2c3ce022e068b37729af072694737d8df0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4abfa2573c5694dbc8a5bd7dab093127959bca78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4abfa2573c5694dbc8a5bd7dab093127959bca78"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
