)]}'
{"id":"openstack%2Fglance~833622","triplet_id":"openstack%2Fglance~master~I945f30fa7dd60e2e93bc248be4529633fb1f6bea","project":"openstack/glance","branch":"master","topic":"fips-compatibility-cs9","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2022-03-29 19:35:39.000000000","reason":"Brian Rosmaita replied on the change"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2022-04-19 18:21:54.000000000","reason":"Change was submitted"},"34120":{"account":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"last_update":"2022-04-19 18:21:54.000000000","reason":"Change was submitted"},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2022-04-19 17:32:34.000000000","reason":"Dan Smith replied on the change"}},"hashtags":[],"change_id":"I945f30fa7dd60e2e93bc248be4529633fb1f6bea","subject":"Move FIPS jobs to experimental and periodic queue","status":"MERGED","created":"2022-03-14 12:56:34.000000000","updated":"2022-04-19 18:24:35.000000000","submitted":"2022-04-19 18:21:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"833622-fips-compatibility-cs9","meta_rev_id":"98794e779f7004ae5499fd3379e475833b54e2d9","_number":833622,"virtual_id_number":833622,"owner":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-19 18:21:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2022-03-29 19:35:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2022-04-19 17:32:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2022-04-19 17:32:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-14 12:57:30.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-14 14:07:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-22 15:04:02.000000000","updated_by":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"reviewer":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"state":"CC"},{"updated":"2022-03-22 15:04:51.000000000","updated_by":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2022-03-22 15:04:51.000000000","updated_by":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2022-03-22 15:05:05.000000000","updated_by":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"}],"messages":[{"id":"75f03abca977796fced82c64c3f0b02c35f26f4b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2022-03-14 12:56:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3ad547aa0f04c85a8ea9b3109a21e01ca12f5120","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-03-14 12:57:30.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/833622/833622/1/check-not-vote/ext-glance-zuul/b59236a : FAILURE in 23s","accounts_in_message":[],"_revision_number":1},{"id":"9a741ce4d1ef8e466544abddb4fcce0c1cdf6c13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-14 14:07:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/7901221022f6480fa5a7fb46c538e3bc : SUCCESS in 1h 05m 05s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/db2e761d4c4c475aae252237d410643a : SUCCESS in 52m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8120fd1f32f4827b853c6041b3db5a1 : SUCCESS in 6m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/56fb319763b84bd7981395013798991c : SUCCESS in 7m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a9f5db78f3754e21adcbe4392af1fef4 : SUCCESS in 7m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aef52b3299a044beaecc199356ffb4d7 : SUCCESS in 8m 03s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/c2647a8d2e88491491639ab700e69a0e : SUCCESS in 13m 01s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2b9a42e04ec64d9ebd6ec2abbc4c566b : SUCCESS in 10m 14s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/cf31ba5e261b41ed88f530ab461066d7 : SUCCESS in 10m 49s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/249a314f6bff4e6dad3768c35baee086 : SUCCESS in 37m 19s","accounts_in_message":[],"_revision_number":1},{"id":"3ef7d8e2a5b0902d6128c32b3c814a30a1875bc3","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2022-03-14 16:09:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"31467748a1724156a840aae2d1ae7c32f799d598","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2022-03-14 16:16:50.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"6b011a3a503d2675897ca4fe3e38f9f713b67a70","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2022-03-14 16:17:31.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"575a41236514519ce9bc9a0f01af29601217ff0d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-03-14 16:17:40.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/833622/833622/2/check-not-vote/ext-glance-zuul/3372cb5 : FAILURE in 21s","accounts_in_message":[],"_revision_number":2},{"id":"9dd5c499487d567e73bcbe70dc420f9bf8ab0066","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-14 16:18:00.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"9a4c51f554f373a6a8dafab389456dec24539f17","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2022-03-14 16:18:22.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"09c77894aca1285cd92281778c19fe80fb540be5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2022-03-14 17:02:34.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"882aeef30874d99db1cf00d39ecd882f4a5152e0","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-03-14 17:03:19.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/833622/833622/4/check-not-vote/ext-glance-zuul/08bb094 : FAILURE in 19s","accounts_in_message":[],"_revision_number":4},{"id":"5c9ba647a1ac68527509194ea38563276538c8fc","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2022-03-14 17:03:21.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"444c28a2a76890581f043c9e9774426c309f582f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-14 18:04:35.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/0ce6d3c661184808b79973436b1fe940 : SUCCESS in 51m 06s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/149ada7ba00c4ebf92d3f5751862c832 : POST_FAILURE in 53m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13ebd99b80a34136ac3a135c6cc3b8ac : SUCCESS in 7m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/46cb9f4f33af45df9e29eb5ac2b0c311 : SUCCESS in 7m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c61cb662f832479f85c38791b229342e : SUCCESS in 7m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2225606fb39d472b981661c45b5b792a : SUCCESS in 8m 40s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/c60e1f2083924a1ea76d45648dabf7bb : SUCCESS in 12m 50s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2d34458dcf0e41a0a1e5819519079a44 : SUCCESS in 11m 01s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/7601c26c7e8b43d7b52e4702e89c4b9b : SUCCESS in 10m 37s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7fc41d85488548f7b985db8c33428cef : SUCCESS in 30m 14s","accounts_in_message":[],"_revision_number":4},{"id":"f02a0e31e110e7c46d53e9e15d14f55566fc534f","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-14 18:43:32.000000000","message":"Patch Set 4:\n\nBuild failed (experimental pipeline).\n\n- glance-tox-py39-glance_store-tips https://zuul.opendev.org/t/openstack/build/93a1eb643c444fd6a03119eba40a44bc : SUCCESS in 8m 45s\n- glance-tox-py36-glance_store-tips https://zuul.opendev.org/t/openstack/build/278837ebe6d84b69a91effbe07449c24 : SUCCESS in 8m 27s\n- glance-tox-functional-py39-glance_store-tips https://zuul.opendev.org/t/openstack/build/c0b6bbb520ad40a0972983361dfa441f : SUCCESS in 11m 46s\n- glance-tox-functional-py36-glance_store-tips https://zuul.opendev.org/t/openstack/build/e5b80c40d8b5462d9e0f33b26d06ec3c : SUCCESS in 11m 28s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/5fc5016235b04d35a2a2a73e5c2f8d7f : SUCCESS in 49m 31s\n- grenade-multinode https://zuul.opendev.org/t/openstack/build/42706b1754eb4c0995fb8f3446ae695a : SUCCESS in 1h 04m 19s\n- glance-multistore-cinder-import-fips https://zuul.opendev.org/t/openstack/build/0435d96ee6464b238d6bd5c3bcb41d3a : FAILURE in 1h 30m 57s","accounts_in_message":[],"_revision_number":4},{"id":"3bf1a057f51504fc9b34415f6a76e12965354683","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2022-03-22 15:04:02.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"19d9bda3173d191c32cb7de29d445b6d608966e0","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-03-22 15:04:44.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/833622/833622/4/check-not-vote/ext-glance-zuul/5170a28 : FAILURE in 16s","accounts_in_message":[],"_revision_number":4},{"id":"40f623b6e4c6708c33fa9519a3ef549acdd02ae2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-22 16:23:31.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/7d7b7137a1c24de18f93a5ba085e6785 : SUCCESS in 1h 10m 29s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7124c93633474e19ab702bce24b54e57 : SUCCESS in 59m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28754e098387479bb9864d634bde03e5 : SUCCESS in 6m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/54f31953b55a41b394c1ec55a136b8a6 : SUCCESS in 7m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b1ad3744806f42178b31ba6e481fa2e4 : SUCCESS in 7m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e14af2bf515c445390ba273170f60928 : SUCCESS in 8m 28s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/63173010b4ae45768adb1ca14fcd25bf : SUCCESS in 13m 18s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6dc52fd750f246afa77686256bafdb63 : SUCCESS in 10m 36s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/c9483b0e849e438195e854b3fa4d1ca9 : SUCCESS in 11m 39s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/544cd64a92d74a32b1f08ace7339c450 : SUCCESS in 37m 40s","accounts_in_message":[],"_revision_number":4},{"id":"393e11fb8365f9e0eaa2714c05b7befc8ee14a10","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2022-03-25 09:41:30.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6f0db05ccdc9478dd1c5e216f47e63e76dbc2180","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-25 11:07:29.000000000","message":"Patch Set 4:\n\nBuild failed (experimental pipeline).\n\n- glance-tox-py39-glance_store-tips https://zuul.opendev.org/t/openstack/build/007539d36ebc4c069637bbc4430ca5ae : SUCCESS in 8m 27s\n- glance-tox-py36-glance_store-tips https://zuul.opendev.org/t/openstack/build/f4788fa2e9944c02ac6adf4bf05166a4 : SUCCESS in 7m 50s\n- glance-tox-functional-py39-glance_store-tips https://zuul.opendev.org/t/openstack/build/2a916a2fb5114e48814ed781ff0e7d9d : SUCCESS in 12m 03s\n- glance-tox-functional-py36-glance_store-tips https://zuul.opendev.org/t/openstack/build/8180408b1eec419b935df9bd7c76e59f : SUCCESS in 11m 19s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/f080200791964c27bbd9c618be0ad8f4 : SUCCESS in 47m 41s\n- grenade-multinode https://zuul.opendev.org/t/openstack/build/a68bce4cfed24751925cb6278ca4490e : SUCCESS in 1h 08m 42s\n- glance-multistore-cinder-import-fips https://zuul.opendev.org/t/openstack/build/b69db72f76004ea4a56837b8f135b822 : FAILURE in 1h 23m 02s","accounts_in_message":[],"_revision_number":4},{"id":"8fb0b2a167918556b58cf5cf690b8f5c59a8545a","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-03-29 19:35:39.000000000","message":"Patch Set 4: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"b6130510060ed7c46f169ab9d62d1ae5b28069d9","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-04-19 17:32:34.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"6b17fa8f18f8f60ee58c051598109fd56f466b6b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 17:33:51.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"7c9aa097de96d70c70ca34ca6aa0f35b67fddcd0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 18:21:54.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3b5fb6f8b3e84ac499cc5f1ff670016c : SUCCESS in 46m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/201ee985474e4745a577e029b7825429 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/426f1e114a964c5d9bed35372dab05c9 : SUCCESS in 5m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eb48c427fafa479fb2ac0163d4b6b743 : SUCCESS in 6m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb9fa9fbdcaf4db3ab2fc007924221ad : SUCCESS in 6m 50s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/da548aaaf2cb4877b37712f1589126d6 : SUCCESS in 7m 49s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f5f3f20bc7d347fc81b9be8791c9ae55 : SUCCESS in 10m 40s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/96456e41138e45d0930475753dc3d3f9 : SUCCESS in 27m 18s","accounts_in_message":[],"_revision_number":4},{"id":"7f629f366ae65a12122131eb64bee602386c4faf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 18:21:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"98794e779f7004ae5499fd3379e475833b54e2d9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 18:24:35.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/801d20c42ebe44348962c7737f3fb3ca : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"8c6a79891a31d3fc88496500ebb7fd2cd5ee2273","revisions":{"51a8acfc397cf6a456fe86d8ca9b9c73534fde51":{"kind":"REWORK","_number":1,"created":"2022-03-14 12:56:34.000000000","uploader":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"ref":"refs/changes/22/833622/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/22/833622/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/22/833622/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/22/833622/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/22/833622/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/22/833622/1"}}},"commit":{"parents":[{"commit":"cb7972b70f0f4b988c28cd5bd0c060729f641d14","subject":"Change test fips job from centos8 to centos-9","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cb7972b70f0f4b988c28cd5bd0c060729f641d14"}]}],"author":{"name":"afariasa","email":"afariasa@redhat.com","date":"2022-03-14 12:56:22.000000000","tz":0},"committer":{"name":"afariasa","email":"afariasa@redhat.com","date":"2022-03-14 12:57:14.000000000","tz":0},"subject":"Move FIPS jobs to experimental and periodic queue","message":"Move FIPS jobs to experimental and periodic queue\n\nChange-Id: I945f30fa7dd60e2e93bc248be4529633fb1f6bea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/51a8acfc397cf6a456fe86d8ca9b9c73534fde51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/51a8acfc397cf6a456fe86d8ca9b9c73534fde51"}]},"branch":"refs/heads/master"},"e3ea9b4ca9f38a00bf335d419c90fd08f0d54db0":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-03-14 16:16:50.000000000","uploader":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"ref":"refs/changes/22/833622/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/22/833622/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/22/833622/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/22/833622/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/22/833622/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/22/833622/2"}}},"commit":{"parents":[{"commit":"c99c3af4ff6a642417846852d5b7d89f9607782b","subject":"Merge \"Document new properties used by Ironic\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c99c3af4ff6a642417846852d5b7d89f9607782b"}]}],"author":{"name":"afariasa","email":"afariasa@redhat.com","date":"2022-03-14 12:56:22.000000000","tz":0},"committer":{"name":"Andre Aranha","email":"afariasa@redhat.com","date":"2022-03-14 16:16:50.000000000","tz":0},"subject":"Move FIPS jobs to experimental and periodic queue","message":"Move FIPS jobs to experimental and periodic queue\n\nChange-Id: I945f30fa7dd60e2e93bc248be4529633fb1f6bea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e3ea9b4ca9f38a00bf335d419c90fd08f0d54db0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e3ea9b4ca9f38a00bf335d419c90fd08f0d54db0"}]},"branch":"refs/heads/master","description":"Rebase"},"1436a831b3804781262e57a41e89650d719a2dd5":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-03-14 16:17:31.000000000","uploader":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"ref":"refs/changes/22/833622/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/22/833622/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/22/833622/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/22/833622/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/22/833622/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/22/833622/3"}}},"commit":{"parents":[{"commit":"67fad0d9d23ea1cf9b9885eee2f3074d43b18f13","subject":"Change test fips job from centos8 to centos-9","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/67fad0d9d23ea1cf9b9885eee2f3074d43b18f13"}]}],"author":{"name":"afariasa","email":"afariasa@redhat.com","date":"2022-03-14 12:56:22.000000000","tz":0},"committer":{"name":"Andre Aranha","email":"afariasa@redhat.com","date":"2022-03-14 16:17:31.000000000","tz":0},"subject":"Move FIPS jobs to experimental and periodic queue","message":"Move FIPS jobs to experimental and periodic queue\n\nChange-Id: I945f30fa7dd60e2e93bc248be4529633fb1f6bea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1436a831b3804781262e57a41e89650d719a2dd5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1436a831b3804781262e57a41e89650d719a2dd5"}]},"branch":"refs/heads/master","description":"Rebase"},"8c6a79891a31d3fc88496500ebb7fd2cd5ee2273":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-03-14 17:02:34.000000000","uploader":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"ref":"refs/changes/22/833622/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/22/833622/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/22/833622/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/22/833622/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/22/833622/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/22/833622/4"}}},"commit":{"parents":[{"commit":"c99c3af4ff6a642417846852d5b7d89f9607782b","subject":"Merge \"Document new properties used by Ironic\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c99c3af4ff6a642417846852d5b7d89f9607782b"}]}],"author":{"name":"afariasa","email":"afariasa@redhat.com","date":"2022-03-14 12:56:22.000000000","tz":0},"committer":{"name":"afariasa","email":"afariasa@redhat.com","date":"2022-03-14 17:03:21.000000000","tz":0},"subject":"Move FIPS jobs to experimental and periodic queue","message":"Move FIPS jobs to experimental and periodic queue\n\nChange-Id: I945f30fa7dd60e2e93bc248be4529633fb1f6bea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8c6a79891a31d3fc88496500ebb7fd2cd5ee2273"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8c6a79891a31d3fc88496500ebb7fd2cd5ee2273"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
