)]}'
{"id":"openstack%2Fglance~835462","triplet_id":"openstack%2Fglance~master~I3e4b472c811bfa789f28ab3cb52a9bc610a4efee","project":"openstack/glance","branch":"master","topic":"default-image-properties","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2022-04-07 05:00:24.000000000","reason":"Abhishek Kekane replied on the change"},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2023-04-11 21:13:44.000000000","reason":"Change was abandoned"},"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2022-05-12 19:28:42.000000000","reason":"Tobias Urdin replied on the change"}},"hashtags":[],"change_id":"I3e4b472c811bfa789f28ab3cb52a9bc610a4efee","subject":"Add default image properties support","status":"ABANDONED","created":"2022-03-28 14:12:54.000000000","updated":"2023-04-11 21:13:44.000000000","total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5661a1bbf4f8a7f7aaeb9d0cda8dd3bc476f933b","_number":835462,"virtual_id_number":835462,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-03-28 23:55:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":-1,"date":"2022-04-06 16:51:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-28 15:23:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-06 16:51:48.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2022-04-06 18:38:37.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"670327257cfbd4f98eef967ea82852ff61654ba0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-03-28 14:12:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1c19324d8eb13e95bfad90d295eb2b7943f1913c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-28 15:23:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/2b2576bd98e74553abe4c88d43e5aa82 : FAILURE in 51m 46s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/3340b853f6314219af04ee5d324a143b : FAILURE in 1h 03m 55s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d28f005ed5d54eec9e81e5aa0d8b8c2a : FAILURE in 29m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f8ae3206eb7f4e379c56f2665a89d23c : POST_FAILURE in 32m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3da5db8c97eb41dc9b71a5bdfb55086e : SUCCESS in 6m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/339c20fd409d43debe614c4249ea6151 : FAILURE in 7m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f4822a8ce11d476bb91749abe0e54c4a : FAILURE in 7m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6552773a990470091316571db2b53a7 : FAILURE in 6m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dd35f1d8b23843db845d5e35b38fe7a0 : SUCCESS in 7m 33s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/f47c32ea72024a88a2ceabf678ef0176 : FAILURE in 10m 47s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b2b7e372055946a4a7e7068823ade884 : FAILURE in 8m 36s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/9cc28ce3b7c74bbd8e3cf86bb163e9e6 : FAILURE in 8m 04s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/60213232d601400d8a5e61cee723cdf2 : FAILURE in 6m 15s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/ff79c021ec464f10aafe1002effa8936 : FAILURE in 32m 32s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e5a46be03ee54815873ad88cc069a114 : POST_FAILURE in 29m 42s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f1343a3cf9e043abb43922b652a7cfe6 : POST_FAILURE in 30m 32s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d7f72ebbf40a4e4aab95369c63c83759 : POST_FAILURE in 31m 21s\n- glance-multistore-cinder-import-fips https://zuul.opendev.org/t/openstack/build/635b76ff8b194194b5e5a3f1d74299b4 : POST_FAILURE in 41m 21s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d1ce3e1afc1f4d758d98911a39fbb076 : FAILURE in 30m 22s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5e22788ab4e34552afd2291875e8ae21 : FAILURE in 31m 48s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/cbbedd50ec3749328d11f6c69b46d3ea : FAILURE in 30m 38s","accounts_in_message":[],"_revision_number":1},{"id":"6025805b40a43e3f39d18665c913a60f0dfb3e8b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-03-28 17:23:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6e4dabeed840acf540f9915d09460ce9f87fa0c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-28 19:33:23.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/fb140386e1444e5bac0909f613d39f0b : SUCCESS in 1h 00m 55s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/d149fca723f94f3c8f6e74a7ba4109c2 : SUCCESS in 1h 04m 19s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d4685fb6aaea48299b238ad1525615b9 : SUCCESS in 1h 19m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9151770ba7c9471cac51ccabdbece2fc : SUCCESS in 49m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/153ef7152bc64bd28d5add76886358d9 : SUCCESS in 6m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e7f43472185415e9dd18d56082576d3 : SUCCESS in 7m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5f9d114d767f4cec9b027d0635a64116 : SUCCESS in 8m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/124f22e5ccb6478aa49fef032c76cfd4 : SUCCESS in 8m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/96025bc7ef784078a7db2d650aadbe9f : SUCCESS in 8m 25s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/d7eefc1fd080497d8240ff54f90f65e5 : SUCCESS in 13m 17s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/73d15e3e19d3475cab94a1e606d7b508 : SUCCESS in 10m 58s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/999d5cc281f04518948ba37572181f6c : SUCCESS in 10m 47s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/74c6990c9cf34ff09d4f19cb13d522f1 : SUCCESS in 6m 38s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/a650e8382f224dea8f314420cadc3b5b : SUCCESS in 1h 15m 09s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/13cf6108b84e412da7c4d48145e9eb10 : SUCCESS in 1h 09m 42s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/36e9f3a2044048ebaa744025207c5633 : SUCCESS in 1h 18m 15s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c61c3b41215842d58353674aadc52d49 : FAILURE in 1h 20m 47s\n- glance-multistore-cinder-import-fips https://zuul.opendev.org/t/openstack/build/1f680539e43d4c43b53b8b1372f0f011 : SUCCESS in 2h 00m 49s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0643bf8ee9c2481c97d4d569d96930fd : SUCCESS in 52m 31s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/eb146a5951764e80b4dc6993b7d1e798 : SUCCESS in 1h 12m 37s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2f727932fee446a0a2baf9f2a588f6cc : SUCCESS in 36m 01s","accounts_in_message":[],"_revision_number":2},{"id":"bf3949afd611dde84397a8c8bede9109d6464145","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-03-28 19:46:36.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c397e6d070ddf82cdb7c4965df9476b51b165eaf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-28 23:55:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/494527b448de41eca3245d6fbf12a684 : SUCCESS in 58m 59s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/008816ff8d014823a7271319227dbd30 : SUCCESS in 1h 07m 42s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a9d21b2fc3f54b72a1536c28acec7408 : SUCCESS in 1h 16m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e5e5efef27504e6ea7e5430bf8ebf03a : SUCCESS in 53m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79969c467a5c4a13a452cf417634f5f4 : SUCCESS in 7m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0dfe9acc2d03427ea46b974f460521eb : SUCCESS in 7m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/78c77c329a094f1d98210f0815d5f2b3 : SUCCESS in 8m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/808709bfabe3417a9638270b1ea1f3f9 : SUCCESS in 8m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/15d7ea2461de44a2a88092b6bed41108 : SUCCESS in 7m 13s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/ca7988a5c25c4b12a026509b653deb77 : SUCCESS in 14m 30s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5046b6008d1e475693ab277975d7a856 : SUCCESS in 11m 33s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/31a6b336b8ee46a3b7085c4d4172f7f1 : SUCCESS in 11m 02s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/456457ae88d940daaf97d263e3496a72 : SUCCESS in 6m 44s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/435834e8f20c4b30b163ef66581ef1cb : SUCCESS in 1h 14m 34s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/48b0c65b7b754ece83a2cb218f4caed6 : SUCCESS in 1h 17m 27s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/8b42580c1db14863b74087f90fcd6b44 : SUCCESS in 1h 17m 56s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/48f396feafd74d3b9231c87d7fb0022c : SUCCESS in 1h 32m 34s\n- glance-multistore-cinder-import-fips https://zuul.opendev.org/t/openstack/build/e8ea55b063504814a244689c2f8e9989 : SUCCESS in 1h 46m 00s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2f3a3b4c8b15490fa71ec0ec3876b587 : SUCCESS in 56m 36s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c0959d9ffeec4d69b536b97041d466aa : SUCCESS in 1h 06m 34s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/74a27c7a0a0e41108183c98ef2df0c1f : SUCCESS in 38m 51s","accounts_in_message":[],"_revision_number":2},{"id":"ae9b290de1fc46d24f188e9308aaf0dcffde2427","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-04-06 16:51:48.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dbff770d3e5d666da65eab3157b3dce4653e94ed","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-04-06 18:38:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"63893d28bb6aa23732d74f2045a16b8ab2bfb470","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-04-07 05:00:24.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3ac727eb12cbd2cdbd64a3cb81a320d73e08ea6b","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-05-12 19:28:42.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5661a1bbf4f8a7f7aaeb9d0cda8dd3bc476f933b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-04-11 21:13:44.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"18082d7652afbd4cc710d0652448d189afdcb04e","revisions":{"59caf4ebb9123b75bdcb84e4daff81225440676e":{"kind":"REWORK","_number":1,"created":"2022-03-28 14:12:54.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/62/835462/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/62/835462/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/835462/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/835462/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/835462/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/62/835462/1"}}},"commit":{"parents":[{"commit":"c99c3af4ff6a642417846852d5b7d89f9607782b","subject":"Merge \"Document new properties used by Ironic\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c99c3af4ff6a642417846852d5b7d89f9607782b"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2022-03-28 13:53:16.000000000","tz":0},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2022-03-28 14:12:52.000000000","tz":0},"subject":"Add default image properties support","message":"Add default image properties support\n\nThis introduces a newer configuration option named\ndefault_additional_image_properties that contains\nkey-value pairs of default properties that operators\ndecide should always be added to images when created.\n\nThese default properties can be explicitly overwritten\nby the properties sent when creating new images.\n\nChange-Id: I3e4b472c811bfa789f28ab3cb52a9bc610a4efee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/59caf4ebb9123b75bdcb84e4daff81225440676e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/59caf4ebb9123b75bdcb84e4daff81225440676e"}]},"branch":"refs/heads/master"},"18082d7652afbd4cc710d0652448d189afdcb04e":{"kind":"REWORK","_number":2,"created":"2022-03-28 17:23:58.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/62/835462/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/62/835462/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/835462/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/835462/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/835462/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/62/835462/2"}}},"commit":{"parents":[{"commit":"c99c3af4ff6a642417846852d5b7d89f9607782b","subject":"Merge \"Document new properties used by Ironic\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c99c3af4ff6a642417846852d5b7d89f9607782b"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2022-03-28 13:53:16.000000000","tz":0},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2022-03-28 17:23:51.000000000","tz":0},"subject":"Add default image properties support","message":"Add default image properties support\n\nThis introduces a newer configuration option named\ndefault_additional_image_properties that contains\nkey-value pairs of default properties that operators\ndecide should always be added to images when created.\n\nThese default properties can be explicitly overwritten\nby the properties sent when creating new images.\n\nChange-Id: I3e4b472c811bfa789f28ab3cb52a9bc610a4efee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/18082d7652afbd4cc710d0652448d189afdcb04e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/18082d7652afbd4cc710d0652448d189afdcb04e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
