)]}'
{"id":"openstack%2Fglance~839491","triplet_id":"openstack%2Fglance~master~If1bf639af4619185d8af01136504c971a78d531f","project":"openstack/glance","branch":"master","topic":"DNM/policy-auth-removal","attention_set":{},"removed_from_attention_set":{"5202":{"account":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"last_update":"2022-07-07 13:39:44.000000000","reason":"Change was abandoned"},"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2022-07-07 13:39:44.000000000","reason":"Change was abandoned"},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2022-07-07 13:39:44.000000000","reason":"Change was abandoned"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2022-07-07 13:39:44.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"If1bf639af4619185d8af01136504c971a78d531f","subject":"DNM: Ensure we can remove auth and policy layer","status":"ABANDONED","created":"2022-04-27 08:18:39.000000000","updated":"2022-07-07 13:39:44.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b9cb148821a5dae8fa0413ab046abcfbabc1cdae","_number":839491,"virtual_id_number":839491,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-04-27 10:03:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-27 08:19:26.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2022-04-27 08:19:26.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2022-04-27 08:19:26.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2022-04-27 08:19:26.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2022-04-27 10:03:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9d7e965f68405281a0faa60388b605387ecc4569","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-04-27 08:18:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8431874e21ce15a1692ff28f189253b4958d58f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 10:03:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/2dd476b2a0eb42d3b4d2a386930f8926 : SUCCESS in 59m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e80bdb76ffc94652876a2b1a5fc6893a : SUCCESS in 1h 10m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c8cba32dfd5341bfb63df6d44d80661f : SUCCESS in 48m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6f2ee4ba6b646058e05f2d5234aec8f : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/35eb076b20db4400b5a6cd486e372e76 : SUCCESS in 6m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/80b5d7a87c5e40018300ce0ec2864b86 : SUCCESS in 6m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7185c5bb08a6448b853ea5da2445d224 : SUCCESS in 8m 06s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/24829ea566c2472e8857e988173deeab : SUCCESS in 10m 52s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/be033471dddc4d77913a46dd07d85792 : SUCCESS in 9m 25s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/81136d38ebe74910b17dc64759d45921 : SUCCESS in 11m 23s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/2b8b4c6fa5f94a17985aab0683edc465 : SUCCESS in 5m 21s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/66ff32eedcaf402cb044db5c9da9412d : SUCCESS in 1h 21m 21s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/78b220ead5d74f039b6a05e1a817e105 : SUCCESS in 1h 10m 51s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/5c807bbd252343b7bb00f5158cf796f1 : SUCCESS in 1h 17m 56s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7a892a9e100a4bb880415abf13848b4d : SUCCESS in 1h 41m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/964a126a65c9460f8070394ab9e2acc7 : SUCCESS in 48m 10s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6a55b3f7c76e4fd2be9cf3b071bc2d66 : SUCCESS in 1h 07m 22s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0b22469082ae4189a7087111f488a18b : SUCCESS in 36m 07s","accounts_in_message":[],"_revision_number":1},{"id":"b9cb148821a5dae8fa0413ab046abcfbabc1cdae","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-07-07 13:39:44.000000000","message":"Abandoned\n\nPolicy layer is removed","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"88cbebd543cca85fc6d15a663e9e03e1a63c3162","revisions":{"88cbebd543cca85fc6d15a663e9e03e1a63c3162":{"kind":"REWORK","_number":1,"created":"2022-04-27 08:18:39.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/91/839491/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/91/839491/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/91/839491/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/91/839491/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/91/839491/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/91/839491/1"}}},"commit":{"parents":[{"commit":"14bd3eb1bb60dafc1ee33cf0e58b0d1e1b4cad67","subject":"Update migration constant","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/14bd3eb1bb60dafc1ee33cf0e58b0d1e1b4cad67"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2022-04-27 07:47:22.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2022-04-27 08:18:18.000000000","tz":0},"subject":"DNM: Ensure we can remove auth and policy layer","message":"DNM: Ensure we can remove auth and policy layer\n\nNOTE: Some of the code is still used from policy layer which we can move\nto a separate module or dedicated glance/api/v2/policy.py module.\n\nChange-Id: If1bf639af4619185d8af01136504c971a78d531f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/88cbebd543cca85fc6d15a663e9e03e1a63c3162"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/88cbebd543cca85fc6d15a663e9e03e1a63c3162"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
