)]}'
{"id":"openstack%2Fglance~839786","triplet_id":"openstack%2Fglance~master~Iae07e1c5abde09874928951ac55c2d0355bcd244","project":"openstack/glance","branch":"master","topic":"fix-docs-use-uc","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2022-05-06 13:37:47.000000000","reason":"Brian Rosmaita replied on the change"},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2022-05-06 15:43:36.000000000","reason":"Change was submitted"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2022-04-29 13:22:30.000000000","reason":"Abhishek Kekane replied on the change"}},"hashtags":[],"change_id":"Iae07e1c5abde09874928951ac55c2d0355bcd244","subject":"[CI] Add upper constraints to install command","status":"MERGED","created":"2022-04-28 16:03:14.000000000","updated":"2022-05-06 15:46:59.000000000","submitted":"2022-05-06 15:43:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"839786-fix-docs-use-uc","meta_rev_id":"a816eb535cc97950c0a55dcc27b8b541be69a23e","_number":839786,"virtual_id_number":839786,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-06 15:43:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2022-05-06 13:37:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2022-05-06 05:44:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-05-06 14:07:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2022-05-06 13:37:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-28 16:27:50.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2022-04-28 17:00:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-29 08:34:17.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2022-04-29 12:49:18.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2022-04-29 12:49:18.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2022-04-29 12:49:18.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2022-05-06 14:07:56.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"}],"messages":[{"id":"08e59e85dae0f685dfb01f6338476a68a12c2980","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-04-28 16:03:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"85d77b717e8511716ca8380614d300a43de3a72b","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-04-28 16:05:19.000000000","message":"Topic set to fix-docs-use-uc","accounts_in_message":[],"_revision_number":1},{"id":"4d2fe1bf0b2eb180b07dcc00f6c9d9e7cf000858","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-04-28 16:27:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f537a2f2b80addbe38635bba8dd9272760183f89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-28 17:00:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/50ecdff199354983876dab30024ce165 : SUCCESS in 4m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/74106d1ef7284aa1896ae71542aefcf3 : SUCCESS in 52m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82d33ba7af1842a7991f0619ab0be1fa : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a8b6df03b7af4c3d9488499bf8a9fd56 : SUCCESS in 5m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/47cef99277a64a218b8a778e7522a9a6 : SUCCESS in 6m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/755b93debafa47109060c9dcef5e2a68 : SUCCESS in 6m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bbb9bd95fd344609951b7cdb4409f71e : SUCCESS in 6m 15s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/c708a899163443c8ae284442e340d518 : SUCCESS in 11m 01s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0204c5878d05407ab3b06d01f48fd59f : SUCCESS in 8m 58s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/52a69f4371424e1e9637c22ab07c2ba1 : SUCCESS in 8m 41s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/410e60009c0d4aee9866540ae84d3eda : SUCCESS in 37m 13s","accounts_in_message":[],"_revision_number":1},{"id":"b7bcd062838d0ab67255566f658d1393b2597d32","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2022-04-29 08:34:17.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"65d16e8b7aee0e18e61c5896ee002bd4f0cdf210","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-04-29 12:49:18.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"216de7935e54d455ea4af7e4f19710b06f0d0c76","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-04-29 13:22:30.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2d4f729626c7f4e61a808525e8e5aa93a233c27d","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-04-29 13:40:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"89da55d06601482bd3c9668ed896b6b478e5f775","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-04-29 14:10:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4bcc34f14d1a2b82c1c4fb50d6d3f0d042f829c7","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-05-05 14:34:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c66478c02d738d3c778ad543abf684d34a348b29","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-05-05 15:18:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"58058a005e60b7e2ffecc87c549dcb397db2b1a9","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-05-05 15:24:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9b44fa13df855935ac33800f8b3c750d661ac146","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 16:20:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/5adf338f13c74589949742f84bf5ce67 : SUCCESS in 5m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/159b7f855e1f45389a2ed9f0beb54869 : SUCCESS in 56m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/141c038554dd48a3b3c69dcea9486612 : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc95b770ac12488c8d1244e332fc3d7d : SUCCESS in 8m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9ccb2a0031d144d59f847034c5efc194 : SUCCESS in 7m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/115f460bd883481a88e2a98d3025fa47 : SUCCESS in 6m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/662563bd82454a85ae4d562c09bf5c61 : SUCCESS in 6m 13s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/004365111b9b450592e453cf72f835fc : SUCCESS in 10m 59s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/409869d8656346e99f906a267e7de637 : SUCCESS in 9m 10s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/c266c4ed420d4c6f9f1bfdfb37faf839 : SUCCESS in 7m 25s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f351a883e634497e8e858c568ac8a998 : SUCCESS in 35m 14s","accounts_in_message":[],"_revision_number":2},{"id":"d6968d7e645a4411237bdf3719c06259149bb4c2","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-05-06 05:44:51.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3c15ec07cac927d608c0117294023393ec5dd603","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-05-06 13:37:47.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bc44c017f566c003737ff4e44bd88b93f4391247","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 13:40:07.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d328815d5d320afd92da18ce1a3147d4999754c4","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2022-05-06 14:07:56.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"9bf258a79c9519d9750e446295594bd86b851d14","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 15:43:36.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/fbdd5dc7639a4009833bd4075e225f6a : SUCCESS in 6m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d45b46d142494698901a69c1b01467d8 : SUCCESS in 55m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/418cfa2aba8c4958bbc22bdd2cfd4893 : SUCCESS in 6m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ad85adae6901497699cfa4b78413a0da : SUCCESS in 8m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7e3bc5a45f7740c1b231faf13fe66d76 : SUCCESS in 9m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85d2b2b3e0184f8a824a07ea167c3bba : SUCCESS in 7m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9006f37afed34b88843c41a279f33485 : SUCCESS in 6m 46s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3f5cd1098a5c47b9a92b348f86655885 : SUCCESS in 10m 46s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/985b57adbcce4b3f80eabaa82a96e86d : SUCCESS in 11m 15s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/710e6fed08b740c786096c2ccff421fc : SUCCESS in 40m 09s","accounts_in_message":[],"_revision_number":2},{"id":"5cdbab01daad086c017ce708adfc9c987a3ba8da","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 15:43:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"a816eb535cc97950c0a55dcc27b8b541be69a23e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 15:46:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/c1535856593e4860a36c61c77c5c45dc : SUCCESS in 55s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfdc9ff6ec174a3bbecd3f447386a634 : SUCCESS in 57s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6c33d5cd16d2489a84c83b73ce1974db : SUCCESS in 59s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"36aedba20e998953333225a9788c084d2e6be6ae","revisions":{"a46a43f063c9c6d5d60b7e53cbbe128bb7228997":{"kind":"REWORK","_number":1,"created":"2022-04-28 16:03:14.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/86/839786/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/86/839786/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/86/839786/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/86/839786/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/86/839786/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/86/839786/1"}}},"commit":{"parents":[{"commit":"14bd3eb1bb60dafc1ee33cf0e58b0d1e1b4cad67","subject":"Update migration constant","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/14bd3eb1bb60dafc1ee33cf0e58b0d1e1b4cad67"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-04-28 16:00:47.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-04-28 16:00:47.000000000","tz":120},"subject":"[CI] Install dependencies for docs target","message":"[CI] Install dependencies for docs target\n\nWhen tox \u0027docs\u0027 target is called, first it installs the dependencies\n(listed in \u0027deps\u0027) in \u0027installdeps\u0027 phase, then it installs glance\n(with its requirements) in \u0027develop-inst\u0027 phase. In the latter case\n\u0027deps\u0027 is not used so that the constraints defined in \u0027deps\u0027 are not\nused. This could lead to failures on stable branches when new packages\nare released that break the build. To avoid this, the simplest solution\nis to pre-install requirements, i.e. add requirements.txt to \u0027docs\u0027 tox\ntarget.\n\nChange-Id: Iae07e1c5abde09874928951ac55c2d0355bcd244\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a46a43f063c9c6d5d60b7e53cbbe128bb7228997"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a46a43f063c9c6d5d60b7e53cbbe128bb7228997"}]},"branch":"refs/heads/master"},"36aedba20e998953333225a9788c084d2e6be6ae":{"kind":"REWORK","_number":2,"created":"2022-05-05 15:18:35.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/86/839786/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/86/839786/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/86/839786/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/86/839786/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/86/839786/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/86/839786/2"}}},"commit":{"parents":[{"commit":"14bd3eb1bb60dafc1ee33cf0e58b0d1e1b4cad67","subject":"Update migration constant","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/14bd3eb1bb60dafc1ee33cf0e58b0d1e1b4cad67"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-04-28 16:00:47.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-05-05 15:16:57.000000000","tz":120},"subject":"[CI] Add upper constraints to install command","message":"[CI] Add upper constraints to install command\n\nIn cases when a package is installed in tox\u0027s develop-inst phase upper\nconstraints are not used if it were not defined in \u0027deps\u0027 of the tox\ntarget. This caused for example \u0027tox -e docs\u0027 to fail in case a package\nintroduced some branch incompatible change.\n\nThere are two option to fix this:\n* add the upper constraints to \u0027deps\u0027 of the target (and pre-install\n  requirements for some cases) OR\n* use it in the install_command\n\nIn the past the latter needed some extra attention (due to e.g. lower\nconstraints tests) and projects were encouraged to remove the\ninstall_command, but nowadays this seems better fitting for some\nproject and easier to use. So this patch fixes the upper constraints\nusage via the custom install_command and by removing it from \u0027deps\u0027.\n\nChange-Id: Iae07e1c5abde09874928951ac55c2d0355bcd244\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/36aedba20e998953333225a9788c084d2e6be6ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/36aedba20e998953333225a9788c084d2e6be6ae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
