)]}'
{"id":"openstack%2Fglance~839814","triplet_id":"openstack%2Fglance~stable%2Fyoga~Iae07e1c5abde09874928951ac55c2d0355bcd244","project":"openstack/glance","branch":"stable/yoga","topic":"fix-docs-use-uc","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2022-05-20 02:42:53.000000000","reason":"Brian Rosmaita replied on the change"},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2022-05-23 15:57:51.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iae07e1c5abde09874928951ac55c2d0355bcd244","subject":"[CI] Add upper constraints to install command","status":"MERGED","created":"2022-04-28 18:29:21.000000000","updated":"2022-05-23 15:59:37.000000000","submitted":"2022-05-23 15:57:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"839814-fix-docs-use-uc","meta_rev_id":"f23997b9401cc96391346b7ffcd0f82b84c1a55b","_number":839814,"virtual_id_number":839814,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-23 15:57:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2022-05-20 02:42:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2022-05-11 05:37:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-05-20 06:21:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2022-05-20 02:42:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-28 19:22:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-11 05:37:08.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2022-05-11 05:37:08.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2022-05-20 06:21:00.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"}],"messages":[{"id":"5f71899da67dac10dd0ac714be698324bc32929a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-04-28 18:29:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d7e8934c6996f42f6652a27256eb0020c43c968f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-28 19:22:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/8e84555b23e24cd1acae042b9c4a6854 : SUCCESS in 51m 21s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1587bb934ce64f4fbe2ec9ae3bc28d9a : SUCCESS in 43m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d42e23817f3401c8e75554bf6a64bbb : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5a121385d17e41fa9cc40ed0759ed6e5 : SUCCESS in 6m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8637f92aaea249f1ad1122754e6e2b2e : SUCCESS in 6m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc0966ba5a904216bf1455146eee2f6f : SUCCESS in 9m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ada4a551b4424ae7a5ba48ee23fa01b9 : SUCCESS in 6m 12s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/235ba25e8184408ab6a538498c6131c1 : SUCCESS in 11m 53s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4b6fead2e87641a2bc11d7212ac82947 : SUCCESS in 11m 35s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/92a5966a5275488780216da309adec60 : SUCCESS in 10m 24s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1207f79af10346f28ad0472522d05b6c : SUCCESS in 36m 06s","accounts_in_message":[],"_revision_number":1},{"id":"94a02e30f4388358bc8fda9482727ffa162de926","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-05-09 13:39:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fc5be1706e6e1d7509c102350983bf14c2e8c36c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-09 14:37:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/08ac13e3b4d6402b83dca4e4b5ad8579 : SUCCESS in 55m 32s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e431b607dc874499a6231edd834b250b : SUCCESS in 46m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0cbd2a6b7ca244598b895b97a9548e8e : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ab5da0dfc2624e2db0747005cae56def : SUCCESS in 5m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/531bbedcf1b84c79acd9894dbc64e188 : SUCCESS in 6m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed79756b4a744c8ea82f05f06a79ced0 : SUCCESS in 7m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/40dc8d47dafe471286bc895dddb62cbe : SUCCESS in 5m 54s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/1a8bf307bbd541109b72329c76dd7301 : SUCCESS in 11m 18s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6cfe6e1d898948829da7a2d3955f5814 : SUCCESS in 9m 15s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/ceff86a319394b77998ce6104336f360 : SUCCESS in 8m 39s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1eb1ea91eacc4a00b9cebfbf687e9b58 : SUCCESS in 28m 37s","accounts_in_message":[],"_revision_number":2},{"id":"4a40d311f442f4301469e904b83be20dd896a721","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-05-11 05:37:08.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"eb18a914d880196dc1f6028af0f70049faa1677a","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-05-20 02:42:53.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e7b986a287081d9bf20c2d75b847cc24d6334d7e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-20 02:43:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3fb6dd1f53ce43bfc15e1fc2d3d650576c81584e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-20 03:30:31.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/28d92f39f4df4a5f84a0948b247ef49b : SUCCESS in 44m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de9e53d524934edfa2ce20a5502abb69 : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e9b2621a2954a309a11a5bfb7826e22 : SUCCESS in 6m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/af44b309240a474bb273ea9fc88e0e68 : SUCCESS in 6m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c10a17ea6c92441db20db414c4f2b9d5 : SUCCESS in 6m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d3d26f1a9587472b8cbeda844648655c : SUCCESS in 7m 27s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cf6c3816591e490898cd30cb0163b2a4 : SUCCESS in 9m 58s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/28c87bad5ae3433ebd47bba3076085ed : SUCCESS in 8m 29s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ba69fa1e7f6647869d2a500cf91f1fb3 : FAILURE in 25m 47s","accounts_in_message":[],"_revision_number":2},{"id":"90431ba29dfd04e817ce64a98002fdb8b0ec7848","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2022-05-20 06:21:00.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c18d9dcd98d05017d875bc8d3f0ecdc52634e85e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-20 07:08:42.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/c7b06eb677e74c6b94795c93ebca81df : SUCCESS in 46m 27s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0e3c265447314cf8b5f4320c1684e9f5 : SUCCESS in 36m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/804ad9b58f484b6eb8a12f8cf95219e7 : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/abd06ffb1aaa46359fbbb649101ca2dd : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1dc1793c8ad44f8da61733eeddd99d40 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74dea45a5e3349eeaec7b860cf6c13df : SUCCESS in 6m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b4cc3ca1d61c4522a233dabb7019a78f : SUCCESS in 5m 39s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/be5d14f4bcc148019e9e64b8bcf5273d : SUCCESS in 8m 51s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d14d09c313464aea8cb32d88d8ffaaaa : SUCCESS in 7m 29s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/722d8b0d50554c72899083258c6652ea : SUCCESS in 7m 54s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/21c14face78e4fd58e564131455da023 : FAILURE in 23m 12s","accounts_in_message":[],"_revision_number":2},{"id":"2a7c1e3e7c11bf8225bf63fb8ee57acfff9fb384","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-05-23 14:14:13.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3163eafddcae911233807144a20ed5a139025e5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 15:09:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/521c2d94caef41a6a301169cc3b4149d : SUCCESS in 52m 46s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6f6fef6166f24815951cd82c3a9ff275 : SUCCESS in 46m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c2858e1929c42fb80f66f9805711f14 : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/48312a1a8e2a4a9b9e382d661933f742 : SUCCESS in 5m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/492f3d47bfd24f36aa66b003192fecb1 : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7476ce5fe9ee4fc7ba0f56fc170d7520 : SUCCESS in 6m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/12f95ce416a94688874ffadfbd8bffc3 : SUCCESS in 5m 24s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/c647e006762242ff944f8b7b0d13c59e : SUCCESS in 10m 33s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/865886895adc4c29aabc74c9c6648bc1 : SUCCESS in 8m 23s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/598c6def9a984ef6b8420ce5d67a55e8 : SUCCESS in 9m 30s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6a988e71ccd64b959f85690b3ffd7e02 : SUCCESS in 33m 11s","accounts_in_message":[],"_revision_number":2},{"id":"7bc64fd90360a9051fa0a460f08c4acef15b6995","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 15:10:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"15810e0adfad5cde2194f720a2b1a308c2141b30","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 15:57:51.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2247d8d95f6548c6a69e322ab8898b8b : SUCCESS in 46m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1b2ebaeb5354dc883ec3a121820f58c : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3ec52c61c1784ad3824a3d3e79139e17 : SUCCESS in 5m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5754abac392e4257b734579ee769b68e : SUCCESS in 5m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a519a6bb7a654c4b823cdd7e48d3df55 : SUCCESS in 6m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/44c8bd2dbc6c49cba228e9f113b32e30 : SUCCESS in 5m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5648c8709a4744d1ad04c0257c2b9a94 : SUCCESS in 10m 19s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7898c1a1a50840e9b03c985195646d74 : SUCCESS in 8m 43s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8bec419447844fc382400bf4a3f3a5f2 : SUCCESS in 35m 18s","accounts_in_message":[],"_revision_number":2},{"id":"4f649239a1b436c019c3f33d6d837f08b19e1eee","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 15:57:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"f23997b9401cc96391346b7ffcd0f82b84c1a55b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 15:59:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdc70637fb6147f1a12c75d47cb6582e : SUCCESS in 1m 19s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3aa915d43d5045d5bbbca7520c63a7fe : SUCCESS in 57s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"49d25474a2cfa1c91649241471c4b7b1b799fd75","revisions":{"d5213071d2ff368f808a18b36e1e64750887749c":{"kind":"REWORK","_number":1,"created":"2022-04-28 18:29:21.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/14/839814/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/14/839814/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/839814/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/839814/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/839814/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/14/839814/1"}}},"commit":{"parents":[{"commit":"1beee78e096ddba4729be10af6ac4ddec8e79da0","subject":"Update TOX_CONSTRAINTS_FILE for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1beee78e096ddba4729be10af6ac4ddec8e79da0"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-04-28 16:00:47.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-04-28 18:27:55.000000000","tz":120},"subject":"[CI] Install dependencies for docs target","message":"[CI] Install dependencies for docs target\n\nWhen tox \u0027docs\u0027 target is called, first it installs the dependencies\n(listed in \u0027deps\u0027) in \u0027installdeps\u0027 phase, then it installs glance\n(with its requirements) in \u0027develop-inst\u0027 phase. In the latter case\n\u0027deps\u0027 is not used so that the constraints defined in \u0027deps\u0027 are not\nused. This could lead to failures on stable branches when new packages\nare released that break the build. To avoid this, the simplest solution\nis to pre-install requirements, i.e. add requirements.txt to \u0027docs\u0027 tox\ntarget.\n\nConflicts:\n  tox.ini\n\nNOTE(elod.illes): conflict is due to branch specific upper constraints\nfile link.\n\nChange-Id: Iae07e1c5abde09874928951ac55c2d0355bcd244\n(cherry picked from commit a46a43f063c9c6d5d60b7e53cbbe128bb7228997)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d5213071d2ff368f808a18b36e1e64750887749c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d5213071d2ff368f808a18b36e1e64750887749c"}]},"branch":"refs/heads/stable/yoga"},"49d25474a2cfa1c91649241471c4b7b1b799fd75":{"kind":"REWORK","_number":2,"created":"2022-05-09 13:39:52.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/14/839814/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/14/839814/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/839814/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/839814/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/839814/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/14/839814/2"}}},"commit":{"parents":[{"commit":"1beee78e096ddba4729be10af6ac4ddec8e79da0","subject":"Update TOX_CONSTRAINTS_FILE for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1beee78e096ddba4729be10af6ac4ddec8e79da0"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-04-28 16:00:47.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-05-09 13:38:27.000000000","tz":120},"subject":"[CI] Add upper constraints to install command","message":"[CI] Add upper constraints to install command\n\nIn cases when a package is installed in tox\u0027s develop-inst phase upper\nconstraints are not used if it were not defined in \u0027deps\u0027 of the tox\ntarget. This caused for example \u0027tox -e docs\u0027 to fail in case a package\nintroduced some branch incompatible change.\n\nThere are two option to fix this:\n* add the upper constraints to \u0027deps\u0027 of the target (and pre-install\n  requirements for some cases) OR\n* use it in the install_command\n\nIn the past the latter needed some extra attention (due to e.g. lower\nconstraints tests) and projects were encouraged to remove the\ninstall_command, but nowadays this seems better fitting for some\nproject and easier to use. So this patch fixes the upper constraints\nusage via the custom install_command and by removing it from \u0027deps\u0027.\n\nConflicts:\n  tox.ini\n\nNOTE(elod.illes): conflict is due to branch specific upper constraints\nlink and variable name change (UPPER_* to TOX_*).\n\nChange-Id: Iae07e1c5abde09874928951ac55c2d0355bcd244\n(cherry picked from commit 36aedba20e998953333225a9788c084d2e6be6ae)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/49d25474a2cfa1c91649241471c4b7b1b799fd75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/49d25474a2cfa1c91649241471c4b7b1b799fd75"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
