)]}'
{"id":"openstack%2Fglance~839815","triplet_id":"openstack%2Fglance~stable%2Fxena~Iae07e1c5abde09874928951ac55c2d0355bcd244","project":"openstack/glance","branch":"stable/xena","topic":"fix-docs-use-uc","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2022-07-09 09:29:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iae07e1c5abde09874928951ac55c2d0355bcd244","subject":"[CI] Add upper constraints to install command","status":"MERGED","created":"2022-04-28 18:30:27.000000000","updated":"2022-07-09 09:31:03.000000000","submitted":"2022-07-09 09:29:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"839815-fix-docs-use-uc","meta_rev_id":"b3d6f54e1dba08afb12bcbe3594ab7b32a9dfad5","_number":839815,"virtual_id_number":839815,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-09 09:29:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"recommended":{"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},"all":[{"value":1,"date":"2022-07-07 15:12:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},{"value":2,"date":"2022-07-07 15:06:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2022-07-07 15:05:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},{"value":1,"date":"2022-07-07 15:06:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-28 19:47:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-07 15:05:43.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2022-07-07 15:06:07.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2022-07-07 15:12:32.000000000","updated_by":{"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},"reviewer":{"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},"state":"REVIEWER"}],"messages":[{"id":"20a18618ec24b14002444b1f4feeb66a14d78433","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-04-28 18:30:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a9d521a1affd01225ae917cedabb35a388d66a44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-28 19:47:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/db5d8c1018114c718d84f7209b444017 : SUCCESS in 1h 09m 23s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b080c8d86914741ae740efa072a43e6 : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6b26adea94d641cb99d176a2024c2b55 : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b72e38bd3ea9410fa67cec78b28983f4 : SUCCESS in 5m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7456df4b0a834f0fb6160f1fe3ea2056 : SUCCESS in 6m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94ca620ce634488f9254a1d94b62e1f0 : SUCCESS in 6m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2d870313d0564acab1d860cdcedcf51d : SUCCESS in 5m 32s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4d1ed4d2bc194123bdc97afe41b4a5ea : SUCCESS in 9m 49s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4c4fbf5ed8e74bbda208f81edf50d601 : SUCCESS in 11m 12s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/3ca0c76ba5204a799fec3cc13964f5d2 : SUCCESS in 8m 15s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2d57db469fb4469eaf3c59904f0a74d4 : SUCCESS in 25m 14s","accounts_in_message":[],"_revision_number":1},{"id":"6cdb88965cddd3ba1657336c3b7f3e7a4d8ca4e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-05-09 15:03:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"03619538de298cf7268968fa66dcb8db8f40b23f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-09 16:08:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/dc1f038c8ae2409ba6fdb4dfbb473b99 : SUCCESS in 1h 04m 37s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc5a4f725b01423a94b8df1f1b8dbc31 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d9712af0f76f4da59001288c4c539c76 : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e7a3990a7d2747fc88e5fd81796b50eb : SUCCESS in 5m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4e0f02e202db4bb6a05564050813cb1a : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f101c5eda4f249779ef6b07da2fc1a0c : SUCCESS in 6m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ae3beceb36344ebdb8a7742b4c384d77 : SUCCESS in 5m 40s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/40fa6b21b03c4fad99e2529a832a1f6e : SUCCESS in 9m 12s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ce3638a2adee4b41acc57963c72a5005 : SUCCESS in 7m 26s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/1bbff618ea834206a2085db2910cdf2e : SUCCESS in 7m 30s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6a3810b588614c5bb28f132103f27d37 : SUCCESS in 35m 09s","accounts_in_message":[],"_revision_number":2},{"id":"6a70431917cf3002c0c09d388dc74697b017f3dd","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-07-07 15:05:43.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"034d5ff6bffbaf5367c39c3d4262fc7759c1e6af","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-07-07 15:06:07.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5b45ab0077f013575e3598ee8f801751d8e8a655","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-07 15:06:56.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c6664a8be84ec6b438e7a6c02065a5e7da07ce87","author":{"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},"date":"2022-07-07 15:12:32.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"43721673bbdd6d6c1929a63940ab00eb498e7f56","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-07 16:19:46.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ec7a966fc25e460f878767af9b514a50\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac8d0dab5a1e43578c61a1c33c924328 : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd91a05fd62b40428e4605c5829639db : SUCCESS in 5m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c0b48ce627b0463f94d61924b0162d8e : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f597152a70b54b9ca20542fafd8ead7a : SUCCESS in 6m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/92cb68a020784f248255a2fb8c43c2ed : SUCCESS in 7m 28s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b498613f197643718e1172ed7e27d62f : SUCCESS in 9m 59s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/415384fb74fd409b988319edf962d23b : SUCCESS in 8m 37s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6a85c6f6cfa7462da3e630f36e5f7d4c : FAILURE in 40m 43s","accounts_in_message":[],"_revision_number":2},{"id":"9fee85285bffdbb2fd9b2d245b1273c914416ac2","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-07-07 17:59:36.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ba2f82f564c14e986a767a61097883ac04b4af3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-07 18:32:05.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e218b2ebd5904c618d9842bea326666a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17c45dc9833949dbb51eb258e1b569ed : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6570e3d7a211448aae7fc28beb7a277a : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/36205f98e0f645d9ad5994b20b2d2c20 : SUCCESS in 4m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9c35d813a0e3471aaac0f2ba0782c952 : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f34a6e9b74145dfa819a2c63c0ff335 : SUCCESS in 7m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f7dd249ad22d4b749ca4d926620de41a : SUCCESS in 5m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e6991ec6d0c145888e7695f732f8191f : SUCCESS in 9m 11s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c16c2b253c474b8c8d20442db35e6579 : SUCCESS in 8m 33s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/9f41d6488a5f422b902022196a728027 : SUCCESS in 8m 46s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/65602202f2c1471d9c2d63503e8d47b9 : FAILURE in 30m 34s","accounts_in_message":[],"_revision_number":2},{"id":"4d17f6ecc5c3d406a3f12e002859cb61ef1eb596","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-07-09 03:04:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"df1738a4315ab3af1fb96e3fa6ae11db2e8e5875","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-09 03:39:48.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/924342e759fc4dd2aaec39c1f1ea5c5c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/810cdd392d6949c2ba533f7ee008775e : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/faf01242438746a98572215314d24631 : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/74a5f93d191d49b18ecc946140082e20 : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f1d7131244c84bba89c90156c72d4828 : SUCCESS in 5m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb05acdd2a4c424a8e5a652ff38ecd16 : SUCCESS in 6m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6d61abf9691745068b7645c6195417d3 : SUCCESS in 5m 17s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/eae184095b1f43a2bd5811729065f7da : TIMED_OUT in 30m 55s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/35f02163b5d944df90cd6d859e591b26 : SUCCESS in 7m 42s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/4d4acad2849d4319a11f1646bc42e320 : SUCCESS in 8m 16s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7f9c35c21a214d28a740d2317fdfaed8 : SUCCESS in 33m 31s","accounts_in_message":[],"_revision_number":2},{"id":"a1a20f98c8e35e367658aa4ba32f11df61690abe","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-07-09 08:11:42.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"97cf42954721f54aac89e858605b011e7d1cdade","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-09 08:47:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c4e029f0e6b467aa7b04328f217a0f5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77f05544eb5747568d157f914c66e2ad : SUCCESS in 7m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ff9579da24b34efb9a37eb654255a640 : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/892d87fb45a74fb1ac1c609d50c2554a : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4f8f9ce4621d481eb2b42826321f17f6 : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/292ac69da06143daace042db2e6d72c7 : SUCCESS in 5m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3ef07c12b86f4a8f9cf1ecea55c8216f : SUCCESS in 4m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e6719dc0f61b48ffbcb572160a778ab6 : SUCCESS in 7m 36s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/adc14970f69948308495413552e0de49 : SUCCESS in 7m 55s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/5857c60d3d1d446c887665c4435207d0 : SUCCESS in 7m 20s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/aff22c6063c04f029bbcc6f7d1e6f0af : SUCCESS in 31m 20s","accounts_in_message":[],"_revision_number":2},{"id":"9b88f22e743b497ee4825d4784c90205d8792aec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-09 08:48:05.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c795be84fd685d53c4a90ca0ae00ef1c2be19c3c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-09 09:29:57.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c37e65549634984972708b472220d5b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0a51a350f38457eb75204451a19d037 : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a1cbaebd0bfa4ffda31d7a242b7c4535 : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/29096cb532614ff5aa3a33ad21f7b760 : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/686b2ff292844c4a9dbf08a561a92144 : SUCCESS in 5m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1a2df8fa88a54d44ad79a3f2aad09839 : SUCCESS in 4m 59s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a8dbe4767a1b4b22900a7034754a967b : SUCCESS in 7m 26s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ea1bf3c6cdd543f9a660befb1f59bf45 : SUCCESS in 7m 52s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6fbb4ac50552452b812c96d124e4e11f : SUCCESS in 36m 04s","accounts_in_message":[],"_revision_number":2},{"id":"0fe731807f82b08e929ecd8988dd68dbc15d1be8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-09 09:29:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"b3d6f54e1dba08afb12bcbe3594ab7b32a9dfad5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-09 09:31:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/839e43b7f1954441839095189c8a45c1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c80998b8e7140b3b10e278bf78b2359 : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/61693033c1304020bb6dda3284b5cdef : SUCCESS in 46s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e1a66af4a894b83f27f90eef0826398445b4df97","revisions":{"d1aea8068037117f53ab1aa776180afc2a48924d":{"kind":"REWORK","_number":1,"created":"2022-04-28 18:30:27.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/15/839815/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/15/839815/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/15/839815/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/15/839815/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/15/839815/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/15/839815/1"}}},"commit":{"parents":[{"commit":"881b24a2ad882ba06e16acd8e5a54eeb9d30e1a5","subject":"Update qcow test to use qcow2 instead","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/881b24a2ad882ba06e16acd8e5a54eeb9d30e1a5"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-04-28 16:00:47.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-04-28 18:30:04.000000000","tz":120},"subject":"[CI] Install dependencies for docs target","message":"[CI] Install dependencies for docs target\n\nWhen tox \u0027docs\u0027 target is called, first it installs the dependencies\n(listed in \u0027deps\u0027) in \u0027installdeps\u0027 phase, then it installs glance\n(with its requirements) in \u0027develop-inst\u0027 phase. In the latter case\n\u0027deps\u0027 is not used so that the constraints defined in \u0027deps\u0027 are not\nused. This could lead to failures on stable branches when new packages\nare released that break the build. To avoid this, the simplest solution\nis to pre-install requirements, i.e. add requirements.txt to \u0027docs\u0027 tox\ntarget.\n\nConflicts:\n  tox.ini\n\nNOTE(elod.illes): conflict is due to branch specific upper constraints\nfile link.\n\nChange-Id: Iae07e1c5abde09874928951ac55c2d0355bcd244\n(cherry picked from commit a46a43f063c9c6d5d60b7e53cbbe128bb7228997)\n(cherry picked from commit d5213071d2ff368f808a18b36e1e64750887749c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d1aea8068037117f53ab1aa776180afc2a48924d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d1aea8068037117f53ab1aa776180afc2a48924d"}]},"branch":"refs/heads/stable/xena"},"e1a66af4a894b83f27f90eef0826398445b4df97":{"kind":"REWORK","_number":2,"created":"2022-05-09 15:03:02.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/15/839815/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/15/839815/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/15/839815/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/15/839815/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/15/839815/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/15/839815/2"}}},"commit":{"parents":[{"commit":"881b24a2ad882ba06e16acd8e5a54eeb9d30e1a5","subject":"Update qcow test to use qcow2 instead","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/881b24a2ad882ba06e16acd8e5a54eeb9d30e1a5"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-04-28 16:00:47.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-05-09 15:02:29.000000000","tz":120},"subject":"[CI] Add upper constraints to install command","message":"[CI] Add upper constraints to install command\n\nIn cases when a package is installed in tox\u0027s develop-inst phase upper\nconstraints are not used if it were not defined in \u0027deps\u0027 of the tox\ntarget. This caused for example \u0027tox -e docs\u0027 to fail in case a package\nintroduced some branch incompatible change.\n\nThere are two option to fix this:\n* add the upper constraints to \u0027deps\u0027 of the target (and pre-install\n  requirements for some cases) OR\n* use it in the install_command\n\nIn the past the latter needed some extra attention (due to e.g. lower\nconstraints tests) and projects were encouraged to remove the\ninstall_command, but nowadays this seems better fitting for some\nproject and easier to use. So this patch fixes the upper constraints\nusage via the custom install_command and by removing it from \u0027deps\u0027.\n\nConflicts:\n  tox.ini\n\nNOTE(elod.illes): conflict is due to branch specific upper constraints\nlink.\n\nChange-Id: Iae07e1c5abde09874928951ac55c2d0355bcd244\n(cherry picked from commit 36aedba20e998953333225a9788c084d2e6be6ae)\n(cherry picked from commit 49d25474a2cfa1c91649241471c4b7b1b799fd75)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e1a66af4a894b83f27f90eef0826398445b4df97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e1a66af4a894b83f27f90eef0826398445b4df97"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
