)]}'
{"id":"openstack%2Fglance~839816","triplet_id":"openstack%2Fglance~stable%2Fwallaby~Iae07e1c5abde09874928951ac55c2d0355bcd244","project":"openstack/glance","branch":"stable/wallaby","topic":"fix-docs-use-uc","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2023-03-02 07:37:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iae07e1c5abde09874928951ac55c2d0355bcd244","subject":"[CI] Add upper constraints to install command","status":"MERGED","created":"2022-04-28 18:31:41.000000000","updated":"2023-03-02 07:38:19.000000000","submitted":"2023-03-02 07:37:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"839816-fix-docs-use-uc","meta_rev_id":"6ebf179be6b2d4e1d34628675e14d5b206197e2e","_number":839816,"virtual_id_number":839816,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-02 07:37:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2023-03-02 06:30:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2023-03-02 06:30:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}],"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-28 19:21:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-02 12:57:55.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"CC"},{"updated":"2022-06-02 14:21:48.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2023-03-02 06:30:42.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"f91b6c0822ec3afb02eab00a396d82a4e5938540","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-04-28 18:31:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a9543b2c63022ff07258af658c4f700f0829d248","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-28 19:21:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/10379b684fde402db86f868376cd0880 : FAILURE in 37m 18s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/474b640063c24aafbacd95c81809bae3 : SUCCESS in 9m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/708134c7762d4ae6a9b0c4a19e093868 : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/274e01d515b44321be10de127ff697aa : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/812b31abecfa4093add30bd70c70055b : SUCCESS in 6m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fcf0cc07afcc43949861cbc1c379fa19 : SUCCESS in 6m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7004c0adfa74c068dbc5ce41abc947c : SUCCESS in 7m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f0c743e4c25e48bfb02fa926baac6306 : SUCCESS in 6m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d92eabd7f2a44217bfba9a20a267a4fc : SUCCESS in 9m 10s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/35266557c2684de5a7720afa274927a2 : SUCCESS in 9m 46s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/4aac5eecb9aa4718923a3355e438684e : SUCCESS in 40m 18s","accounts_in_message":[],"_revision_number":1},{"id":"a7e8e2f2c75489179990e78b9b5e967c441085d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-05-09 16:22:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ef0fe8c069b76c53590a675c17ba2c3bfd160170","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-09 17:03:01.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/c3118fa7c75c47c69c58a5903fc7b2f2 : FAILURE in 39m 44s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6a1374cc699e42e48a05d1c5cab35cf5 : SUCCESS in 9m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8919bee935b0433b81cabf6f09be6dae : TIMED_OUT in 31m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d575e3d97fdb4bd2ab6fd683d888f512 : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/99fc1094e5d14f8bb58b609ead0ed0e6 : SUCCESS in 6m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/60fc517c7ad548b284f2268f0d431894 : SUCCESS in 6m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d8ee2a09d5e4d9e8100b7cb6f82bca5 : SUCCESS in 6m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/816f33fd80fe49ee8e44de82702b39a8 : SUCCESS in 5m 18s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fa3aab0e600a4e4bb6f9de1367b1ee5c : TIMED_OUT in 31m 13s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/69afc59b57ee4e2c9923c8a85dc9f5cf : SUCCESS in 9m 16s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8a0e7b90e2c7433dbc6063c407e85c90 : SUCCESS in 30m 19s","accounts_in_message":[],"_revision_number":2},{"id":"845b0c308cc363916193f0dea77969153ccc40ab","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-05-09 20:09:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"af69c04fbe4d5c46c6cbbad4b9f76f9828ff5c42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-09 21:03:32.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/c788fbfa060f4328bff725789e76c864 : FAILURE in 47m 39s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f11d986d328745b7b372f7f1f1b36bd5 : SUCCESS in 6m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6cfa18a822404342be599a1c5b9f7ef7 : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6571fc4cd5224b399c38aef5a019c6df : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ecc680c5449548f898f3f8aa24f60fda : SUCCESS in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e9827ca2b94b402fb4124cd3c4b0e002 : SUCCESS in 5m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/326bcc0cd01d4623a8c21f70b23b8623 : SUCCESS in 5m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e295c2c733f44a7cb12904885d98550e : SUCCESS in 6m 10s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4b757d689b02433594033c44c4b1637a : TIMED_OUT in 31m 01s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/82548c57cf2043ae947670f10ce914ab : SUCCESS in 6m 56s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/af3e4261a229434c8949af4c5f71e1ec : SUCCESS in 23m 38s","accounts_in_message":[],"_revision_number":2},{"id":"983e30b5762e811868f856a7272cc2561bfc0e87","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-06-02 12:57:55.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"4bbe6c7f4cfc3634a1e44a4b803de1d753c3a2ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 13:36:32.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6d4d828cefa2401cab34e0ad859f3cd6 : SUCCESS in 7m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e17e97a7ddd43f0bae892168125f7c8 : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/64830b6d4000440386fe72e60cbf90b7 : SUCCESS in 6m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8378c9e3c7974ea08162d8cf2fa15a2d : SUCCESS in 4m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/83596584bd9c4f7ca454edbe468569fc : SUCCESS in 6m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c6371712a8e49c7934af27ddd60ef44 : SUCCESS in 6m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/10409b29bec4452a86d5ac7590b0b804 : SUCCESS in 5m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2142f2a2061e44b6893cf24735ef4306 : TIMED_OUT in 31m 10s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/738f48c295ac45f3a7972cb4bde9e62d : SUCCESS in 7m 43s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/abd8b4b2b115446fb3d85874c59ba07f : SUCCESS in 33m 38s","accounts_in_message":[],"_revision_number":2},{"id":"44a3df96964c1e0c21d720e6a06ad32d0aa516ed","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-06-02 14:21:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d72ae057fe83c032a3d6690b8ee485a849a6dcf9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 15:06:00.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4c2bd6f3aa2745c6af55d17993ba904a : SUCCESS in 7m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eef52527d0b44260858a21717aa22b5a : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/631a8d55deb6463a801c8b004ae6b741 : SUCCESS in 6m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/64c833456b514736b4f1f5ccefbaf81e : SUCCESS in 5m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b70c141cca2e4006b54a191f5aa9e4fb : SUCCESS in 5m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/579e497bfdd24ee49e7bd054cb4740af : SUCCESS in 8m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8e81804a8a6144c49d5076c528b7e3d1 : SUCCESS in 6m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d865ad8980264ca2a942845d1f58c3bf : TIMED_OUT in 31m 03s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ee32ede5bd384e41a5215972cc04511c : FAILURE in 8m 09s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ffe480ff654549f4b275718d60352048 : SUCCESS in 31m 30s","accounts_in_message":[],"_revision_number":2},{"id":"c2aecf8d0c3e6b35054e9e8c4c270e055d020fa1","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-06-06 14:16:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0e2754ff014912ec98661b4d753c4d849b507759","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-06 14:56:58.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/020c2ce3086c4852968fd46a1574da9b : SUCCESS in 7m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/136330bdd7b94b3ea1aa9160ad68ab42 : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bf1e7eacae8943c2a07687b5c2c6e10a : SUCCESS in 5m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/71dd8ea30f37423a847838b5e472a2b5 : SUCCESS in 4m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b85fda3176d1458e80d1f44dce5eaf81 : SUCCESS in 6m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fa3e56de4ab465ab0a62f251e017897 : SUCCESS in 5m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c53259085eda44428d116f64ff8d2143 : SUCCESS in 5m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7e043f577c2f4b169d8a3bc8334021f8 : TIMED_OUT in 30m 50s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0ad7139a0f9d4b2f98bed4a54a520980 : SUCCESS in 7m 35s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5dbcda77dff44825b6187b6869fc35a2 : SUCCESS in 35m 21s","accounts_in_message":[],"_revision_number":2},{"id":"163f36c3aaf57b37eb37f6ea2940917488006da7","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-07-07 15:06:10.000000000","message":"Abandoned\n\nstable/wallaby is too old for this kind of change, only merging in yoga and xena.","accounts_in_message":[],"_revision_number":2},{"id":"f3932e8770b289612b506198935266d41ca67e1e","tag":"autogenerated:gerrit:restore","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-03-02 05:58:23.000000000","message":"Restored","accounts_in_message":[],"_revision_number":2},{"id":"e1ff291469f4f19806340aca4452a5e37a4a9a20","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-03-02 05:58:43.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5392be0ef67ccde0e318054c61e7a01288d571e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 06:29:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41b7a7b202c342cda799371023804e51\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bd5808fc73b049138c25aac982e27df0 : SUCCESS in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd2e8ad479ea475db1ebc1a3ba3aefff : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3066bdf11dbd4cec9239b5b45ea38f06 : SUCCESS in 6m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c760039fcb744fe4973812a38441d473 : SUCCESS in 4m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/92f0abc0c7744bdda14a74d52d0a319e : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a95e411ff3b4404cbd6362de7068d7f3 : SUCCESS in 5m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f4981f7a063a40428c3081ad17873a7c : SUCCESS in 6m 06s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6881bb38bd2a4ad7b98487d2d2b76405 : SUCCESS in 7m 34s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/bc642c2808764dd3aa8c9761065f98f8 : SUCCESS in 7m 32s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b6d1cc349d954c9cacfe87d445d845db : SUCCESS in 28m 31s","accounts_in_message":[],"_revision_number":2},{"id":"514fa4c11994a8809213e40e3bf0e0a2416cb03d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-03-02 06:30:42.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"aada76cd2b00cf72233a04d19b0e0aa753821d73","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 06:31:14.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b5f13e5173c13211f88308b4c1d7ff9b592116bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 07:37:10.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7035cbfc25ff45109b0541ee19d4d038\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6e284b357c3a481a87aaa77e2639d9f7 : SUCCESS in 6m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9870e9be4c14c939af6467d109f4747 : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2da3c629985a475d8cf70de27f10d395 : SUCCESS in 5m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c6fc704b8ada4270b36139e0b1ff5018 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/696b36882cec4e3293f11bd79a2272a0 : SUCCESS in 5m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/05efdea86c7042d9a067278dbfca728a : SUCCESS in 5m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7550b94ca222436d8fca7ec30f777511 : SUCCESS in 7m 38s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b06352b6b5f340babb0765d8423fbd75 : SUCCESS in 6m 58s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/69ab32cf375b4a9cb31f782c2d675283 : SUCCESS in 36m 17s","accounts_in_message":[],"_revision_number":2},{"id":"4a319696ceef1de0dab41749b30fc2e190f222a9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 07:37:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"6ebf179be6b2d4e1d34628675e14d5b206197e2e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 07:38:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e789611d10174e36a726d2ea471f8689\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e69da01d277a4670b393b83337b90640 : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6a950cdbc0ef41fc82aae35beca66550 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5948af7c14055c246bdec923c9d75389725dbce5","revisions":{"cfa0604c80d79d0fed09dd21e70d06099624a15c":{"kind":"REWORK","_number":1,"created":"2022-04-28 18:31:41.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/16/839816/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/16/839816/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/16/839816/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/16/839816/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/16/839816/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/16/839816/1"}}},"commit":{"parents":[{"commit":"b1d635c287f950b15df207455d64d97228963e5f","subject":"Fix failing copy_image flow init","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b1d635c287f950b15df207455d64d97228963e5f"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-04-28 16:00:47.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-04-28 18:31:20.000000000","tz":120},"subject":"[CI] Install dependencies for docs target","message":"[CI] Install dependencies for docs target\n\nWhen tox \u0027docs\u0027 target is called, first it installs the dependencies\n(listed in \u0027deps\u0027) in \u0027installdeps\u0027 phase, then it installs glance\n(with its requirements) in \u0027develop-inst\u0027 phase. In the latter case\n\u0027deps\u0027 is not used so that the constraints defined in \u0027deps\u0027 are not\nused. This could lead to failures on stable branches when new packages\nare released that break the build. To avoid this, the simplest solution\nis to pre-install requirements, i.e. add requirements.txt to \u0027docs\u0027 tox\ntarget.\n\nConflicts:\n  tox.ini\n\nNOTE(elod.illes): conflict is due to branch specific upper constraints\nfile link.\n\nChange-Id: Iae07e1c5abde09874928951ac55c2d0355bcd244\n(cherry picked from commit a46a43f063c9c6d5d60b7e53cbbe128bb7228997)\n(cherry picked from commit d5213071d2ff368f808a18b36e1e64750887749c)\n(cherry picked from commit d1aea8068037117f53ab1aa776180afc2a48924d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cfa0604c80d79d0fed09dd21e70d06099624a15c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cfa0604c80d79d0fed09dd21e70d06099624a15c"}]},"branch":"refs/heads/stable/wallaby"},"5948af7c14055c246bdec923c9d75389725dbce5":{"kind":"REWORK","_number":2,"created":"2022-05-09 16:22:13.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/16/839816/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/16/839816/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/16/839816/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/16/839816/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/16/839816/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/16/839816/2"}}},"commit":{"parents":[{"commit":"b1d635c287f950b15df207455d64d97228963e5f","subject":"Fix failing copy_image flow init","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b1d635c287f950b15df207455d64d97228963e5f"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-04-28 16:00:47.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-05-09 16:20:03.000000000","tz":120},"subject":"[CI] Add upper constraints to install command","message":"[CI] Add upper constraints to install command\n\nIn cases when a package is installed in tox\u0027s develop-inst phase upper\nconstraints are not used if it were not defined in \u0027deps\u0027 of the tox\ntarget. This caused for example \u0027tox -e docs\u0027 to fail in case a package\nintroduced some branch incompatible change.\n\nThere are two option to fix this:\n* add the upper constraints to \u0027deps\u0027 of the target (and pre-install\n  requirements for some cases) OR\n* use it in the install_command\n\nIn the past the latter needed some extra attention (due to e.g. lower\nconstraints tests) and projects were encouraged to remove the\ninstall_command, but nowadays this seems better fitting for some\nproject and easier to use. So this patch fixes the upper constraints\nusage via the custom install_command and by removing it from \u0027deps\u0027.\n\nConflicts:\n  tox.ini\n\nNOTE(elod.illes): conflict is due to branch specific upper constraints\nlink. Also, lower-constraints is not yet removed from this branch so to\navoid silently using wrong upper constraints instead of lower ones, the\ninstall_command needs to be redefined in lower-constraints target.\n\nChange-Id: Iae07e1c5abde09874928951ac55c2d0355bcd244\n(cherry picked from commit 36aedba20e998953333225a9788c084d2e6be6ae)\n(cherry picked from commit 49d25474a2cfa1c91649241471c4b7b1b799fd75)\n(cherry picked from commit e1a66af4a894b83f27f90eef0826398445b4df97)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5948af7c14055c246bdec923c9d75389725dbce5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5948af7c14055c246bdec923c9d75389725dbce5"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
