)]}'
{"id":"openstack%2Fglance~839846","triplet_id":"openstack%2Fglance~master~Id328be8711abda4d157241e672299d30d82a63ef","project":"openstack/glance","branch":"master","topic":"fix-docs-use-uc","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2022-04-29 12:48:37.000000000","reason":"Change was abandoned"},"5202":{"account":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"last_update":"2022-04-29 12:48:37.000000000","reason":"Change was abandoned"},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2022-04-29 12:48:26.000000000","reason":"yatin replied on the change"},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2022-04-29 12:48:37.000000000","reason":"Change was abandoned"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2022-04-29 12:48:37.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Id328be8711abda4d157241e672299d30d82a63ef","subject":"Docs build: Add requirements.txt to deps","status":"ABANDONED","created":"2022-04-29 05:41:21.000000000","updated":"2022-04-29 12:48:37.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9e83ce83901811025b92d1accf164d94f71bc46a","_number":839846,"virtual_id_number":839846,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-04-29 06:46:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"recommended":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"disliked":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":-1,"date":"2022-04-29 12:35:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":2,"date":"2022-04-29 07:01:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2022-04-29 08:33:38.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-29 06:46:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-29 07:01:12.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2022-04-29 07:01:12.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2022-04-29 07:01:12.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2022-04-29 07:01:12.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2022-04-29 07:01:12.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2022-04-29 08:33:38.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"}],"messages":[{"id":"c52c47100ae51a940bce9f42a4017a1570b231b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-04-29 05:41:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"48c82b386ef5513b0322d1accfc97c813ab03d4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 06:46:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/18deba545061449ca9282e7aafdef87c : SUCCESS in 4m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/81ec2071f9ed44f1af7e7aaa6b2c31a8 : SUCCESS in 1h 00m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d7661a8c1664c748e03842ebb6e8e40 : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f53bfc960cbc4ab3a4bd5cefe160f3c0 : SUCCESS in 5m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/82e070029d3b461eb03541cdc13d1e25 : SUCCESS in 5m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7af3e6667581457798840ba1bb17f0a4 : SUCCESS in 6m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d94243639269408f92c03802c0d6cf76 : SUCCESS in 6m 48s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/e5f3da39a1894cf3be920b86c51cdb9b : SUCCESS in 12m 35s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a2e411f61dc24bcd9954ac392fab0cfa : SUCCESS in 8m 58s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/3351e9d045e34a94aaf31684c9209168 : SUCCESS in 9m 15s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/858c0dea964d47feab28320688464874 : SUCCESS in 37m 45s","accounts_in_message":[],"_revision_number":1},{"id":"10c40b29d63d05f92e2407e097b1cde94a66bb73","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-04-29 07:01:12.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"51fb7410dba63b5fec48a8105e8ba3e08979b0cb","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2022-04-29 08:33:38.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"967e6de20906490c131d7d131b06fa7ca2a7eb4f","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-04-29 12:35:34.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ca237fb37135a84c7c7ce33428bdac5a646a7dde","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-04-29 12:48:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9e83ce83901811025b92d1accf164d94f71bc46a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-04-29 12:48:37.000000000","message":"Abandoned\n\nIn favor of https://review.opendev.org/c/openstack/glance/+/839786","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5463df79a045c89ed2d773b2dc0628f144057b6d","revisions":{"5463df79a045c89ed2d773b2dc0628f144057b6d":{"kind":"REWORK","_number":1,"created":"2022-04-29 05:41:21.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/46/839846/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/46/839846/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/46/839846/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/46/839846/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/46/839846/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/46/839846/1"}}},"commit":{"parents":[{"commit":"14bd3eb1bb60dafc1ee33cf0e58b0d1e1b4cad67","subject":"Update migration constant","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/14bd3eb1bb60dafc1ee33cf0e58b0d1e1b4cad67"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-04-29 05:40:35.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-04-29 05:40:38.000000000","tz":330},"subject":"Docs build: Add requirements.txt to deps","message":"Docs build: Add requirements.txt to deps\n\nSince we use usedevelop\u003dTrue, as a result glance\ninstallation happens at \u0027develop-inst\u0027 phase and it installs\nrequirements from requirements.txt without considering\ntox constraints, This can lead to issue where broken\nrequirements are installed in docs job and as a result\nit will fail.\n\nIn order to get constraints be considered for this case, add\nrequirements.txt to deps for docs env.\n\nChange-Id: Id328be8711abda4d157241e672299d30d82a63ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5463df79a045c89ed2d773b2dc0628f144057b6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5463df79a045c89ed2d773b2dc0628f144057b6d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
