)]}'
{"id":"openstack%2Fglance~840576","triplet_id":"openstack%2Fglance~master~If5c31e53f43d3d11e67ce0e1dd0fb5d502dd8ba9","project":"openstack/glance","branch":"master","topic":"dnm-rally-job","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2022-05-05 06:57:35.000000000","reason":"Abhishek Kekane replied on the change"}},"hashtags":[],"change_id":"If5c31e53f43d3d11e67ce0e1dd0fb5d502dd8ba9","subject":"[DNM] Glance rally create and delete image scenario with Ceph","status":"ABANDONED","created":"2022-05-05 05:56:58.000000000","updated":"2022-05-11 05:37:42.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4b6f4434e26a76819e1dfe8db0b86431904473cf","_number":840576,"virtual_id_number":840576,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-05-05 08:15:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-05 05:57:17.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-05 05:57:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b77e198c0ab74decc9d843b9858b0ff74e6da1d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-05-05 05:56:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2ca7c8bb18d29592e738600133e647b17bb424b9","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-05-05 05:57:17.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/glance on branch master.  The error was:\n\n  mapping values are not allowed in this context\n  in \"openstack/glance/.zuul.yaml@master\", line 21, column 13","accounts_in_message":[],"_revision_number":1},{"id":"bb3ac2eb5e86707b3f840683e16bc7c2ed981810","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 05:57:37.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/glance on branch master.  The error was:\n\n  mapping values are not allowed in this context\n  in \"openstack/glance/.zuul.yaml@master\", line 21, column 13","accounts_in_message":[],"_revision_number":1},{"id":"233a4f67ae11744be9f877e00a1f459766444b4a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-05-05 06:05:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"00ce16d0f7e3459ab9f20c6e5c617f96be60c50a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 06:47:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/55b437387e1545d5879eee60d21ad68e : SUCCESS in 41m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7d2dc23434a4441abb3da96113f1b93 : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d2f85a8030254448b6985be781795ea9 : SUCCESS in 5m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/beb502044eba480788f7baa80b68b02e : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b0a4defd6e6416da5199939a2f5cdc1 : SUCCESS in 6m 32s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/72a99e16fca344cba6166af2cee642c3 : SUCCESS in 11m 00s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/702ad6763c894ae1a78b9743c2d5151a : SUCCESS in 8m 40s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/7b9193c837104f76b3734940fd8b3624 : SUCCESS in 11m 36s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/432823de27d142418a849cca0feabfda : SUCCESS in 27m 48s\n- glance-rally-task https://zuul.opendev.org/t/openstack/build/27bd1071bc864ef996bcf4e6625b6b31 : FAILURE in 38m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b24b3344a3c311a02263c90d809c29bc376a334f","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-05-05 06:57:35.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5a68f825c64dc4022ac166a6bf5f59a73dd9b7db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-05-05 07:16:42.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"44b5199f7c0012bba7229cdacd01deaff614ed7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 08:15:09.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/31cf42d3dd024d559d8dbcde71841101 : SUCCESS in 53m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/065090cefdcf41a9a3c7a47b5f9427f4 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aabef101079b4d4a8351912e6805d1cb : SUCCESS in 4m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bfb6d20780bb44c1b1e0333a211237ea : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53f08388ac394b7a997892c67c049851 : SUCCESS in 6m 07s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/0842bff4cdc04e33a58e9b3be25b5b57 : SUCCESS in 10m 24s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5c629a56ac6f4715b0d06858fb49081d : SUCCESS in 11m 53s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/f3c0788923fd4e699a839d82c31d9c6e : FAILURE in 12m 11s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7205b76dc5f24f3e88fbf7c00894318d : SUCCESS in 33m 03s\n- glance-rally-task https://zuul.opendev.org/t/openstack/build/bd77d2c9c754449e84e6acc2f38519f9 : FAILURE in 34m 37s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4b6f4434e26a76819e1dfe8db0b86431904473cf","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-05-11 05:37:42.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"5ea1e4e4599324fc1b4646ae1e66448b57f8beba","revisions":{"8faac95e95176d9287df5f9322fc36bf68fdb11f":{"kind":"REWORK","_number":1,"created":"2022-05-05 05:56:58.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/76/840576/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/76/840576/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/840576/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/840576/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/840576/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/76/840576/1"}}},"commit":{"parents":[{"commit":"8271cde1e2850ff1870b310801e3fa8c7fbbf998","subject":"[api-ref] update cache manage API info","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8271cde1e2850ff1870b310801e3fa8c7fbbf998"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2022-05-05 05:43:00.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2022-05-05 05:43:43.000000000","tz":0},"subject":"[DNM] Glance rally create and delete image scenario with Ceph","message":"[DNM] Glance rally create and delete image scenario with Ceph\n\nDepends-On: https://review.opendev.org/c/openstack/rally-openstack/+/840575\nChange-Id: If5c31e53f43d3d11e67ce0e1dd0fb5d502dd8ba9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8faac95e95176d9287df5f9322fc36bf68fdb11f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8faac95e95176d9287df5f9322fc36bf68fdb11f"}]},"branch":"refs/heads/master"},"d9d3ab240d92c4eb712fba8e40eafd9ca54b5ae8":{"kind":"REWORK","_number":2,"created":"2022-05-05 06:05:03.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/76/840576/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/76/840576/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/840576/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/840576/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/840576/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/76/840576/2"}}},"commit":{"parents":[{"commit":"8271cde1e2850ff1870b310801e3fa8c7fbbf998","subject":"[api-ref] update cache manage API info","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8271cde1e2850ff1870b310801e3fa8c7fbbf998"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2022-05-05 05:43:00.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2022-05-05 05:54:39.000000000","tz":0},"subject":"[DNM] Glance rally create and delete image scenario with Ceph","message":"[DNM] Glance rally create and delete image scenario with Ceph\n\nDepends-On: https://review.opendev.org/c/openstack/rally-openstack/+/840575\nChange-Id: If5c31e53f43d3d11e67ce0e1dd0fb5d502dd8ba9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d9d3ab240d92c4eb712fba8e40eafd9ca54b5ae8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d9d3ab240d92c4eb712fba8e40eafd9ca54b5ae8"}]},"branch":"refs/heads/master"},"5ea1e4e4599324fc1b4646ae1e66448b57f8beba":{"kind":"REWORK","_number":3,"created":"2022-05-05 07:16:42.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/76/840576/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/76/840576/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/840576/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/840576/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/840576/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/76/840576/3"}}},"commit":{"parents":[{"commit":"8271cde1e2850ff1870b310801e3fa8c7fbbf998","subject":"[api-ref] update cache manage API info","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8271cde1e2850ff1870b310801e3fa8c7fbbf998"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2022-05-05 05:43:00.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2022-05-05 07:06:56.000000000","tz":0},"subject":"[DNM] Glance rally create and delete image scenario with Ceph","message":"[DNM] Glance rally create and delete image scenario with Ceph\n\nDepends-On: https://review.opendev.org/c/openstack/rally-openstack/+/840575\nChange-Id: If5c31e53f43d3d11e67ce0e1dd0fb5d502dd8ba9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5ea1e4e4599324fc1b4646ae1e66448b57f8beba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5ea1e4e4599324fc1b4646ae1e66448b57f8beba"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
