)]}'
{"id":"openstack%2Fglance~841034","triplet_id":"openstack%2Fglance~master~I48f88c188f830f61c03ec5570cd95731636a0396","project":"openstack/glance","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-05-18 12:44:59.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2022-05-18 11:33:46.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I48f88c188f830f61c03ec5570cd95731636a0396","subject":"Migrate CentOS Stream 8 job to CentOS Stream 9","status":"MERGED","created":"2022-05-08 13:32:03.000000000","updated":"2022-05-18 12:49:44.000000000","submitted":"2022-05-18 12:44:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"841034","meta_rev_id":"0dd1b66af65773631435387395d71cd315bc1bbc","_number":841034,"virtual_id_number":841034,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-18 12:44:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2022-05-18 06:36:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2022-05-18 11:33:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2022-05-18 11:33:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-08 14:20:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-17 20:59:40.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"CC"},{"updated":"2022-05-18 06:36:08.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2022-05-18 11:33:46.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"fe845b619112278777a85ff53f956eadcd4f2766","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-05-08 13:32:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1ec6eda5166653032718c0ffac3f230779c88c56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-08 14:20:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b50f3694f65e47e0834c5d387f612ab7 : SUCCESS in 47m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/818db9046b6043c8b9ff139466828224 : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/72ba302c974c4e388ce82ecefefc7b41 : SUCCESS in 5m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/66fbc3e94cfc4f3f86ea0e38fe0be17a : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82f1dc0e27f04bd1865261c4ee83de48 : SUCCESS in 5m 31s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/aed435da0f6846e3b8fffe5ecc1e9eda : SUCCESS in 9m 02s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0bf4f0ffb7594b9c9cca4b398177793a : SUCCESS in 7m 57s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/ab2977c8c0a0403ea25ac6fbff779d3b : SUCCESS in 9m 15s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/cf4820b37f3f472eb6590293ae39e2a9 : SUCCESS in 26m 58s","accounts_in_message":[],"_revision_number":1},{"id":"426d9d57fa07fb401005aa6c6115b15dd3b58dc7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-05-09 10:47:18.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"3ba14d288d064ac908b473dac69860c9b154c21b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-09 11:32:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ecfdeb10d36049f88fcee1ecfae48ace : SUCCESS in 38m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a56c8c31cdb34251b7d2277f6ebf1c62 : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c734647a9ec9411fa9d987bce3bbefdc : SUCCESS in 6m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cc5ced3b2a8c4850a7a5da76a2b96740 : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0eb2e1b8c8ac4d1cbe945005c7122582 : SUCCESS in 5m 56s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/c5cf42f92c0047c198b624c59983b709 : SUCCESS in 8m 24s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e216ec496076490394bfdd229846063c : SUCCESS in 9m 01s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/106229e7302b44a8bb3bea11d58624da : SUCCESS in 7m 39s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1b99c784b4c44e96894cc2a547fef135 : SUCCESS in 37m 51s","accounts_in_message":[],"_revision_number":2},{"id":"ba009847870383c938a839f55eeefd2b25ac3f66","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-05-17 20:59:40.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e797af2b6fc81a7c1dab612338cda6a945dfa95f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-05-17 23:36:45.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d4a5323f435a69f21e5117f755935a08d897fac7","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-05-17 23:38:33.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6ba57275b1dd973941f25a8b9d45a20198139df6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-18 00:33:45.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/73a2f2176ff549ee96d2c75ca1f12ac8 : SUCCESS in 51m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6fd320b3424c446ba2ffa4985eda93d6 : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/336fe5850eac4733a950cf93b3456d41 : SUCCESS in 5m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/74225f7884f44836a90735475dd9b10a : SUCCESS in 5m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae276043f98444d7b274c60220b31708 : SUCCESS in 7m 22s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/6586e5a8bf114a30b7bc591d673d588f : SUCCESS in 10m 27s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/631d684d579e48c0a0056948cfab9367 : SUCCESS in 10m 06s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/13dbc6340cb048a68f77269d5b55d1ff : SUCCESS in 9m 27s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/9dd7b30d04e941689fc7f5aeadf0c6a2 : SUCCESS in 30m 41s","accounts_in_message":[],"_revision_number":3},{"id":"3d495085fd6a783c94eefd8ae32fda26731d99d7","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-05-18 06:36:08.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ff4453c5b7367ce5bb330ee685ed70ae00a104d8","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-05-18 11:33:46.000000000","message":"Patch Set 3: Code-Review+2\n\nOK, thanks for the clarification!","accounts_in_message":[],"_revision_number":3},{"id":"155eaa089d69e47d516ad52d8ff7b23ecbf8667b","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-05-18 11:33:59.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"07ff727449095a4e46a0552b915229da1a5bc6d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-18 11:35:01.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ef33d1456ee721a8d7833dbf81d20f495dba5a19","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-18 12:44:59.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bc8213c9070a46a2a43b59fea233fce1 : SUCCESS in 47m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/274da6639f414c7ca99de71343a7e94b : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2f4e7a49106c460aa0c29880b6490b2e : SUCCESS in 6m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d5238428985041ff95edba92c5423849 : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d80b8b8867cd44f7b84faebe149faf92 : SUCCESS in 8m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5c13eb2cd8954cd7afec89975a0e331f : SUCCESS in 8m 25s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1b82922414b84426b79e5c163805c274 : SUCCESS in 10m 33s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2a0a816c9803411a99f522e066565992 : SUCCESS in 34m 41s","accounts_in_message":[],"_revision_number":3},{"id":"526a696138cc54e72ffff50fdbf5485cd588e1b6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-18 12:44:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"0dd1b66af65773631435387395d71cd315bc1bbc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-18 12:49:44.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0885bdfd0cc94b188ca4fa0765f4a373 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ce8fd56e1a501c83c5a8bce9923b15e25810ac75","revisions":{"0bc13f9f39b87be6b03414a9b575bc8b674fd855":{"kind":"REWORK","_number":1,"created":"2022-05-08 13:32:03.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/34/841034/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/34/841034/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/34/841034/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/34/841034/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/34/841034/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/34/841034/1"}}},"commit":{"parents":[{"commit":"cb65cc8e44523cd04ecace02296061a6ddc3982d","subject":"Internal server error if shared member tries to stage data to image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cb65cc8e44523cd04ecace02296061a6ddc3982d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-05-08 13:30:11.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-05-08 13:30:11.000000000","tz":540},"subject":"Migrate CentOS Stream 8 job to CentOS Stream 9","message":"Migrate CentOS Stream 8 job to CentOS Stream 9\n\nSome projects like cinder has dropped support for Python 3.6 and is no\nlonger compatible with CentOS Stream 8 which uses Python 3.6 as\nthe default runtime.\n\nChange-Id: I48f88c188f830f61c03ec5570cd95731636a0396\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0bc13f9f39b87be6b03414a9b575bc8b674fd855"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0bc13f9f39b87be6b03414a9b575bc8b674fd855"}]},"branch":"refs/heads/master"},"05a72b6c6cc984d9d2105d7e97704e9c21a79e81":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-05-09 10:47:18.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/34/841034/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/34/841034/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/34/841034/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/34/841034/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/34/841034/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/34/841034/2"}}},"commit":{"parents":[{"commit":"cb65cc8e44523cd04ecace02296061a6ddc3982d","subject":"Internal server error if shared member tries to stage data to image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cb65cc8e44523cd04ecace02296061a6ddc3982d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-05-08 13:30:11.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-05-09 10:47:18.000000000","tz":0},"subject":"Migrate CentOS Stream 8 job to CentOS Stream 9","message":"Migrate CentOS Stream 8 job to CentOS Stream 9\n\nSome projects like cinder have dropped support for Python 3.6 and are\nno longer compatible with CentOS Stream 8 which uses Python 3.6 as\nthe default runtime.\n\nChange-Id: I48f88c188f830f61c03ec5570cd95731636a0396\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/05a72b6c6cc984d9d2105d7e97704e9c21a79e81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/05a72b6c6cc984d9d2105d7e97704e9c21a79e81"}]},"branch":"refs/heads/master","description":"Edit commit message"},"ce8fd56e1a501c83c5a8bce9923b15e25810ac75":{"kind":"REWORK","_number":3,"created":"2022-05-17 23:36:45.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/34/841034/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/34/841034/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/34/841034/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/34/841034/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/34/841034/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/34/841034/3"}}},"commit":{"parents":[{"commit":"9bc4ea498940cbae9c7f88d7f29381a0b6364992","subject":"Merge \"Override GLANCE_USE_IMPORT_WORKFLOW in cinder jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9bc4ea498940cbae9c7f88d7f29381a0b6364992"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-05-08 13:30:11.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-05-17 23:36:10.000000000","tz":540},"subject":"Migrate CentOS Stream 8 job to CentOS Stream 9","message":"Migrate CentOS Stream 8 job to CentOS Stream 9\n\nSome projects like cinder have dropped support for Python 3.6 and are\nno longer compatible with CentOS Stream 8 which uses Python 3.6 as\nthe default runtime.\n\nChange-Id: I48f88c188f830f61c03ec5570cd95731636a0396\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ce8fd56e1a501c83c5a8bce9923b15e25810ac75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ce8fd56e1a501c83c5a8bce9923b15e25810ac75"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
