)]}'
{"id":"openstack%2Fglance~841167","triplet_id":"openstack%2Fglance~master~I987ca42a589a3af8e52fbe4f9e68d345fb8bdf5a","project":"openstack/glance","branch":"master","topic":"multistores-typo","attention_set":{},"removed_from_attention_set":{"34781":{"account":{"_account_id":34781,"name":"Ethan Myers","display_name":"ethanmyers","email":"ethan.myers@canonical.com","username":"ethanmyers"},"last_update":"2022-05-10 18:22:31.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I987ca42a589a3af8e52fbe4f9e68d345fb8bdf5a","subject":"fixed typo in multistore documentation","status":"MERGED","created":"2022-05-09 20:18:47.000000000","updated":"2022-05-10 18:24:05.000000000","submitted":"2022-05-10 18:22:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"841167-multistores-typo","meta_rev_id":"d322e580927f77b4fa9e3a8674ffa6d70c4eefba","_number":841167,"virtual_id_number":841167,"owner":{"_account_id":34781,"name":"Ethan Myers","display_name":"ethanmyers","email":"ethan.myers@canonical.com","username":"ethanmyers"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-10 18:22:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":2,"date":"2022-05-10 17:37:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":1,"date":"2022-05-10 12:44:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":1,"date":"2022-05-10 17:37:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-09 20:25:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-10 12:44:20.000000000","updated_by":{"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},"reviewer":{"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},"state":"REVIEWER"},{"updated":"2022-05-10 17:37:17.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"}],"messages":[{"id":"55062cc9132e494c7e09f1694bccc99352b0887a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34781,"name":"Ethan Myers","display_name":"ethanmyers","email":"ethan.myers@canonical.com","username":"ethanmyers"},"date":"2022-05-09 20:18:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"06e3baf20c722ecedbdf5ac54d17a1d4f17e7aec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-09 20:25:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1aaffc2aae5146929b46ef04e3b0b1b2 : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92042f45c47c4bf8a5d83460e471f229 : SUCCESS in 5m 28s","accounts_in_message":[],"_revision_number":1},{"id":"a3f992717bf1cdfc308d4fd2a2199ee733d3449b","author":{"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},"date":"2022-05-10 12:44:20.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"8dfdc668ac1e66e9d5e5234ebd0eb8822226b8f8","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2022-05-10 17:37:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"342085634981c79a10fd1ffb84043293330aed84","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-10 17:37:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b54cce8ef694640f782caf2859ec923bbceb437b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-10 18:22:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ddd70747cdfe4f839aa8b9fb2ec47716 : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/042cb3c5d0e94c9f969c8aafcfaf4cf9 : SUCCESS in 6m 34s","accounts_in_message":[],"_revision_number":1},{"id":"99b706b66872ef559272e914018d0d36b125ed8b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-10 18:22:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d322e580927f77b4fa9e3a8674ffa6d70c4eefba","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-10 18:24:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d751b16be21446bb0e7aa0f0cd9e467 : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"18bebc188122a8b41e02eea69ea91ad51ec218c3","revisions":{"18bebc188122a8b41e02eea69ea91ad51ec218c3":{"kind":"REWORK","_number":1,"created":"2022-05-09 20:18:47.000000000","uploader":{"_account_id":34781,"name":"Ethan Myers","display_name":"ethanmyers","email":"ethan.myers@canonical.com","username":"ethanmyers"},"ref":"refs/changes/67/841167/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/67/841167/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/67/841167/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/67/841167/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/67/841167/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/67/841167/1"}}},"commit":{"parents":[{"commit":"f40fdba514ed10f1c9b3ce78bc6ac39e5bee8423","subject":"Merge \"[CI] Add upper constraints to install command\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f40fdba514ed10f1c9b3ce78bc6ac39e5bee8423"}]}],"author":{"name":"Ethan Myers","email":"ethan@ethanmye.rs","date":"2022-05-09 20:13:26.000000000","tz":-360},"committer":{"name":"Ethan Myers","email":"ethan@ethanmye.rs","date":"2022-05-09 20:13:26.000000000","tz":-360},"subject":"fixed typo in multistore documentation","message":"fixed typo in multistore documentation\n\nChange-Id: I987ca42a589a3af8e52fbe4f9e68d345fb8bdf5a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/18bebc188122a8b41e02eea69ea91ad51ec218c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/18bebc188122a8b41e02eea69ea91ad51ec218c3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
