)]}'
{"id":"openstack%2Fglance~841279","triplet_id":"openstack%2Fglance~stable%2Fyoga~I81156e32517b03577c09d489b6eb5d19769600a3","project":"openstack/glance","branch":"stable/yoga","topic":"bug/1972666-stable/yoga","attention_set":{},"removed_from_attention_set":{"33765":{"account":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"last_update":"2022-06-02 18:25:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I81156e32517b03577c09d489b6eb5d19769600a3","subject":"Added cli_opts and cache_opts","status":"MERGED","created":"2022-05-11 05:56:13.000000000","updated":"2022-06-02 18:26:36.000000000","submitted":"2022-06-02 18:25:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"841279-bug/1972666-stable/yoga","cherry_pick_of_change":841249,"cherry_pick_of_patch_set":3,"meta_rev_id":"7eb13795b1f2a51c51157013c48668f8db70d720","_number":841279,"virtual_id_number":841279,"owner":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-02 18:25:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2022-05-24 10:16:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2022-06-02 12:47:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2022-06-02 12:47:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-11 07:29:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-11 14:43:13.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"CC"},{"updated":"2022-05-20 02:41:47.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"CC"},{"updated":"2022-05-23 14:14:26.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2022-05-24 10:16:13.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2022-06-02 12:47:24.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"f41d8a966973b89fd7c77c9ad54fdd24b1256f57","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"date":"2022-05-11 05:56:13.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"fd5850f4335a587e0affd426883e1d490b8f44ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 07:29:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/35d09757a30b4500b6ae21f130529626 : SUCCESS in 55m 04s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/03176fb4f3764a359713aebad2cbc603 : SUCCESS in 55m 15s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c10dc772622c490095193c800284992d : SUCCESS in 1h 21m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/03ed6bae70d24ac6a76e188c756b00ca : SUCCESS in 44m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/effeda4ffce94dd0a04bd16e329a237a : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a4d26f80e154b8caff6f2e90dd13f69 : SUCCESS in 6m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1c0f61f4688a4b64a0b62c29a65f1c5e : SUCCESS in 6m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/353096f1e4974273bda6b1eae2304bb9 : SUCCESS in 6m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/601c3674d71f430e9950c63185428dc1 : SUCCESS in 5m 50s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/3e7c588f88c54ad282788f8f1ff9c230 : SUCCESS in 10m 23s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/cd6371e9844349a6bdb8355d80baf7d7 : SUCCESS in 9m 25s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/3030bb965b5f444cb378ab39855ab548 : SUCCESS in 8m 51s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/e755db1a01d24cc1b1f4236f5af2b92d : SUCCESS in 4m 47s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/b509daeb98c548f496025ed481ae2240 : SUCCESS in 1h 07m 06s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/11bb6b1e93b941d1a7316e00245a8e9b : SUCCESS in 1h 14m 42s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/2751774de8b14716a64f7abf4ee6040d : SUCCESS in 1h 10m 28s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a8b9af15724d45cfb75e5f6f4817ec85 : FAILURE in 25m 54s\n- glance-multistore-cinder-import-fips https://zuul.opendev.org/t/openstack/build/a011683aa24d474cb2f5b74002476a4f : FAILURE in 1h 31m 06s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5559bbf195734f2592c515acbcda754f : SUCCESS in 48m 21s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9d7ef3d0fc114aacbfb51001d75dbd8a : SUCCESS in 1h 04m 27s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c6823184b039497aa9a61199d33d505e : SUCCESS in 31m 48s","accounts_in_message":[],"_revision_number":1},{"id":"3c4cf4ce7027e4e319b1962b6bd50dc06b001844","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2022-05-11 14:43:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"320c8c7871baa2697dcf711fefe308acb0db4ace","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 16:20:42.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/b643bb36c3de4e4d81adc4e326d6dd6f : SUCCESS in 46m 38s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/2adfca397a684bcbb419c1376b48983d : SUCCESS in 1h 00m 57s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7f53f9a897e04672b3f9beceba5f2c54 : SUCCESS in 1h 08m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/245bb1ca65f44dbca300b678456a9e5c : SUCCESS in 45m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f0af908314e4d0389096550cbc0bcdd : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f18c87b9ce4a49518e2b02b5b43a4cfe : SUCCESS in 6m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b64ff6ca0be949888bfb8c7f9fce1114 : SUCCESS in 6m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67687d3c2163418b8b35770523954487 : SUCCESS in 6m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5dd861e45af844d7beb8e91654a0f839 : SUCCESS in 6m 29s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/f78cae50f884438298ef03e624e91e36 : SUCCESS in 10m 39s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/977653550c984475b0b5e78f3aee1282 : SUCCESS in 10m 31s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/b86d043df1ac4f16a95d8bcb4c56a162 : SUCCESS in 12m 01s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/79e36ff8af394b308242ff28e5b734f7 : SUCCESS in 6m 31s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/7c4b4255a4c04ea4b59ed05c6c7e603f : FAILURE in 59m 51s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/77c882c3e0f34a9fb2f512aaada016a5 : SUCCESS in 1h 09m 12s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/9d0fb6893ddf45c2bc455534e4217ad7 : SUCCESS in 1h 12m 33s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b2d654e862654a0aa098cb8152cafd53 : FAILURE in 21m 16s\n- glance-multistore-cinder-import-fips https://zuul.opendev.org/t/openstack/build/e1c2c1a6003245f996cab9cd410b7514 : FAILURE in 1h 35m 08s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/be41b30eddc84c20a7f719961123e052 : SUCCESS in 51m 21s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c28e14217d3045ababf67b5ff0a47965 : SUCCESS in 1h 27m 21s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/fbd270d5920445a285bb6dde9c3b9a20 : SUCCESS in 33m 36s","accounts_in_message":[],"_revision_number":1},{"id":"84a01fbc0003ac416112837a724e6f325ad0336e","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-05-20 02:41:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dadde68d104fc691b9afaa73e689a836cf8ebfd9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-20 04:25:13.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/5abf9e477e6c4f2881ef6426b8e2d7f8 : SUCCESS in 46m 43s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/8e4f90bb318840078af74fdc2bbcff05 : SUCCESS in 1h 06m 29s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c26375f758f340d2909fee564a8f5018 : SUCCESS in 1h 14m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/680619242e8e4838ba1ec592fdd25fa8 : SUCCESS in 53m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ee032b99f6449e08e8e00d64e86feda : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9ea6fc8c1f6042ea8a2080c14c9ab462 : SUCCESS in 5m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f112a21ac7a04f09a712e76edcad66ff : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97170e504514458682e2256c26a7be61 : SUCCESS in 7m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8e143fcaf3384f40bb8b11677a982621 : SUCCESS in 6m 22s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/3154c54d56244123b8e7c04895498c49 : SUCCESS in 11m 54s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3eb258b6f8524faa832587ed76d8fd3a : SUCCESS in 8m 53s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/a250ecea39674488b64e77e35c3cc00e : SUCCESS in 9m 52s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/b064edbf9ef74730aec54c45cedfa3ae : SUCCESS in 4m 36s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f800dba7bee644e391dfe21101234ff9 : SUCCESS in 1h 14m 30s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b427ac5edf4f4489ac51b5c7efc5d8c8 : SUCCESS in 1h 30m 31s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/21bbe96637ea4040bf90eaf2a04f0ee8 : SUCCESS in 1h 09m 13s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/069ab1706f944b54add90448e0a99e8a : FAILURE in 1h 26m 48s\n- glance-multistore-cinder-import-fips https://zuul.opendev.org/t/openstack/build/30ce93c4acb048898654def5156eefdf : FAILURE in 39m 38s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/933b5b7970344859880fb6030f23c0e1 : SUCCESS in 59m 15s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/486ed997d17f4043aab2ebcf966e264c : SUCCESS in 1h 17m 27s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/23f1f3ba7a834d65bc010a87bae5e194 : FAILURE in 45m 54s","accounts_in_message":[],"_revision_number":1},{"id":"d1907f847cf166f516f25c7bad04e1f111bf397e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-05-23 14:14:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d9d093fc441adeb8c17be5a55fe41fd6a0b72180","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 15:56:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/cd3426263e9548659a336eb145c23ee5 : SUCCESS in 1h 06m 35s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/2fb7787e1806459fb81c5fdf569d26e9 : SUCCESS in 1h 01m 21s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3ed77255339643e1be573d7dfab34d81 : SUCCESS in 1h 25m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0792b542a383467fa9b13454b28467f9 : SUCCESS in 1h 01m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/986788b29f9a4f11aae6dc99124dc27e : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/512c719a75b44c3ca0add8b735805fd9 : SUCCESS in 6m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1daaef3f72294eb89a1b28d6183b6450 : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf204a631dd14c5eb46f19ea844b7a40 : SUCCESS in 7m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c26ee597681475cae5937e281f17c1c : SUCCESS in 6m 06s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/6c6b80c2ecb143d6bbde9fe63570703f : SUCCESS in 12m 53s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d0c11752daea4504b561ae0be4b925b1 : SUCCESS in 10m 55s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/43687ddbd1c14e04917e9db9b2692c8b : SUCCESS in 10m 29s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/236b3b5be90e4d7199d4ff5bb2e6924b : SUCCESS in 4m 13s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/8a3c430e924b45fe9c535db6d97e1109 : SUCCESS in 1h 10m 04s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7301b2f4646b443da879c7d8c4259f0b : SUCCESS in 1h 10m 11s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/3d866a2286ff465ead3c887e1de85405 : SUCCESS in 1h 11m 21s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/196ff531755347ff908a7907f2db31d5 : SUCCESS in 1h 35m 30s\n- glance-multistore-cinder-import-fips https://zuul.opendev.org/t/openstack/build/1686db14c0ab42a296e6cfad5de07fc8 : FAILURE in 45m 20s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8206eeac3a30495bbbac574da8515988 : SUCCESS in 51m 28s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/67dcff174c554316af9918f618663dc2 : SUCCESS in 1h 18m 35s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/cf818875e4f84d11ba98fa18b376b4ea : SUCCESS in 34m 55s","accounts_in_message":[],"_revision_number":1},{"id":"cb36c5f98bf427fa6a49992e2264be6b413798fe","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-05-24 10:16:13.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4b74a9ea3c9425a670d44ef3c095015bd2c2a6cd","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-06-02 12:47:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"feaaa642ac3de2668b674f16bfd3514486771109","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 12:48:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ccb5c72afe9f76aa63894b95ebd50936cccf6281","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 18:25:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/92d4998e9aea499c82f6a21fd936226b : SUCCESS in 50m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b7d8ed687da8426ea4dab31ab6f56226 : SUCCESS in 1h 10m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/10aab4aba42b4995bf069536bca5d85a : SUCCESS in 1h 00m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5751605eba40494381ac95d27bfc5805 : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4844fd9045594b6ca56d6e09e8c5c14b : SUCCESS in 8m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a0b53d5cfaa84dc3b3d98d74b9706af2 : SUCCESS in 10m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd9fd3406dfd40109967e3676b3c99c1 : SUCCESS in 7m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/37ffe49c49124fd1b66a698b1df5e9aa : SUCCESS in 6m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d875031f3366406db789aa25b4c4b1cb : SUCCESS in 9m 52s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3e9ca0fa338940cb8d24d2d389d2805d : SUCCESS in 16m 04s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/87ed0d780d5e4e149e2df638cedd84f5 : SUCCESS in 1h 13m 08s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/bfe5b19a2f8e481992727e3e39f760c5 : SUCCESS in 1h 10m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0ef998ebb2e34ed3bcb338d2d48b3e0c : SUCCESS in 45m 43s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/229309444ea048bcb19ce56c5af6b2c6 : SUCCESS in 1h 04m 11s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c21a101bfb324a269ab0c57b98c73df4 : SUCCESS in 39m 51s","accounts_in_message":[],"_revision_number":1},{"id":"fd2436aa9aa3ad158752f5d6f176bca755b5d2be","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 18:25:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7eb13795b1f2a51c51157013c48668f8db70d720","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 18:26:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a07ffbb8d3784879af35fef7bf43a96a : SUCCESS in 59s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9fd6ca346951462ba4f585f927497b32 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e943645fb4e67384c17bbaf9330453624abaa1d3","revisions":{"e943645fb4e67384c17bbaf9330453624abaa1d3":{"kind":"REWORK","_number":1,"created":"2022-05-11 05:56:13.000000000","uploader":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"ref":"refs/changes/79/841279/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/79/841279/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/79/841279/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/79/841279/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/79/841279/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/79/841279/1"}}},"commit":{"parents":[{"commit":"1beee78e096ddba4729be10af6ac4ddec8e79da0","subject":"Update TOX_CONSTRAINTS_FILE for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1beee78e096ddba4729be10af6ac4ddec8e79da0"}]}],"author":{"name":"Mridula Joshi","email":"mrjoshi@redhat.com","date":"2022-05-10 09:47:35.000000000","tz":0},"committer":{"name":"Mridula Joshi","email":"mrjoshi@redhat.com","date":"2022-05-11 05:56:13.000000000","tz":0},"subject":"Added cli_opts and cache_opts","message":"Added cli_opts and cache_opts\n\nIt is observed that \u0027cli_opts\u0027 and \u0027cache_opts\u0027 from\nglance/common/wsgi.py are not picked up by the configgen nor listed\nthrough the functions in glance/opts.py.\nThis patch adds these options to support configgen to pick all groups\nfrom wsgi.py.\n\nCloses-Bug: #1972666\nChange-Id: I81156e32517b03577c09d489b6eb5d19769600a3\n(cherry picked from commit c9d7ecdb7701f85787d9de61dad37eadb6d2bc8f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e943645fb4e67384c17bbaf9330453624abaa1d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e943645fb4e67384c17bbaf9330453624abaa1d3"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
