)]}'
{"id":"openstack%2Fglance~841350","triplet_id":"openstack%2Fglance~master~Id0813d9dc553dd424732079039349b42f6f3201b","project":"openstack/glance","branch":"master","topic":"drop-py3-6-7","attention_set":{},"removed_from_attention_set":{"19138":{"account":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"last_update":"2022-05-19 21:47:33.000000000","reason":"Change was submitted"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2022-05-19 21:47:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id0813d9dc553dd424732079039349b42f6f3201b","subject":"Update python testing as per zed cycle teting runtime","status":"MERGED","created":"2022-05-11 00:10:33.000000000","updated":"2022-05-19 21:50:02.000000000","submitted":"2022-05-19 21:47:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"841350-drop-py3-6-7","meta_rev_id":"7a1c85d892730663629165e301157399a49da7aa","_number":841350,"virtual_id_number":841350,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-19 21:47:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2022-05-19 19:37:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":2,"date":"2022-05-19 18:19:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2022-05-19 19:37:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"}],"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-11 02:09:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-11 05:31:56.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2022-05-11 06:50:21.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"CC"},{"updated":"2022-05-11 12:09:57.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2022-05-11 13:41:40.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2022-05-19 18:19:43.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2022-05-19 19:37:55.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"}],"messages":[{"id":"b93522003eebb2c7ba8008ae35288e8f3a07ca93","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-05-11 00:10:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f379de75aaa899000b91202434081ad4fd5630f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 02:09:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ab81a7293e814af7988bb7c345011f9e : SUCCESS in 4m 50s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e24d80c1b4a44f08a2eccaef8c190b47 : SUCCESS in 4m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/944a1eae17e44f668cb8630b202eedd8 : SUCCESS in 1h 06m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/84fb54c516d9435c89203d47f1ffd055 : SUCCESS in 1h 21m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c3d5b2933e7c42458a3de6d328043dd6 : SUCCESS in 51m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89708521a01d481295eb31b49f5cf5f4 : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1d7eeccdd366406ea8adad1e19d996bb : SUCCESS in 4m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1d4595fcb8e1495d9a4f724423e339c8 : SUCCESS in 6m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0a4212ae70ab4433b13314ab8de6a29b : SUCCESS in 7m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44b3f4fc44834869a024aa826d4fc783 : SUCCESS in 5m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/80ccda15e1504f33a66b9145b004b6f4 : SUCCESS in 5m 36s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/2abc214f95c64c4ab93516864552883f : FAILURE in 6m 44s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ec65695598324970907d0651c4045483 : SUCCESS in 10m 55s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/37871124a5364616b7ea800da0dba5a2 : SUCCESS in 8m 44s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/0959629b59d24bc38afb3c6ecefb1758 : FAILURE in 1h 40m 15s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b5dc63c221394853a370012c84243e8c : SUCCESS in 1h 35m 42s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f0959276892d44cf9fea3e0eba3fcba9 : SUCCESS in 1h 21m 04s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/773ce6f4105a4e48a7880eea777be5e0 : SUCCESS in 1h 51m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/274f31351aa04bba97e39ad35b8082f3 : SUCCESS in 41m 00s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5d4ee7e63f9343379716c3ef23e47b37 : SUCCESS in 1h 14m 06s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2a3daac78ee848a4ae1912a55916c2cc : SUCCESS in 43m 20s","accounts_in_message":[],"_revision_number":1},{"id":"64af7babc3569b9db646ae15e187d57b4aab5220","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-05-11 05:31:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8b954418a776f8f4e7e24d687744a2f4113fc08c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2022-05-11 05:38:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8053e68ce73f933bb392113ef1cb2f41c6ebb164","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 05:38:46.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/glance on branch master.  The error was:\n\n  Job openstack-tox-functional-py38-fips not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - check-requirements\n        - integrated-gate-storage\n        - openstack-python3-zed-jobs\n        - periodic-stable-jobs\n        - publish-openstack-docs-pti\n        - release-notes-jobs-python3\n      check:\n        jobs:\n          - openstack-tox-functional-py38-fips\n          - openstack-tox-functional-py39\n          - glance-tox-functional-py39-rbac-defaults\n          - glance-code-constants-check\n          - glance-ceph-thin-provisioning:\n              voting: false\n              irrelevant-files: \u0026tempest-irrelevant-files\n                - ^(test-|)requirements.txt$\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^glance/hacking/.*$\n                - ^glance/locale/.*$\n                - ^glance/tests/.*$\n                - ^doc/.*$\n                - ^releasenotes/.*$\n                - ^tools/.*$\n                - ^tox.ini$\n                - ^\\.zuul\\.yaml$\n          - tempest-integrated-storage:\n              irrelevant-files: *tempest-irrelevant-files\n          - tempest-integrated-storage-import:\n              irrelevant-files: *tempest-irrelevant-files\n          - tempest-integrated-storage-import-standalone:\n              irrelevant-files: *tempest-irrelevant-files\n          - glance-multistore-cinder-import:\n              irrelevant-files: *tempest-irrelevant-files\n          - grenade:\n              irrelevant-files: *tempest-irrelevant-files\n          - tempest-ipv6-only:\n              irrelevant-files: *tempest-irrelevant-files\n          - nova-ceph-multistore:\n              irrelevant-files: *tempest-irrelevant-files\n          - glance-secure-rbac-protection-functional\n  \n      gate:\n        jobs:\n          - openstack-tox-functional-py38\n          - openstack-tox-functional-py39\n          - tempest-integrated-storage:\n              irrelevant-files: *tempest-irrelevant-files\n          - tempest-integrated-storage-import:\n              irrelevant-files: *tempest-irrelevant-files\n          - tempest-integrated-storage-import-standalone:\n              irrelevant-files: *tempest-irrelevant-files\n          - grenade:\n              irrelevant-files: *tempest-irrelevant-files\n          - grenade-skip-level:\n              irrelevant-files: *tempest-irrelevant-files\n          - tempest-ipv6-only:\n              irrelevant-files: *tempest-irrelevant-files\n          - nova-ceph-multistore:\n              irrelevant-files: *tempest-irrelevant-files\n          - glance-secure-rbac-protection-functional\n      experimental:\n        jobs:\n          - glance-tox-py39-glance_store-tips\n          - glance-tox-py38-glance_store-tips\n          - glance-tox-functional-py39-glance_store-tips\n          - glance-tox-functional-py38-glance_store-tips\n          - barbican-tempest-plugin-simple-crypto\n          - grenade-multinode\n          - tempest-pg-full:\n              irrelevant-files: *tempest-irrelevant-files\n          - glance-multistore-cinder-import-fips:\n              irrelevant-files: *tempest-irrelevant-files\n      periodic:\n        jobs:\n          # NOTE(rosmaita): we only want the \"tips\" jobs to be run against\n          # master, hence the \u0027branches\u0027 qualifiers below.  Without them, when\n          # a stable branch is cut, the tests would be run against the stable\n          # branch as well, which is pointless because these libraries are\n          # frozen (more or less) in the stable branches.\n          #\n          # The \"tips\" jobs can be removed from the stable branch .zuul.yaml\n          # files if someone is so inclined, but that would require manual\n          # maintenance, so we do not do it by default.  Another option is\n          # to define these jobs in the openstack/project-config repo.\n          # That would make us less agile in adjusting these tests, so we\n          # aren\u0027t doing that either.\n          - glance-tox-functional-py39-oslo-tips:\n              branches: master\n          - glance-tox-functional-py38-oslo-tips:\n              branches: master\n          - glance-tox-py39-keystone-tips:\n              branches: master\n          - glance-tox-py38-keystone-tips:\n              branches: master\n          - glance-tox-functional-py39-keystone-tips:\n              branches: master\n          - glance-tox-functional-py38-keystone-tips:\n              branches: master\n          - glance-tox-py39-glance_store-tips:\n              branches: master\n          - glance-tox-py38-glance_store-tips:\n              branches: master\n          - glance-tox-functional-py39-glance_store-tips:\n              branches: master\n          - glance-tox-functional-py38-glance_store-tips:\n              branches: master\n          - glance-tox-py39-cursive-tips:\n              branches: master\n          - glance-tox-py38-cursive-tips:\n              branches: master\n          - glance-tox-functional-py39-cursive-tips:\n              branches: master\n          - glance-tox-functional-py38-cursive-tips:\n              branches: master\n          - glance-multistore-cinder-import-fips:\n              branches: master\n\n  in \"openstack/glance/.zuul.yaml@master\", line 310, column 3","accounts_in_message":[],"_revision_number":2},{"id":"0533fe4a2c3870dd9292a0b266866932be22b22b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2022-05-11 06:41:55.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"4e0e8ec0f5ec873032f35ccf25e618ba16ba8c28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 06:42:33.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":3},{"id":"2fd767b59d41c25f0ae1ce3eb7d780007ca026cd","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2022-05-11 06:50:21.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"aedbdc6376802957f9c52818c9d3e7b7a9790a6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 08:34:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ee06a5b2b1d74e939e815edfd035910d : SUCCESS in 3m 00s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/00ede3af5936488c8af8938de2b17e9a : SUCCESS in 4m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/3c13e785140346a7bc357fd062315739 : SUCCESS in 1h 01m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/53b0bad15b6841c9be9a9568edc5c7cc : SUCCESS in 53m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/399728a4420d41ea8da5d23b8de2c921 : SUCCESS in 43m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c88898cb82964c458e2a4b9c093126b7 : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f6bf4e8b079041d1810d8872720becea : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2162df202f7e4cc7bb194988422b4d2f : SUCCESS in 6m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/221d1f1aa965445b9da44b5da3f5cca2 : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad6d385befe84445a9598575fb4e0c1f : SUCCESS in 5m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c8633a1ed69a41e5990d85850829eae6 : SUCCESS in 5m 39s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/83f5fcf879924bad9ba013ae5427bb1d : SUCCESS in 10m 37s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/65dbec85c91144b5bde110e17886279f : SUCCESS in 8m 47s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/dcacc539816e4055916c20ed283cc303 : SUCCESS in 9m 49s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/0379c0558a7e44298bb791b577f4131e : FAILURE in 1h 42m 14s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/5dcfdaa3c2624fc5ba81d0399303d864 : SUCCESS in 1h 11m 09s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/8b55dd32216c4867b10391ca7c205514 : SUCCESS in 1h 11m 47s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/527607fc95f34e33be2cfea6e1094c27 : SUCCESS in 1h 31m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/09e84a8324024ed9ae4de5ba59d7b328 : SUCCESS in 59m 15s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/91f5f2d1b6484755a64b75398f7e6289 : SUCCESS in 1h 20m 19s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ac0e5d018edd4a2aabf79b46e3d5a877 : SUCCESS in 33m 48s","accounts_in_message":[],"_revision_number":3},{"id":"776ff022b9730311732924ed7d3d75a48d5c1582","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2022-05-11 12:09:57.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"838d3bbf0820f7ad41e8f67ef26f96b0297f3229","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-05-11 13:41:40.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"b8b2c7c9078b4f72c5cbd9a2a2d90b894bffbbfd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 13:41:53.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n\nWarning:\n  Change 841368 in project openstack/openstack-zuul-jobs does not share a change queue with 841350 in project openstack/glance","accounts_in_message":[],"_revision_number":3},{"id":"f5d91b38862cb6486fcf0847cc1a7dcd019cdf2f","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-05-11 13:51:21.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"04ac5fa50b88d73c80c3ff27bb094a16af983201","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-05-19 17:05:32.000000000","message":"Abandoned\n\nThis patch is not required any more as we decided to drop py36 jobs for now and this patch also has a dependency which is not merged yet.\n\nWe are going forward with, https://review.opendev.org/c/openstack/glance/+/842549 to resolve gate conflicts for glance.\n\nThank you for your efforts!","accounts_in_message":[],"_revision_number":3},{"id":"13e9fb0739048b034586f5ddb24944f55c63147a","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-05-19 17:25:29.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"14a2827fb22d206467d63e24d184b4d20d5b3916","tag":"autogenerated:gerrit:restore","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-05-19 17:40:50.000000000","message":"Restored","accounts_in_message":[],"_revision_number":3},{"id":"b75a0983e7971552a1ea3e2a2cc4e031d7ae551f","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-05-19 17:40:56.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"fc47fe521a08310a6b751c26cdec228db97d8b56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-05-19 17:57:47.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e410006bbed265ccfbee0ddce4222d9404f8e151","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-05-19 18:19:43.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"89380cf8e86882f7fde2a3a04465eb47b7473276","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-19 19:36:58.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d3b6909a19424e8788aae22117273ff8 : SUCCESS in 2m 15s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/25dfe5c4dd7e4e1c992937dbce5523ad : SUCCESS in 5m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/d15302a1d3864362b885978ca0909697 : SUCCESS in 1h 09m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bc47730728fb4039a6fd3db5be0acebb : SUCCESS in 1h 30m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/92d4b90a50814a7f8317e18d840747a5 : SUCCESS in 53m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/868686d1c5234ea9aefa4a1edd461b24 : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4c3907d87dd841209d5c1ef009a2edbd : SUCCESS in 6m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/139dd7f342a3403eafc1b4cdb0bb812c : SUCCESS in 5m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/07dc1313623f4c4ca18e048423de4386 : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c08ae621d47467f9f8c4fdacc9663b1 : SUCCESS in 6m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0c08c1de65dd459c9d507763f62f4f05 : SUCCESS in 6m 18s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/08f23c15e34a4183bc6c440e1e030166 : SUCCESS in 10m 11s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/a5a4d985c8d84dd29264f27f361d7d45 : SUCCESS in 12m 31s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/919b4244b715406499f982f27d717812 : SUCCESS in 1h 26m 05s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/9d13107079cf4c6f8834a1620cabac5b : SUCCESS in 1h 22m 36s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/7d9278c779df49c18709960c7beb6fa1 : SUCCESS in 1h 19m 29s\n- glance-multistore-cinder https://zuul.opendev.org/t/openstack/build/ba7ce78e495f4184ba60ac0d3963c216 : SUCCESS in 1h 30m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ef8d55ab08cd44f4957d6d91be697023 : SUCCESS in 52m 01s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/051ba33436d4457aae75e863442cd3df : SUCCESS in 1h 04m 36s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/51d01e69ccec422dbe7cbc2a3b2692a0 : SUCCESS in 36m 10s","accounts_in_message":[],"_revision_number":4},{"id":"fb6e105f1487c5fe23af33ad3bde2c8f5f4b614c","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-05-19 19:37:55.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"da51dd149ccbf822f4a61d7f51bd8ebe7b61435d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-19 19:39:01.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"bae678784c2729dfa13e66ac74e019ff4016cdb7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-19 21:47:33.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/91155ca811eb45dba13a585963506758 : SUCCESS in 2m 53s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/cbee76567a1e40a390badb499af654a2 : SUCCESS in 4m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/dc94c30e992442968f5e646a6a74d369 : SUCCESS in 1h 15m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/613632508ad14a7e9a8b2a7bdd82b61a : SUCCESS in 1h 18m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0c5519b5fd274780904ce9869f624907 : SUCCESS in 47m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6fb3ab8d03c4469da4872ad23af01979 : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/858528d574a741b280bd9f5418968456 : SUCCESS in 5m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4abe82be002346628144b6ecf3826b8a : SUCCESS in 6m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfdf81aadd9a48a999ad1fd5504b4d6d : SUCCESS in 7m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f324cc1079d34cb59d67c4af74f71d5e : SUCCESS in 6m 51s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4d8d4a6d0f6b4239b0655efc190bb2be : SUCCESS in 9m 12s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/63e83602f51b44518fd2d7c825f9c286 : SUCCESS in 10m 33s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/44a14f1853b34a108c6e11cc39252ab4 : SUCCESS in 1h 19m 30s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/88101ac0e537407798ee616d6161e9ee : SUCCESS in 59m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a6f343c8aeed465e89ad2b299bbcc7bf : SUCCESS in 51m 18s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/bc8014880e94458a82cc8771cad89b63 : SUCCESS in 1h 14m 52s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/04232d675f8944438888198e9495d95a : SUCCESS in 32m 36s","accounts_in_message":[],"_revision_number":4},{"id":"edc5e79b61f849b621ab2389819f84ebed6891c8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-19 21:47:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"7a1c85d892730663629165e301157399a49da7aa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-19 21:50:02.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/0e6335b84c144aae92e31dea2d852640 : SUCCESS in 57s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89ce3e7e154c4c2686e9fb64954a746a : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3173a74095e74f3fb2442e916ccd1d05 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"d55d250d336edb1892f0f40cf8be6063be3315a4","revisions":{"e07f81f825942bd56a057c15953973a06e253322":{"kind":"REWORK","_number":1,"created":"2022-05-11 00:10:33.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/50/841350/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/50/841350/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/50/841350/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/50/841350/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/50/841350/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/50/841350/1"}}},"commit":{"parents":[{"commit":"331ce59a50524e65630aed2d888af30f4494563d","subject":"Merge \"Fix tests for logging connection info\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/331ce59a50524e65630aed2d888af30f4494563d"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-05-11 00:04:16.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-05-11 00:10:07.000000000","tz":-300},"subject":"Update python testing as per zed cycle teting runtime","message":"Update python testing as per zed cycle teting runtime\n\nIn Zed cycle, we have dropped the python 3.6/3.7[1] testing\nand its support. Moving the py36 job to py38 based as well as\nupdating the python classifier also to reflect the same.\n\nopenstack-tox-functional-py36-fips job is left which can be moved\nto py38|py39 based once that job is defined in openatck-zuul-jobs\nrepo.\n\n[1] https://governance.openstack.org/tc/reference/runtimes/zed.html\n\nChange-Id: Id0813d9dc553dd424732079039349b42f6f3201b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e07f81f825942bd56a057c15953973a06e253322"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e07f81f825942bd56a057c15953973a06e253322"}]},"branch":"refs/heads/master"},"cfd83c1d9b486bf7d6ad3fbce9c7065693869cfb":{"kind":"REWORK","_number":2,"created":"2022-05-11 05:38:15.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/50/841350/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/50/841350/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/50/841350/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/50/841350/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/50/841350/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/50/841350/2"}}},"commit":{"parents":[{"commit":"331ce59a50524e65630aed2d888af30f4494563d","subject":"Merge \"Fix tests for logging connection info\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/331ce59a50524e65630aed2d888af30f4494563d"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-05-11 00:04:16.000000000","tz":-300},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2022-05-11 05:22:35.000000000","tz":0},"subject":"Update python testing as per zed cycle teting runtime","message":"Update python testing as per zed cycle teting runtime\n\nIn Zed cycle, we have dropped the python 3.6/3.7[1] testing\nand its support. Moving the py36 job to py38 based as well as\nupdating the python classifier also to reflect the same.\n\nopenstack-tox-functional-py36-fips job is left which can be moved\nto py38|py39 based once that job is defined in openatck-zuul-jobs\nrepo.\n\n[1] https://governance.openstack.org/tc/reference/runtimes/zed.html\n\nChange-Id: Id0813d9dc553dd424732079039349b42f6f3201b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cfd83c1d9b486bf7d6ad3fbce9c7065693869cfb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cfd83c1d9b486bf7d6ad3fbce9c7065693869cfb"}]},"branch":"refs/heads/master"},"0cc93d96325af3b6ca7a0659c0539a4d1c563a48":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-05-11 06:41:55.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/50/841350/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/50/841350/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/50/841350/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/50/841350/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/50/841350/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/50/841350/3"}}},"commit":{"parents":[{"commit":"331ce59a50524e65630aed2d888af30f4494563d","subject":"Merge \"Fix tests for logging connection info\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/331ce59a50524e65630aed2d888af30f4494563d"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-05-11 00:04:16.000000000","tz":-300},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2022-05-11 06:24:47.000000000","tz":0},"subject":"Update python testing as per zed cycle testing runtime","message":"Update python testing as per zed cycle testing runtime\n\nIn Zed cycle, we have dropped the python 3.6/3.7[1] testing\nand its support. Moving the py36 job to py38 based as well as\nupdating the python classifier also to reflect the same.\n\n[1] https://governance.openstack.org/tc/reference/runtimes/zed.html\n\nDepends-On: I3be69c3b42e9920e0de0299ae7e97026cdfd12e3\nChange-Id: Id0813d9dc553dd424732079039349b42f6f3201b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0cc93d96325af3b6ca7a0659c0539a4d1c563a48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0cc93d96325af3b6ca7a0659c0539a4d1c563a48"}]},"branch":"refs/heads/master"},"d55d250d336edb1892f0f40cf8be6063be3315a4":{"kind":"REWORK","_number":4,"created":"2022-05-19 17:57:47.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/50/841350/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/50/841350/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/50/841350/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/50/841350/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/50/841350/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/50/841350/4"}}},"commit":{"parents":[{"commit":"331ce59a50524e65630aed2d888af30f4494563d","subject":"Merge \"Fix tests for logging connection info\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/331ce59a50524e65630aed2d888af30f4494563d"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-05-11 00:04:16.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-05-19 17:56:56.000000000","tz":-300},"subject":"Update python testing as per zed cycle teting runtime","message":"Update python testing as per zed cycle teting runtime\n\nIn Zed cycle, we have dropped the python 3.6/3.7[1] testing\nand its support. Moving the py36 job to py38 based as well as\nupdating the python classifier also to reflect the same.\n\nopenstack-tox-functional-py36-fips job is left which can be moved\nto py38|py39 based once that job is defined in openatck-zuul-jobs\nrepo.\n\nopenstack-tox-functional-py36-fips job will be migrated to py38 or py39\nin followup patch as that need openstack-zuul-config changes too.\n\n[1] https://governance.openstack.org/tc/reference/runtimes/zed.html\n\nChange-Id: Id0813d9dc553dd424732079039349b42f6f3201b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d55d250d336edb1892f0f40cf8be6063be3315a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d55d250d336edb1892f0f40cf8be6063be3315a4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
