)]}'
{"id":"openstack%2Fglance~841548","triplet_id":"openstack%2Fglance~master~I524dfeb05c078773aa77020d4a6a9991a7eb75c2","project":"openstack/glance","branch":"master","topic":"bug/1973136","attention_set":{},"removed_from_attention_set":{"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2022-05-14 03:59:22.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I524dfeb05c078773aa77020d4a6a9991a7eb75c2","subject":"Disable import workflow in glance cinder jobs","status":"MERGED","created":"2022-05-12 07:05:20.000000000","updated":"2022-05-23 08:17:55.000000000","submitted":"2022-05-14 03:59:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"841548-bug/1973136","meta_rev_id":"1807acfc90e1efeb43d5400b632e2e08e8bcd6a1","_number":841548,"virtual_id_number":841548,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-14 03:59:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2022-05-14 01:10:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2022-05-13 07:35:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-05-14 03:00:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2022-05-14 03:04:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-12 08:54:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-13 07:35:17.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2022-05-14 01:10:09.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2022-05-14 03:00:27.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"}],"messages":[{"id":"218f7fea82edcbf08f61cb9cdd57b10db11e162a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-05-12 07:05:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e4e4c8918d5658ae8f9913560fe483d4ed1beaeb","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-05-12 07:09:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"91acbb3258f845cf668241394526e6c5b912bc62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-12 08:54:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b72f6c81a1cc41e68bbba24b26b8c6b3 : SUCCESS in 56m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54e1fa9503a144b8abf8013b73a48ed0 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/39de721248c74fe0b73db77b323bc80d : SUCCESS in 5m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eb4c5f2f5e634d8e9a5a0f0f06bf5934 : SUCCESS in 6m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e738fc00b1dc411e9a6dd7f9c6a05e7f : SUCCESS in 5m 59s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/d2eef46da8a8479c83035c2e81b9f702 : SUCCESS in 12m 50s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a50d6921b49d48e59474f96745ac52ee : SUCCESS in 9m 15s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/e3cc3da8a78749fe85f71746f44f46aa : SUCCESS in 10m 09s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/eaeddac931f44970bcfc89c55608b3bf : POST_FAILURE in 1h 44m 15s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/97160ba72c684d139d38570655f2dd14 : SUCCESS in 41m 48s","accounts_in_message":[],"_revision_number":1},{"id":"1080aa682e57312d29f5ab456aaf0c15271f7f9e","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-05-12 09:02:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e0fc0c335e92b834fa9302c50851e4587a06d835","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-12 10:19:08.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ae3feebd8641429d86aa7557b5aabd90 : SUCCESS in 47m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d34be600d43b494cb25f6f41d5d43554 : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/df9345719739452a928c5ad9ac3a2501 : SUCCESS in 5m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b0a4388cb95444548b432528042559b5 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f52a9cc2c3fe4962ba29f4dddd9717fc : SUCCESS in 6m 07s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/a1935eb1a55f4b1497d74700b7a46d3c : SUCCESS in 9m 39s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c6017eb3cd6844a4a8342955138d1a76 : SUCCESS in 7m 16s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/8909c7520378459cb8496cf94e78dacc : SUCCESS in 8m 14s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8228e3ca6bf24eef92a63a8416589ae9 : POST_FAILURE in 1h 12m 43s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8b48cf71fcdf425fbc6f766d63f9e337 : SUCCESS in 24m 34s","accounts_in_message":[],"_revision_number":1},{"id":"ce4217bc993b1c002f9ece6d1f579bfdb887bc83","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-05-12 12:16:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ac8c9e3f55d9b237cc4d9cec1e9ccc9897777900","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-05-13 05:59:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b240811a2fd568459157293b1c434fbf6f46ab1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-13 07:22:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/96d553f4d84c489089373484f12e0c23 : SUCCESS in 58m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8da1baf0794c45f28b8517c48ad86427 : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/324c92a7ff274b45a41ea4786b00cddc : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d7c8179e396e4df78d51955eb0f62cf4 : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/195f57320a704651886e8eb37a7448b6 : SUCCESS in 6m 02s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/3456feeac5e84466ac974887bd4e3c84 : SUCCESS in 10m 56s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7d3193aa34fb4c03812203655b720dc6 : SUCCESS in 8m 21s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/b84ba611adc843f59360af5c8aea9a4f : SUCCESS in 10m 23s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/6bc5289e3de542a398095fc5432e2f81 : SUCCESS in 1h 21m 45s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/9ebdf421771744ae9c56fdb8a6d9d6fb : SUCCESS in 36m 17s","accounts_in_message":[],"_revision_number":2},{"id":"03bbe419bf299925c50062fbde15461adb5ff25a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-05-13 07:35:17.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e15219fb89a7bcea1355aab781f18ffc2324c03e","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-05-14 01:10:09.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"74ba354c978062f9d930352b69d864cbb21a6f6d","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2022-05-14 03:00:27.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"9a18cf1de7a8f3c61b82a397fc642a25a69913f6","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-05-14 03:04:16.000000000","message":"Patch Set 2: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bb4a1a24b99e3ca0ecff7e914730209eccdc9a5e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-14 03:04:57.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"dbc7d23a32385746cff4363d057dec3f596ae5e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-14 03:59:21.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/08bfcfc7ba9c4c3ba7d1ae5050db8845 : SUCCESS in 50m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65935cc03a194dcc9e70ea1041e39f84 : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c4a4a5a8921745bfbecc1c8bf7ba7258 : SUCCESS in 5m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/287aed60ec9d43f68ec275abef3f26cc : SUCCESS in 6m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/575b3d6192364651bea676c402cb914f : SUCCESS in 6m 07s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cfe57f5177944a10ac4302ce909489cb : SUCCESS in 8m 40s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e89bf2eb783c47d7b93d01b49fcbd4b8 : SUCCESS in 9m 33s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/227fa976f1754a02a9f6b89a3ec57b95 : SUCCESS in 32m 49s","accounts_in_message":[],"_revision_number":2},{"id":"4bf9b1541ff608ed435ce3d0ac926149938bd4e2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-14 03:59:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"b28b4947b2859e23949bed18f581de209af712e9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-14 04:01:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f72086af7bb4482bbff79289efe54ba : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":2},{"id":"09dd6b4fafd583794b63000b9d37f47b706a9c16","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-05-16 04:22:00.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"28e1390b6cddd29795a179ee3a9fb680c76747af","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-05-17 12:56:47.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1807acfc90e1efeb43d5400b632e2e08e8bcd6a1","tag":"autogenerated:gerrit:revert","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-05-23 08:17:55.000000000","message":"Created a revert of this change as I214c9a6017f66d3eb6589496726e8c8f895d56aa","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d7fa7a0321ea5a56ec130aa0bd346749459ccaf2","revisions":{"9e3657114e545d5881d451889479e6656aa31c40":{"kind":"REWORK","_number":1,"created":"2022-05-12 07:05:20.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/48/841548/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/48/841548/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/48/841548/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/48/841548/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/48/841548/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/48/841548/1"}}},"commit":{"parents":[{"commit":"331ce59a50524e65630aed2d888af30f4494563d","subject":"Merge \"Fix tests for logging connection info\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/331ce59a50524e65630aed2d888af30f4494563d"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2022-05-12 06:54:06.000000000","tz":330},"committer":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2022-05-12 07:05:16.000000000","tz":330},"subject":"Disable import workflow in glance cinder jobs","message":"Disable import workflow in glance cinder jobs\n\nRecently, glance-multistore-cinder-import job started failing.\nAs per the RCA done here[1], the reason is glance is using\nimport workflow to create images which is an async operation.\nAs in case of glance cinder configuration, there are a lot of\nexternal APIs (cinder) called like volume create, attachment\ncreate, attachment update, attachment delete etc which takes\ntime to process hence the image doesn\u0027t get available in the\nexpected time (as per devstack) hence the failure.\nDisabling import workflow will cause the images to be created\nsynchronously which should pass the glance cinder jobs.\n\n[1] https://review.opendev.org/c/openstack/glance/+/841278/1#message-456096e48b28e5b866deb8bf53e9258ee08219a0\n\nCloses-Bug: 1973136\nChange-Id: I524dfeb05c078773aa77020d4a6a9991a7eb75c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9e3657114e545d5881d451889479e6656aa31c40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9e3657114e545d5881d451889479e6656aa31c40"}]},"branch":"refs/heads/master"},"d7fa7a0321ea5a56ec130aa0bd346749459ccaf2":{"kind":"REWORK","_number":2,"created":"2022-05-13 05:59:19.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/48/841548/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/48/841548/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/48/841548/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/48/841548/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/48/841548/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/48/841548/2"}}},"commit":{"parents":[{"commit":"331ce59a50524e65630aed2d888af30f4494563d","subject":"Merge \"Fix tests for logging connection info\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/331ce59a50524e65630aed2d888af30f4494563d"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2022-05-12 06:54:06.000000000","tz":330},"committer":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2022-05-13 05:56:52.000000000","tz":330},"subject":"Disable import workflow in glance cinder jobs","message":"Disable import workflow in glance cinder jobs\n\nRecently, glance-multistore-cinder-import job started failing.\nAs per the RCA done here[1], the reason is glance is using\nimport workflow to create images which is an async operation.\nAs in case of glance cinder configuration, there are a lot of\nexternal APIs (cinder) called like volume create, attachment\ncreate, attachment update, attachment delete etc which takes\ntime to process hence the image doesn\u0027t get available in the\nexpected time (as per devstack) hence the failure.\nDisabling import workflow will cause the images to be created\nsynchronously which should pass the glance cinder jobs.\nTo disable import workflow, we are inheriting from\ntempest-integrated-storage and not\ntempest-integrated-storage-import (which has import plugin enabled).\n\n[1] https://review.opendev.org/c/openstack/glance/+/841278/1#message-456096e48b28e5b866deb8bf53e9258ee08219a0\n\nCloses-Bug: 1973136\nChange-Id: I524dfeb05c078773aa77020d4a6a9991a7eb75c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d7fa7a0321ea5a56ec130aa0bd346749459ccaf2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d7fa7a0321ea5a56ec130aa0bd346749459ccaf2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
