)]}'
{"id":"openstack%2Fglance~841695","triplet_id":"openstack%2Fglance~master~I764a71623b79c32b10b638b0448b4bc02c2dc7bf","project":"openstack/glance","branch":"master","attention_set":{},"removed_from_attention_set":{"31412":{"account":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"last_update":"2022-05-19 14:19:38.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I764a71623b79c32b10b638b0448b4bc02c2dc7bf","subject":"Update python testing as per zed cycle testing runtime","status":"ABANDONED","created":"2022-05-13 08:58:00.000000000","updated":"2022-05-19 14:19:38.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fd24e03acd010b3098245ee5ab1ce1f825d16ad8","_number":841695,"virtual_id_number":841695,"owner":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-05-13 10:41:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-13 10:41:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e06ad998e5b259be82df6f3717842ccde49a1d09","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2022-05-13 08:58:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ed31ffa54b2845878286993aff35fc4c908c02de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-13 10:41:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3faf17b603b0487a83c6577493811670 : SUCCESS in 2m 50s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/52d6c9378efd4e47a8b53ce2f8013120 : SUCCESS in 5m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/0c59cf1daf9e470f828974b997672600 : SUCCESS in 56m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d7f95c1589d14369a4f842e9bd8f88d1 : SUCCESS in 1h 07m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6058292225b0459ea604f70346aa2b82 : SUCCESS in 53m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfebbedbaca043a18cd5fbdd946ab0a2 : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cefb9d7464b94b3c8ebc863534e4bd45 : FAILURE in 6m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b305156d890b4dd79ce780e76f36cb85 : SUCCESS in 6m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/054da1c0f4f94662869b227de0753486 : SUCCESS in 7m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8ca249ea907142bbb249fab36db77592 : SUCCESS in 5m 46s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c2665738392d4bbcbce3a6c9ed504899 : SUCCESS in 9m 44s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/328f49bb7ffb4a1fb281d2b2c8ed4171 : SUCCESS in 10m 29s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/2dd45ad8c56b48c2a8c7bd63e8d386a8 : SUCCESS in 1h 10m 05s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/59a9166220fe46eda681958965212454 : SUCCESS in 1h 09m 07s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/5b922dac50754af99ae1dda577ac80d8 : SUCCESS in 1h 12m 17s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/fc34d89ba81a44979f4aeac353c8588f : SUCCESS in 1h 37m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/31608f5ae0bd4a90ae59270ed26c99ca : SUCCESS in 56m 11s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/edff3c2ea27b4a948877c3d153a1d822 : SUCCESS in 1h 17m 09s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/295ea566e8f54d4fbe41f538cd869df6 : SUCCESS in 44m 36s","accounts_in_message":[],"_revision_number":1},{"id":"fd24e03acd010b3098245ee5ab1ce1f825d16ad8","tag":"autogenerated:gerrit:abandon","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-05-19 14:19:38.000000000","message":"Abandoned\n\nDuplicates https://review.opendev.org/c/openstack/glance/+/841350","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"999779e09380d9ad6e7970fa90561c8c50a10f7c","revisions":{"999779e09380d9ad6e7970fa90561c8c50a10f7c":{"kind":"REWORK","_number":1,"created":"2022-05-13 08:58:00.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/95/841695/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/95/841695/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/841695/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/841695/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/841695/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/95/841695/1"}}},"commit":{"parents":[{"commit":"331ce59a50524e65630aed2d888af30f4494563d","subject":"Merge \"Fix tests for logging connection info\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/331ce59a50524e65630aed2d888af30f4494563d"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2022-05-13 08:56:15.000000000","tz":480},"committer":{"name":"songwenping","email":"songwenping@inspur.com","date":"2022-05-13 08:57:54.000000000","tz":480},"subject":"Update python testing as per zed cycle testing runtime","message":"Update python testing as per zed cycle testing runtime\n\nIn Zed cycle, we have dropped the python 3.6/3.7[1] testing\nand its support. Removing the py36 centos8 job as well as\nupdating the python classifier also to reflect the same.\n\n[1] https://governance.openstack.org/tc/reference/runtimes/zed.html\n\nChange-Id: I764a71623b79c32b10b638b0448b4bc02c2dc7bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/999779e09380d9ad6e7970fa90561c8c50a10f7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/999779e09380d9ad6e7970fa90561c8c50a10f7c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
