)]}'
{"id":"openstack%2Fglance~841747","triplet_id":"openstack%2Fglance~stable%2Fyoga~I524dfeb05c078773aa77020d4a6a9991a7eb75c2","project":"openstack/glance","branch":"stable/yoga","topic":"bug/1973136-stable/yoga","attention_set":{},"removed_from_attention_set":{"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2022-05-23 08:17:03.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I524dfeb05c078773aa77020d4a6a9991a7eb75c2","subject":"Disable import workflow in glance cinder jobs","status":"ABANDONED","created":"2022-05-16 04:17:31.000000000","updated":"2022-05-23 08:17:03.000000000","total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":841548,"cherry_pick_of_patch_set":2,"meta_rev_id":"d68e060bfe792f85e31df5e34e0d374a55b0dc3b","_number":841747,"virtual_id_number":841747,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-05-17 17:36:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":-1,"date":"2022-05-20 02:45:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-16 05:20:23.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2022-05-16 05:39:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-20 02:45:49.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"}],"messages":[{"id":"e5500b55d328a01f11ffb491f79de2b57ada3ad1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-05-16 04:17:31.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"9de8794b6d2dcdebdd8b80fbd545521ef7b33748","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-05-16 05:20:23.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"622babc16af2bdd22fc73b2c4b6e85dbc1d99d17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-16 05:39:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/7c8e385d790d45f881606618f340c8ad : SUCCESS in 1h 09m 55s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6ab7aad0fc73452fa6de6521af221fa1 : SUCCESS in 1h 03m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c05881eb135f49bc974f7f210c0778bd : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d5a42c8f8c540558c8c0245e56c042b : SUCCESS in 4m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b603190c096a40d8aed0dd991b5d5043 : SUCCESS in 5m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e919c3a9ede4c3c8f7dee897da3e3c0 : SUCCESS in 6m 42s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/2be1236d5af8497daa58a2a0e3dc2652 : SUCCESS in 9m 17s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/274ccff9293f459ea479548b2c208f0b : SUCCESS in 7m 16s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/20cae1e585604df493c490fabe571015 : SUCCESS in 7m 39s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/57ca65dead374e73b601a8fd3f4ece45 : FAILURE in 19m 23s\n- glance-multistore-cinder-import-fips https://zuul.opendev.org/t/openstack/build/901fd3714efb464b8cd45fb2de5cad9f : FAILURE in 1h 14m 36s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2323b50997af48d8a55523365dd232a7 : SUCCESS in 23m 24s","accounts_in_message":[],"_revision_number":1},{"id":"a1c91307ddb88a42a2a19c69e0a1ad8ddd8f3122","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-05-17 15:57:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a044dae9d310e12f06a1040fc10bbaca92850623","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-05-17 16:04:23.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a6a9564e4e44740e29bfe769a32741d69dba0f05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-17 17:36:58.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/953a866015e441ce9c925a29dbeae781 : SUCCESS in 59m 09s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/16820a77a09f4613993d5a968879f9f1 : SUCCESS in 47m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14bf709b333e404bb7cd58f7bdb1c6a7 : SUCCESS in 6m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e2065de4d0194ff7a4cd490e883bc9e1 : SUCCESS in 5m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0c910ba8ccd342169d963d504044972f : SUCCESS in 7m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7897866857994dceba7ea8acc282de83 : SUCCESS in 6m 32s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/0a45d2231e754dd7a12cbb5307d897ad : SUCCESS in 11m 02s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/16e82e9d96b64c21a0af313033ddda16 : FAILURE in 16m 35s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/5aee7d8fb3a94c92b4dc29cbc279b187 : SUCCESS in 17m 51s (non-voting)\n- glance-multistore-cinder https://zuul.opendev.org/t/openstack/build/05e682abf63d454baa4c0206a4cc15a6 : SUCCESS in 1h 29m 27s\n- glance-multistore-cinder-fips https://zuul.opendev.org/t/openstack/build/34e2e2a82d0447e89924f388a1474535 : FAILURE in 1h 22m 34s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6ff425bd90084d7290a3dd232ed12164 : SUCCESS in 37m 41s","accounts_in_message":[],"_revision_number":3},{"id":"bc14ff4f171e05a8f76e2083013f1080ae16d9ec","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-05-17 17:42:09.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"727da1cfcb82f8f6580a64b1b284bd31ad7cccb9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-17 19:11:11.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/5d25c7b77e9d4edb8b1abe1c578baaca : SUCCESS in 48m 26s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/248d277c36184211811520b842d61062 : SUCCESS in 39m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2da831411d44428db3f26f839a9b18ec : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f971890fb1f94487b45ef8c42e52c5c4 : SUCCESS in 5m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ba58fae9101344959d4631ba8452efb8 : SUCCESS in 6m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d13a42f1d4a549499950cba8c9757550 : SUCCESS in 5m 54s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/c2782c52dd014cd999b561d5e8e59be8 : SUCCESS in 10m 40s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7259364c439f4aa9ac7988c7413d8223 : SUCCESS in 9m 24s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/6fd83ee4d58041d68bd39dfce17a4f8c : SUCCESS in 9m 13s (non-voting)\n- glance-multistore-cinder https://zuul.opendev.org/t/openstack/build/dc0b70ce85d944f9bafe8b8ad3a91019 : SUCCESS in 1h 10m 01s\n- glance-multistore-cinder-fips https://zuul.opendev.org/t/openstack/build/cc7e3e08366342f4a0ef3bd7033d9e4f : FAILURE in 1h 21m 35s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/49c6d872553a477ebae09c0ff58c969e : FAILURE in 26m 19s","accounts_in_message":[],"_revision_number":3},{"id":"5a9a594ad492804b1de974437b01255aca6f0d73","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-05-18 06:16:43.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"398b91bfd9918008e24dc5656a76ffc7d2b1aca2","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-05-20 02:45:49.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d68e060bfe792f85e31df5e34e0d374a55b0dc3b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-05-23 08:17:03.000000000","message":"Abandoned\n\nIn favor of https://review.opendev.org/c/openstack/devstack/+/842390","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"40583bbef9fd5610cf91ba7c0ea59d5a6d932f05","revisions":{"f52a71da1d18b8dae2f90149f58a89de09bbe998":{"kind":"REWORK","_number":1,"created":"2022-05-16 04:17:31.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/47/841747/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/841747/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/841747/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/841747/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/841747/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/841747/1"}}},"commit":{"parents":[{"commit":"1beee78e096ddba4729be10af6ac4ddec8e79da0","subject":"Update TOX_CONSTRAINTS_FILE for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1beee78e096ddba4729be10af6ac4ddec8e79da0"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2022-05-12 06:54:06.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2022-05-16 04:17:31.000000000","tz":0},"subject":"Disable import workflow in glance cinder jobs","message":"Disable import workflow in glance cinder jobs\n\nRecently, glance-multistore-cinder-import job started failing.\nAs per the RCA done here[1], the reason is glance is using\nimport workflow to create images which is an async operation.\nAs in case of glance cinder configuration, there are a lot of\nexternal APIs (cinder) called like volume create, attachment\ncreate, attachment update, attachment delete etc which takes\ntime to process hence the image doesn\u0027t get available in the\nexpected time (as per devstack) hence the failure.\nDisabling import workflow will cause the images to be created\nsynchronously which should pass the glance cinder jobs.\nTo disable import workflow, we are inheriting from\ntempest-integrated-storage and not\ntempest-integrated-storage-import (which has import plugin enabled).\n\n[1] https://review.opendev.org/c/openstack/glance/+/841278/1#message-456096e48b28e5b866deb8bf53e9258ee08219a0\n\nCloses-Bug: 1973136\nChange-Id: I524dfeb05c078773aa77020d4a6a9991a7eb75c2\n(cherry picked from commit d7fa7a0321ea5a56ec130aa0bd346749459ccaf2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f52a71da1d18b8dae2f90149f58a89de09bbe998"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f52a71da1d18b8dae2f90149f58a89de09bbe998"}]},"branch":"refs/heads/stable/yoga"},"0554aa6493f64a21a53e68bcd7b652e1ee342e19":{"kind":"REWORK","_number":2,"created":"2022-05-17 15:57:06.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/47/841747/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/841747/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/841747/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/841747/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/841747/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/841747/2"}}},"commit":{"parents":[{"commit":"1beee78e096ddba4729be10af6ac4ddec8e79da0","subject":"Update TOX_CONSTRAINTS_FILE for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1beee78e096ddba4729be10af6ac4ddec8e79da0"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2022-05-12 06:54:06.000000000","tz":330},"committer":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2022-05-17 15:57:01.000000000","tz":330},"subject":"Disable import workflow in glance cinder jobs","message":"Disable import workflow in glance cinder jobs\n\nRecently, glance-multistore-cinder-import job started failing.\nAs per the RCA done here[1], the reason is glance is using\nimport workflow to create images which is an async operation.\nAs in case of glance cinder configuration, there are a lot of\nexternal APIs (cinder) called like volume create, attachment\ncreate, attachment update, attachment delete etc which takes\ntime to process hence the image doesn\u0027t get available in the\nexpected time (as per devstack) hence the failure.\nDisabling import workflow will cause the images to be created\nsynchronously which should pass the glance cinder jobs.\nTo disable import workflow, we are inheriting from\ntempest-integrated-storage and not\ntempest-integrated-storage-import (which has post-run checks for import)\nand overriding GLANCE_USE_IMPORT_WORKFLOW\nto False in cinder jobs as the parent job tempest-integrated-storage\nalso has import plugin enabled[2].\nAlso removing -import from job names since it\u0027s not using\nimport plugin anymore.\n\nAlso includes squash of change\nChange-Id: I2775e007f942feed8fa6ae5e385c03992859edc1\n\n[1] https://review.opendev.org/c/openstack/glance/+/841278/1#message-456096e48b28e5b866deb8bf53e9258ee08219a0\n[2] https://opendev.org/openstack/tempest/src/branch/master/zuul.d/integrated-gate.yaml#L195\n\nCloses-Bug: 1973136\n\nChange-Id: I524dfeb05c078773aa77020d4a6a9991a7eb75c2\n(cherry picked from commit d7fa7a0321ea5a56ec130aa0bd346749459ccaf2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0554aa6493f64a21a53e68bcd7b652e1ee342e19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0554aa6493f64a21a53e68bcd7b652e1ee342e19"}]},"branch":"refs/heads/stable/yoga"},"40583bbef9fd5610cf91ba7c0ea59d5a6d932f05":{"kind":"REWORK","_number":3,"created":"2022-05-17 16:04:23.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/47/841747/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/841747/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/841747/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/841747/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/841747/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/841747/3"}}},"commit":{"parents":[{"commit":"1beee78e096ddba4729be10af6ac4ddec8e79da0","subject":"Update TOX_CONSTRAINTS_FILE for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1beee78e096ddba4729be10af6ac4ddec8e79da0"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2022-05-12 06:54:06.000000000","tz":330},"committer":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2022-05-17 15:59:55.000000000","tz":330},"subject":"Disable import workflow in glance cinder jobs","message":"Disable import workflow in glance cinder jobs\n\nRecently, glance-multistore-cinder-import job started failing.\nAs per the RCA done here[1], the reason is glance is using\nimport workflow to create images which is an async operation.\nAs in case of glance cinder configuration, there are a lot of\nexternal APIs (cinder) called like volume create, attachment\ncreate, attachment update, attachment delete etc which takes\ntime to process hence the image doesn\u0027t get available in the\nexpected time (as per devstack) hence the failure.\nDisabling import workflow will cause the images to be created\nsynchronously which should pass the glance cinder jobs.\nTo disable import workflow, we are inheriting from\ntempest-integrated-storage and not\ntempest-integrated-storage-import (which has post-run checks for import)\nand overriding GLANCE_USE_IMPORT_WORKFLOW\nto False in cinder jobs as the parent job tempest-integrated-storage\nalso has import plugin enabled[2].\nAlso removing -import from job names since it\u0027s not using\nimport plugin anymore.\n\nAlso includes squash of change\nChange-Id: I2775e007f942feed8fa6ae5e385c03992859edc1\n\n[1] https://review.opendev.org/c/openstack/glance/+/841278/1#message-456096e48b28e5b866deb8bf53e9258ee08219a0\n[2] https://opendev.org/openstack/tempest/src/branch/master/zuul.d/integrated-gate.yaml#L195\n\nConflicts:\n    .zuul.yaml\n\nCloses-Bug: 1973136\n\nChange-Id: I524dfeb05c078773aa77020d4a6a9991a7eb75c2\n(cherry picked from commit d7fa7a0321ea5a56ec130aa0bd346749459ccaf2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/40583bbef9fd5610cf91ba7c0ea59d5a6d932f05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/40583bbef9fd5610cf91ba7c0ea59d5a6d932f05"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
