)]}'
{"id":"openstack%2Fglance~841805","triplet_id":"openstack%2Fglance~master~Ied143f61b275f1b1dcc4dbab6a56aaacce971543","project":"openstack/glance","branch":"master","topic":"add-import-wait","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2022-05-18 21:22:06.000000000","reason":"Brian Rosmaita replied on the change"}},"hashtags":[],"change_id":"Ied143f61b275f1b1dcc4dbab6a56aaacce971543","subject":"DNM: testing devstack lib/tempest import wait","status":"ABANDONED","created":"2022-05-14 01:23:10.000000000","updated":"2022-05-18 21:22:39.000000000","total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3be1ddc1ecb04fdec751b951c0d3d1b578616ed5","_number":841805,"virtual_id_number":841805,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-05-18 05:46:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-14 02:56:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-17 21:41:09.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"fb05166f104740b56f5daf5c972e8f0273a69b92","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-05-14 01:23:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"315f82c373565437cbe8c33d9896706f2600824a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-14 02:56:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/13898330b9394ed89dda2aed1b545d41 : SUCCESS in 1h 22m 30s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/9a2e848736b640788c462fbd6bd378a9 : SUCCESS in 1h 15m 05s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a433ed1eb693450fb1cc19d887885ce4 : SUCCESS in 1h 31m 19s\n- glance-multistore-cinder-import-1 https://zuul.opendev.org/t/openstack/build/704d690afd074c9cb9263e5dd2094da7 : SUCCESS in 1h 25m 13s\n- glance-multistore-cinder-import-2 https://zuul.opendev.org/t/openstack/build/251dbb14da2c4158b050bc4806a1fb92 : SUCCESS in 1h 31m 34s\n- glance-multistore-cinder-import-3 https://zuul.opendev.org/t/openstack/build/27f597e419af4d11a819d4b635bef584 : SUCCESS in 1h 25m 57s\n- glance-multistore-cinder-import-4 https://zuul.opendev.org/t/openstack/build/6b7168093dfe40afbfb8537be5b183e8 : SUCCESS in 1h 23m 01s","accounts_in_message":[],"_revision_number":1},{"id":"dede828da04e8646d42f79216d6a0b11d813d566","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-05-17 13:13:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"884f7825afafc9507c064de32d0f520181af1a28","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-05-17 21:39:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3f4ab6572a73a0220f034b1958ab4e42f4aedffe","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-05-17 21:40:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a9581d17c368f00322feffaef554412b66b507fd","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-05-17 21:41:09.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"14d65ed08d557e0e553368af1643e461cf624f58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-17 21:41:16.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/glance for 841805,1","accounts_in_message":[],"_revision_number":1},{"id":"1bba55a8b5d29e98e39d2efdb89af241999e6e6e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-05-17 21:49:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6b6ab9cf676d08b1ee07b5e8382e24497da15c41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-17 23:55:01.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9699d13a68d649a582580679676862aa : SUCCESS in 1h 31m 57s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/308e7ec1855a4a3eae4c6d78196fc96a : SUCCESS in 1h 26m 53s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8ca576db782e4fb693b91b2e93e41c3a : SUCCESS in 1h 52m 51s\n- glance-multistore-cinder-import-1 https://zuul.opendev.org/t/openstack/build/55be7b68a50b499a809961c90a54ba7c : SUCCESS in 1h 57m 53s\n- glance-multistore-cinder-import-2 https://zuul.opendev.org/t/openstack/build/a83ec8a4850a4539b3f62b54c1e53c51 : FAILURE in 1h 33m 35s\n- glance-multistore-cinder-import-3 https://zuul.opendev.org/t/openstack/build/fa356df6dc3a49b69899646082a72e97 : SUCCESS in 1h 45m 01s\n- glance-multistore-cinder-import-4 https://zuul.opendev.org/t/openstack/build/953234c9ddcc445c92964565dd1ffffb : FAILURE in 1h 44m 27s","accounts_in_message":[],"_revision_number":2},{"id":"2c7091164a32b82528960d4d113888775d991abc","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-05-18 03:36:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e017d392d7c3ade7384ac2d6f3fbb69f608f158e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-05-18 03:50:28.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5002d6ac73d79f568ad80c9369a16f006d663c57","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-05-18 03:51:13.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"553479058f0baf782bc50da860601e9d41237d84","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-05-18 03:54:52.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"361435c84f433b88f1a3677e5c651889d16593f2","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-05-18 03:55:36.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0ae8ff59810bf10d31595d927565b53961af00a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-18 05:46:57.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1b21dc2f95474870bb2999d061cd7ed8 : SUCCESS in 1h 17m 06s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b6f6c03445984f0a880900e714a2309a : SUCCESS in 1h 15m 34s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a8e8a1fd830d402abde0a409605298d1 : SUCCESS in 1h 48m 53s\n- glance-multistore-cinder-import-1 https://zuul.opendev.org/t/openstack/build/0c37f826c3414c9995198a711eb374bb : SUCCESS in 1h 25m 52s\n- glance-multistore-cinder-import-2 https://zuul.opendev.org/t/openstack/build/b13b1bbaab904440aaa74f8f28df9bda : SUCCESS in 1h 45m 34s\n- glance-multistore-cinder-import-3 https://zuul.opendev.org/t/openstack/build/704674d047224763b7c0a1b90f756def : FAILURE in 1h 14m 55s\n- glance-multistore-cinder-import-4 https://zuul.opendev.org/t/openstack/build/6dc34ad5b29049bba8992f96317f42bf : FAILURE in 1h 29m 30s\n- glance-multistore-cinder-import-negative https://zuul.opendev.org/t/openstack/build/04378578260b4a68ad93c933b6042fbb : FAILURE in 26m 40s","accounts_in_message":[],"_revision_number":4},{"id":"ccfb973619e0aaf5b63c57cd1511283a7b58ed2d","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-05-18 12:27:14.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"82839d617fd492389f3f9d30c3b462772d3dd329","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-18 14:11:18.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/21881b04f9ad46f6aafd247ed17e56f0 : SUCCESS in 1h 21m 58s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/dc221b3b3ff34ae8ae03368e0fbfb3d0 : SUCCESS in 1h 24m 25s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7be42e99857f4e919075a2edf4f8df88 : SUCCESS in 1h 26m 33s\n- glance-multistore-cinder-import-1 https://zuul.opendev.org/t/openstack/build/8fbbdfba18534279bb089ba182374c32 : SUCCESS in 1h 40m 30s\n- glance-multistore-cinder-import-2 https://zuul.opendev.org/t/openstack/build/6f6ad11de01248c1914e0d95c38bfa48 : SUCCESS in 1h 35m 07s\n- glance-multistore-cinder-import-3 https://zuul.opendev.org/t/openstack/build/d89b31ce795e401ab553448e7f0470d2 : SUCCESS in 1h 20m 48s\n- glance-multistore-cinder-import-4 https://zuul.opendev.org/t/openstack/build/f2fdcee9fab344d690fd3a10c33f0c84 : SUCCESS in 1h 31m 04s\n- glance-multistore-cinder-import-negative https://zuul.opendev.org/t/openstack/build/68ad5e7f743c40dc8ed78a0c75829904 : FAILURE in 29m 56s","accounts_in_message":[],"_revision_number":4},{"id":"ae1f589762b8b5a4846d0aaed8c908cc9afca241","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-05-18 21:22:06.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3be1ddc1ecb04fdec751b951c0d3d1b578616ed5","tag":"autogenerated:gerrit:abandon","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-05-18 21:22:39.000000000","message":"Abandoned\n\nNo longer needed, https://review.opendev.org/c/openstack/devstack/+/841804 has merged.","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"240ff10a7e016f9dcb5682d233e8a24a75bf7930","revisions":{"34899d0fbfcda85e138ecd63b9309b6cff0acaf0":{"kind":"REWORK","_number":1,"created":"2022-05-14 01:23:10.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/05/841805/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/05/841805/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/05/841805/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/05/841805/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/05/841805/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/05/841805/1"}}},"commit":{"parents":[{"commit":"331ce59a50524e65630aed2d888af30f4494563d","subject":"Merge \"Fix tests for logging connection info\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/331ce59a50524e65630aed2d888af30f4494563d"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-05-14 01:21:49.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-05-14 01:21:49.000000000","tz":-240},"subject":"DNM: testing devstack lib/tempest import wait","message":"DNM: testing devstack lib/tempest import wait\n\nDepends-on: https://review.opendev.org/c/openstack/devstack/+/841804\nChange-Id: Ied143f61b275f1b1dcc4dbab6a56aaacce971543\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/34899d0fbfcda85e138ecd63b9309b6cff0acaf0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/34899d0fbfcda85e138ecd63b9309b6cff0acaf0"}]},"branch":"refs/heads/master"},"f1041d02a0c62fab1713f7976477300fc09586dc":{"kind":"REWORK","_number":2,"created":"2022-05-17 21:49:57.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/05/841805/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/05/841805/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/05/841805/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/05/841805/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/05/841805/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/05/841805/2"}}},"commit":{"parents":[{"commit":"9bc4ea498940cbae9c7f88d7f29381a0b6364992","subject":"Merge \"Override GLANCE_USE_IMPORT_WORKFLOW in cinder jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9bc4ea498940cbae9c7f88d7f29381a0b6364992"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-05-17 21:49:40.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-05-17 21:49:40.000000000","tz":-240},"subject":"DNM: testing devstack lib/tempest import wait","message":"DNM: testing devstack lib/tempest import wait\n\nDepends-on: https://review.opendev.org/c/openstack/devstack/+/841804\nChange-Id: Ied143f61b275f1b1dcc4dbab6a56aaacce971543\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f1041d02a0c62fab1713f7976477300fc09586dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f1041d02a0c62fab1713f7976477300fc09586dc"}]},"branch":"refs/heads/master"},"190fa9189c829f505e0d4fb50eac9a2d2a642aab":{"kind":"REWORK","_number":3,"created":"2022-05-18 03:50:28.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/05/841805/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/05/841805/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/05/841805/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/05/841805/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/05/841805/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/05/841805/3"}}},"commit":{"parents":[{"commit":"9bc4ea498940cbae9c7f88d7f29381a0b6364992","subject":"Merge \"Override GLANCE_USE_IMPORT_WORKFLOW in cinder jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9bc4ea498940cbae9c7f88d7f29381a0b6364992"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-05-17 21:49:40.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-05-18 03:50:16.000000000","tz":-240},"subject":"DNM: testing devstack lib/tempest import wait","message":"DNM: testing devstack lib/tempest import wait\n\nDepends-on: https://review.opendev.org/c/openstack/devstack/+/841804\nChange-Id: Ied143f61b275f1b1dcc4dbab6a56aaacce971543\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/190fa9189c829f505e0d4fb50eac9a2d2a642aab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/190fa9189c829f505e0d4fb50eac9a2d2a642aab"}]},"branch":"refs/heads/master"},"240ff10a7e016f9dcb5682d233e8a24a75bf7930":{"kind":"REWORK","_number":4,"created":"2022-05-18 03:54:52.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/05/841805/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/05/841805/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/05/841805/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/05/841805/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/05/841805/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/05/841805/4"}}},"commit":{"parents":[{"commit":"9bc4ea498940cbae9c7f88d7f29381a0b6364992","subject":"Merge \"Override GLANCE_USE_IMPORT_WORKFLOW in cinder jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9bc4ea498940cbae9c7f88d7f29381a0b6364992"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-05-17 21:49:40.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-05-18 03:54:45.000000000","tz":-240},"subject":"DNM: testing devstack lib/tempest import wait","message":"DNM: testing devstack lib/tempest import wait\n\nDepends-on: https://review.opendev.org/c/openstack/devstack/+/841804\nChange-Id: Ied143f61b275f1b1dcc4dbab6a56aaacce971543\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/240ff10a7e016f9dcb5682d233e8a24a75bf7930"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/240ff10a7e016f9dcb5682d233e8a24a75bf7930"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
