)]}'
{"id":"openstack%2Fglance~841837","triplet_id":"openstack%2Fglance~master~I2775e007f942feed8fa6ae5e385c03992859edc1","project":"openstack/glance","branch":"master","topic":"bug/1973136","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2022-05-16 14:02:21.000000000","reason":"Abhishek Kekane replied on the change"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2022-05-17 14:19:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2775e007f942feed8fa6ae5e385c03992859edc1","subject":"Override GLANCE_USE_IMPORT_WORKFLOW in cinder jobs","status":"MERGED","created":"2022-05-16 07:05:19.000000000","updated":"2022-05-19 08:26:47.000000000","submitted":"2022-05-17 14:19:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"841837-bug/1973136","meta_rev_id":"7440caa5a56b0dd29ea62b3091d620742c6f6193","_number":841837,"virtual_id_number":841837,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-17 14:19:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2022-05-17 13:15:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2022-05-17 05:36:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2022-05-17 13:15:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-16 07:06:06.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-16 07:15:51.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2022-05-16 09:19:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-17 13:15:34.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"}],"messages":[{"id":"553bd0d324c23b0e7bc83927ddb4eb1a690732ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-05-16 07:05:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e6ced7b13adc0f421da2a425f3db3695ba5915d7","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-05-16 07:06:06.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/841837/841837/1/check-not-vote/ext-glance-zuul/4c42834 : FAILURE in 20s","accounts_in_message":[],"_revision_number":1},{"id":"5fa7af4c3f83bb0d7b1aba31f768f5eb1ec1060c","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-05-16 07:15:51.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6cf891b6e1eebb856ab236a75d1e5e255d3315c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-05-16 08:31:39.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c26cb137c3d7f484bf49bcee06b945d52ca2ee18","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-05-16 08:32:24.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/841837/841837/2/check-not-vote/ext-glance-zuul/a20460a : FAILURE in 19s","accounts_in_message":[],"_revision_number":2},{"id":"fda24f9644a14437c62b3ffb8d820f6673104842","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-16 09:19:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1b2d6182fe734aafb6c83ed3cb51c54c : SUCCESS in 46m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4264faa27e4840bf897aae70be4b505b : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4be3b176465548dfb82c7a6c72237b44 : SUCCESS in 5m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1d09a9600bd0445bb0d522ccd2c2e95b : SUCCESS in 5m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8ff32eb865d46409ab54e6bf5373084 : SUCCESS in 6m 25s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/446a7862b27142a983a435d75814c63d : SUCCESS in 11m 03s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/025b5bb784d84fbf8881dfc67e55b7ce : SUCCESS in 10m 21s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/ded05659d53047b7a5793279291934c9 : SUCCESS in 9m 14s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/9a52e2243a0549d094fa0eadd22457db : SUCCESS in 31m 35s","accounts_in_message":[],"_revision_number":2},{"id":"f3fe31846ed0f8b8e7c2a2e349c33b41324cfc95","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-05-16 09:30:52.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d9f11ec62806bad26b3541bc35efe34cc1fa3682","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-05-16 09:36:18.000000000","message":"Patch Set 2: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"541d7dba1ddfb586a038715a76d5d5c26d51c1b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-05-16 10:38:23.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d0718c5050a2cd26b3e4f645930188577a315f1e","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-05-16 10:39:12.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"c394b231298c76d7beaaecf5f33fde38cd13387e","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-05-16 10:39:45.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/841837/841837/3/check-not-vote/ext-glance-zuul/b0292c8 : FAILURE in 54s","accounts_in_message":[],"_revision_number":3},{"id":"6bebd8576297657abf83de5545870bcc09a8fccd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-16 11:59:08.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c7ce269e640143e28567125c20a0a415 : SUCCESS in 50m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6ad91b6eb4045829949db38d0bae07f : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cf1eeb96a48e4f42821b318b8c8daf02 : SUCCESS in 6m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9a3d3b27ac274c3f8f21a80d20d3cae0 : SUCCESS in 6m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b8dda5b89104f999a24d573b858005d : SUCCESS in 7m 02s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/b7ffed7e5acb49d4927e08bb5e912002 : SUCCESS in 10m 34s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/88e959791865419b944b8fb5ce4f13fb : SUCCESS in 10m 36s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/fdae9b876f5649089e3bf3cd39411387 : SUCCESS in 9m 41s (non-voting)\n- glance-multistore-cinder https://zuul.opendev.org/t/openstack/build/2a36cc25360e4f96a3ffb9e5c4982f28 : SUCCESS in 1h 19m 06s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d3ff8ae172ac4b2c94c0896d3e305ce4 : FAILURE in 32m 39s","accounts_in_message":[],"_revision_number":3},{"id":"6c0af6c1c05ad6e396c11d93b92a34811cf583f4","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-05-16 12:52:52.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"006df2bdb6a7a5a205e50d28645e38b5a0637741","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-05-16 12:53:41.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/841837/841837/3/check-not-vote/ext-glance-zuul/62d924c : FAILURE in 18s","accounts_in_message":[],"_revision_number":3},{"id":"1e006d44babd02574fd3241e4ca76a51f5418120","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-16 13:57:02.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c54bfe6138094cf68cc146ce20b83ee8 : SUCCESS in 35m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8087aed28bda475696a4ff1b5d0b0d14 : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/71efeed48b174997b7e2f10c38651bd3 : SUCCESS in 5m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fac7c13027be4dc88cedf116b2f6fed1 : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9297ff8fe434417a573b47be03bada7 : SUCCESS in 6m 26s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/84aff8da7a3f4d19aec129db0c98ab0a : SUCCESS in 9m 34s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3b07a4978013424e97fcb6c50bc05b0e : SUCCESS in 7m 49s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/f169ca52f7784e77bd777f2ba638cff6 : SUCCESS in 8m 19s (non-voting)\n- glance-multistore-cinder https://zuul.opendev.org/t/openstack/build/ae14f98166fb44f3ad6c40ec5bd35d85 : SUCCESS in 1h 02m 27s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/85658e9e913247c795de1dec6ea409e6 : FAILURE in 24m 03s","accounts_in_message":[],"_revision_number":3},{"id":"bada47c6cb717f8846fcffef103ac01520198ab5","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-05-16 14:02:21.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"303a64ce5e8184becf1ffea1dbc13ca587c982eb","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-05-16 14:03:03.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/841837/841837/3/check-not-vote/ext-glance-zuul/44b7bbf : FAILURE in 18s","accounts_in_message":[],"_revision_number":3},{"id":"d92ef85361bb98baddf5f7fb844aef8da49e6e89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-16 15:26:12.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f0b7955d575047dd928cf9a4312817c5 : SUCCESS in 48m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b78af605e08b41dc93eccbfebf409eb4 : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7ac5233687bc49a4836138e4fcf43153 : SUCCESS in 6m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bff3dce4b56b4c6c9776ccbfe91111c1 : SUCCESS in 5m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d81a65514fc48d4a2133491418f7319 : SUCCESS in 6m 17s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/e75a0b9301ab4f5e8b2274c87a8b5251 : SUCCESS in 12m 37s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3298fca5dc1c4e0f99f6df6ab1e2579e : SUCCESS in 10m 08s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/c3b5baf54afd4524af3df91c1d6a77e4 : SUCCESS in 9m 02s (non-voting)\n- glance-multistore-cinder https://zuul.opendev.org/t/openstack/build/0e969e7837b34fa887ec860abb575a6f : SUCCESS in 1h 21m 53s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/46acdec9efd94041837e813857631828 : FAILURE in 31m 39s","accounts_in_message":[],"_revision_number":3},{"id":"61b843993d8a104671db2f9d90a3e126c69db8fd","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-05-17 05:36:28.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"965e03e74ee189c34ba24b6925be6929d726955f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-17 06:51:44.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4cef7781c05e403d8523e3470d676f39 : SUCCESS in 38m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7de7c8ab660d4de39a6ed40cb89265e2 : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c07652372d64c57ac289400fceccd89 : SUCCESS in 4m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a018c244d6384c7e808e06f6b5c31fc9 : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/830a35f1a476443f849744fa9be5ad6e : SUCCESS in 5m 23s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/690dc41703f645e4bcb2d0c233ba6260 : SUCCESS in 9m 11s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/46a468a8adf84c28b1a2ba5385d84e93 : SUCCESS in 8m 15s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/ba0efc684c8549ac97d34803a6d1d674 : SUCCESS in 8m 24s (non-voting)\n- glance-multistore-cinder https://zuul.opendev.org/t/openstack/build/b6b839a15e7b44a6895e075f158aa23f : SUCCESS in 1h 13m 49s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d390be8228cf4121a1e4b57340e95014 : SUCCESS in 24m 48s","accounts_in_message":[],"_revision_number":3},{"id":"f53dd18a92327a7bbb6e8b21a740f6b984b8cff7","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-05-17 13:15:34.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c5689e097f480b642b959bca8a4fb9a8b5509661","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-17 13:16:34.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a94450d2d0c3788fe982229b3b301cd2fbd110f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-17 14:19:35.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/edfec77391234ccd8df20c227f81a4de : SUCCESS in 57m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c1156d8cc6941039a51e63ce157a7d1 : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/af620aed0d224ab4a26426cf14229967 : SUCCESS in 5m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f7d09acc8ea34b45b6c0b3817763a7f7 : SUCCESS in 6m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b959a36c9164dde929d3accbd55ef84 : SUCCESS in 7m 28s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9956494ffaff4b728a9f1b7d51e25ad2 : SUCCESS in 9m 26s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1f0113c0cbfb4e12addd251450c370a2 : SUCCESS in 9m 47s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2498d5d46bc949729f8cf949c885cadf : SUCCESS in 39m 11s","accounts_in_message":[],"_revision_number":3},{"id":"edf6ee200c106911e3250d102ae78061cf64fb9a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-17 14:19:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"8c1587c6cb13ddc110939e83c963957023e65882","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-17 14:22:49.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f332dc4ef6f44b69566b7bbe81c93af : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":3},{"id":"7440caa5a56b0dd29ea62b3091d620742c6f6193","tag":"autogenerated:gerrit:revert","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-05-19 08:26:47.000000000","message":"Created a revert of this change as Ie72a2531172208e32626ad2060737257663eba46","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"025ba83c470fd58425273ad10975a1eb3d095452","revisions":{"9fb442e2f3faef320438980be79946dad78d2624":{"kind":"REWORK","_number":1,"created":"2022-05-16 07:05:19.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/37/841837/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/37/841837/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/37/841837/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/37/841837/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/37/841837/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/37/841837/1"}}},"commit":{"parents":[{"commit":"d7fa7a0321ea5a56ec130aa0bd346749459ccaf2","subject":"Disable import workflow in glance cinder jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d7fa7a0321ea5a56ec130aa0bd346749459ccaf2"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2022-05-16 06:55:49.000000000","tz":330},"committer":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2022-05-16 07:05:15.000000000","tz":330},"subject":"Override GLANCE_USE_IMPORT_WORKFLOW in cinder jobs","message":"Override GLANCE_USE_IMPORT_WORKFLOW in cinder jobs\n\nWith change[1], we moved glance cinder jobs parent from\ntempest-integrated-storage-import to regular\ntempest-integrated-storage job but this job also has\nimport workflow enabled[2].\nIn this patch we are overriding GLANCE_USE_IMPORT_WORKFLOW\nto False in cinder jobs to ensure that import workflow\nis disabled in those jobs.\n\n[1] https://review.opendev.org/c/openstack/glance/+/841548\n[2] https://opendev.org/openstack/tempest/src/branch/master/zuul.d/integrated-gate.yaml#L195\n\nRelated-Bug: #1973136\nChange-Id: I2775e007f942feed8fa6ae5e385c03992859edc1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9fb442e2f3faef320438980be79946dad78d2624"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9fb442e2f3faef320438980be79946dad78d2624"}]},"branch":"refs/heads/master"},"1edf7f26e0ee7ef12e0d2a3eb0c497f083504b08":{"kind":"REWORK","_number":2,"created":"2022-05-16 08:31:39.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/37/841837/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/37/841837/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/37/841837/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/37/841837/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/37/841837/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/37/841837/2"}}},"commit":{"parents":[{"commit":"d7fa7a0321ea5a56ec130aa0bd346749459ccaf2","subject":"Disable import workflow in glance cinder jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d7fa7a0321ea5a56ec130aa0bd346749459ccaf2"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2022-05-16 06:55:49.000000000","tz":330},"committer":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2022-05-16 08:31:19.000000000","tz":330},"subject":"Override GLANCE_USE_IMPORT_WORKFLOW in cinder jobs","message":"Override GLANCE_USE_IMPORT_WORKFLOW in cinder jobs\n\nWith change[1], we moved glance cinder jobs parent from\ntempest-integrated-storage-import to regular\ntempest-integrated-storage job but this job also has\nimport workflow enabled[2].\nIn this patch we are overriding GLANCE_USE_IMPORT_WORKFLOW\nto False in cinder jobs to ensure that import workflow\nis disabled in those jobs.\nAlso removing -import from job names since it\u0027s not using\nimport plugin anymore.\n\n[1] https://review.opendev.org/c/openstack/glance/+/841548\n[2] https://opendev.org/openstack/tempest/src/branch/master/zuul.d/integrated-gate.yaml#L195\n\nRelated-Bug: #1973136\nChange-Id: I2775e007f942feed8fa6ae5e385c03992859edc1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1edf7f26e0ee7ef12e0d2a3eb0c497f083504b08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1edf7f26e0ee7ef12e0d2a3eb0c497f083504b08"}]},"branch":"refs/heads/master"},"025ba83c470fd58425273ad10975a1eb3d095452":{"kind":"REWORK","_number":3,"created":"2022-05-16 10:38:23.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/37/841837/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/37/841837/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/37/841837/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/37/841837/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/37/841837/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/37/841837/3"}}},"commit":{"parents":[{"commit":"d7fa7a0321ea5a56ec130aa0bd346749459ccaf2","subject":"Disable import workflow in glance cinder jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d7fa7a0321ea5a56ec130aa0bd346749459ccaf2"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2022-05-16 06:55:49.000000000","tz":330},"committer":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2022-05-16 10:38:07.000000000","tz":330},"subject":"Override GLANCE_USE_IMPORT_WORKFLOW in cinder jobs","message":"Override GLANCE_USE_IMPORT_WORKFLOW in cinder jobs\n\nWith change[1], we moved glance cinder jobs parent from\ntempest-integrated-storage-import to regular\ntempest-integrated-storage job but this job also has\nimport workflow enabled[2].\nIn this patch we are overriding GLANCE_USE_IMPORT_WORKFLOW\nto False in cinder jobs to ensure that import workflow\nis disabled in those jobs.\nAlso removing -import from job names since it\u0027s not using\nimport plugin anymore.\n\n[1] https://review.opendev.org/c/openstack/glance/+/841548\n[2] https://opendev.org/openstack/tempest/src/branch/master/zuul.d/integrated-gate.yaml#L195\n\nRelated-Bug: #1973136\nChange-Id: I2775e007f942feed8fa6ae5e385c03992859edc1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/025ba83c470fd58425273ad10975a1eb3d095452"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/025ba83c470fd58425273ad10975a1eb3d095452"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
