)]}'
{"id":"openstack%2Fglance~841935","triplet_id":"openstack%2Fglance~stable%2Fyoga~I09fc9164a08f42507d2aec44c5b382a72f232571","project":"openstack/glance","branch":"stable/yoga","topic":"fix-rs-type-404-stable/yoga","attention_set":{},"removed_from_attention_set":{"5202":{"account":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"last_update":"2022-05-23 14:11:04.000000000","reason":"Change was submitted"},"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2022-05-20 02:41:05.000000000","reason":"Brian Rosmaita replied on the change"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2022-05-23 14:11:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I09fc9164a08f42507d2aec44c5b382a72f232571","subject":"Fix failing namespace list delete race","status":"MERGED","created":"2022-05-17 06:16:34.000000000","updated":"2022-05-23 14:12:32.000000000","submitted":"2022-05-23 14:11:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"841935-fix-rs-type-404-stable/yoga","cherry_pick_of_change":841970,"cherry_pick_of_patch_set":2,"meta_rev_id":"933a5b908ceaf06fbd2b64777b25f2ebbf92d8f0","_number":841935,"virtual_id_number":841935,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-23 14:11:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2022-05-23 06:08:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2022-05-23 12:40:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2022-05-23 12:40:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-17 06:16:34.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"CC"},{"updated":"2022-05-17 06:27:02.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2022-05-17 06:27:02.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2022-05-17 07:46:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-23 12:40:37.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"22d1e075451a4c2375f526b33728996c068ba0db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-05-17 06:16:34.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"1c0dc499dd7e667f5c5636c70e7a8421d8b6d6fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-17 07:46:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/9e20153087cc4149b3ac713373472fa3 : SUCCESS in 1h 02m 46s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/c67966e0cb4c4213b06f0899b0e8d358 : SUCCESS in 1h 11m 23s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/eda31600555e49758e6e6e13e94ef1c0 : SUCCESS in 1h 10m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4a11b8e95f5a4d649c47578a1f41044d : SUCCESS in 45m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/966d4856790648fea54428845191924d : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/35ead599b5e5442c8e46846b0730a17d : SUCCESS in 5m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/31f12a26094c4492b9cf32874aad8b19 : SUCCESS in 6m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/702095048cf0415fac2196e43b8e9c3b : SUCCESS in 6m 08s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/ba797a3b2dbe45dfa0854d1c94b884c7 : SUCCESS in 11m 14s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3b297fe0236e4761bdf70a6ecbed8015 : SUCCESS in 9m 01s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/7e3e309a533743b981a91667d257b171 : SUCCESS in 9m 04s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/e63dc3a28853469a97f6630774597967 : SUCCESS in 4m 23s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/22e2733307594821ac0469cec148ebb5 : SUCCESS in 1h 09m 06s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ec71f280f32448dba6f64be91b8b8778 : SUCCESS in 1h 05m 40s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/2c79a96b204b40a69be7767b11e50c5e : SUCCESS in 1h 23m 33s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c27e3fcb42784389b402e675beb59e96 : FAILURE in 24m 18s\n- glance-multistore-cinder-import-fips https://zuul.opendev.org/t/openstack/build/cb8f55756ac5454caf9e286a84e46813 : FAILURE in 1h 22m 31s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e22e5c62c03d495ca514296946a95f52 : SUCCESS in 59m 50s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/63aaf4acd58a444eb5f6c51c23fb994a : SUCCESS in 1h 08m 10s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/80c9d952c7c346f1b26091a08bdfdce1 : SUCCESS in 29m 53s","accounts_in_message":[],"_revision_number":1},{"id":"c96dfcf5d50e1819f0a56bffff3345ab66958714","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-05-20 02:41:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3377b6bd792e704f9cf5d3e6367f27356a75c188","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-20 04:17:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/9a8f6bfe99a84ab2ab9fe72a9e43e0da : SUCCESS in 1h 05m 31s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/b959d9eb2f06459db1479013712e3f12 : SUCCESS in 1h 12m 01s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/73e0b3406bf94c15be2d3b172fd4c85d : SUCCESS in 1h 13m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b14b9674062041feb13f57c36bb71308 : SUCCESS in 57m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7fe1e5bdf17145de813d3e737234eca3 : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be61ea9443714143b116dd7286b57abb : SUCCESS in 5m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f7664863d6ef45aa93f0825833f22430 : SUCCESS in 6m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd47b80280c349e5ba151fa788df0030 : SUCCESS in 6m 34s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/80c640fd7ac9437a9c025aab05c523a2 : SUCCESS in 9m 51s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a91b6436adef4eb4904447291310cd7c : SUCCESS in 9m 15s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/522b3f32f02b4bd28a768712f94b6f6c : SUCCESS in 10m 45s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/d861734fb1f046f6876d226cb433abf9 : SUCCESS in 6m 12s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f9daa41c4f8c4ae9873f3f88a50933ef : SUCCESS in 1h 05m 25s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/fd846f8710fb4b778ceee64be7303601 : SUCCESS in 1h 18m 31s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/a9f554d7826b480abc9d2b6657c88fbb : SUCCESS in 1h 12m 44s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/595c332760894c7180cc85986baeb710 : SUCCESS in 1h 29m 31s\n- glance-multistore-cinder-import-fips https://zuul.opendev.org/t/openstack/build/2333f5d6636d4723a1f565d064c16a6c : FAILURE in 40m 59s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c67078f5001249e2878429b56cf1fb31 : SUCCESS in 54m 56s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/507cb8e4a92d44b28928c1fd57eaec90 : SUCCESS in 1h 10m 19s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b4d56cfd71e44d428d5316389d70acea : SUCCESS in 31m 11s","accounts_in_message":[],"_revision_number":1},{"id":"82bcfe1c20c62936493fd0e0341bc3f8304bab81","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-05-23 06:08:32.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e73b04b60ab0886c19c87ebbeca63e28de9e4952","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-05-23 12:40:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"480c205eee27a4a4b12f94f79c67592db5163f17","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 12:41:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"41d3de2c035d4d9166e05ead65bc7909f7e0b8a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 14:11:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/5ecc33e2dcca43b1b413eb43b4761ef9 : SUCCESS in 55m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/30de60ba92004893af7ffe33752296ac : SUCCESS in 1h 11m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/348b42dbc2654d57b9591b0bbe70a7ac : SUCCESS in 48m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab7b99b771e24651b2745fc5f71c9ffd : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c570ed78aaaa45ba9479eafc7967a4d7 : SUCCESS in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6a32cd30efdb40ea90c951643b8cb869 : SUCCESS in 8m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/133872b3ac424c64b99a5dcc81493357 : SUCCESS in 8m 41s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/87a38a041957494791a06fb8d631ba99 : SUCCESS in 9m 48s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5b08a500cc6d4329a238b73864ed60b1 : SUCCESS in 11m 05s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/119832ddc3e648a38a664b0504268870 : SUCCESS in 1h 21m 12s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/6ce6ac113b104498953ea0e6a6f1dea3 : SUCCESS in 1h 21m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3d249a051cc94604b783f915dc23378c : SUCCESS in 46m 51s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5fa71bf37ba24893a85355fa2d392d49 : SUCCESS in 1h 16m 07s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f87fe4b458214844b13f728fddd8a590 : SUCCESS in 32m 53s","accounts_in_message":[],"_revision_number":1},{"id":"b71426454a22ca52bb63e31032c556d948f15072","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 14:11:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"933a5b908ceaf06fbd2b64777b25f2ebbf92d8f0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 14:12:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/398bb5e563d146be926f7f081f596782 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cc98dbef31be30372b59b2d18b78694ce0879aa0","revisions":{"cc98dbef31be30372b59b2d18b78694ce0879aa0":{"kind":"REWORK","_number":1,"created":"2022-05-17 06:16:34.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/35/841935/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/35/841935/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/841935/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/841935/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/841935/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/35/841935/1"}}},"commit":{"parents":[{"commit":"1beee78e096ddba4729be10af6ac4ddec8e79da0","subject":"Update TOX_CONSTRAINTS_FILE for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1beee78e096ddba4729be10af6ac4ddec8e79da0"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2022-05-16 18:58:05.000000000","tz":-420},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2022-05-17 06:16:34.000000000","tz":0},"subject":"Fix failing namespace list delete race","message":"Fix failing namespace list delete race\n\nIf a namespace is deleted by another client while we are doing a\nnamespace list operation, we will fail the list with NotFound if we\ntry to pull the resource_type_associations list. The latter re-queries\nthe DB for the namespace and will raise NotFound to us. This is\nespecially bad because the namespace being deleted need not even\nbelong to the caller of the list, as is the case in a tempest run.\n\nThis makes us catch the failure and continue the operation, reporting\nno associations so that the client gets a consistent view and no\nerror.\n\nCloses-Bug: #1973631\nChange-Id: I09fc9164a08f42507d2aec44c5b382a72f232571\n(cherry picked from commit 35e6c57bc90365796f45ada9271e532068557b8a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cc98dbef31be30372b59b2d18b78694ce0879aa0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cc98dbef31be30372b59b2d18b78694ce0879aa0"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
