)]}'
{"id":"openstack%2Fglance~842070","triplet_id":"openstack%2Fglance~master~I9ab3f1b7595e22dbb03af95168314352a44eb930","project":"openstack/glance","branch":"master","topic":"bp/instant_caching","attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2022-09-07 14:47:37.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2022-07-19 12:07:43.000000000","reason":"\u003cGERRIT_ACCOUNT_5314\u003e replied on the change","reason_account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2022-07-19 14:32:42.000000000","reason":"\u003cGERRIT_ACCOUNT_4393\u003e replied on the change","reason_account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2022-07-19 15:30:45.000000000","reason":"\u003cGERRIT_ACCOUNT_8122\u003e replied on the change","reason_account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}}},"hashtags":[],"change_id":"I9ab3f1b7595e22dbb03af95168314352a44eb930","subject":"Immediate caching of an image","status":"MERGED","created":"2022-05-17 09:46:25.000000000","updated":"2022-09-07 14:47:37.000000000","submitted":"2022-07-19 17:01:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":86,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"842070-bp/instant_caching","meta_rev_id":"9eab077b7d85034d2414d63d7a067e0e463dc97d","_number":842070,"virtual_id_number":842070,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-19 17:01:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2022-07-19 12:07:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":2,"date":"2022-07-19 15:30:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":1,"date":"2022-07-19 15:30:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}],"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-17 09:47:17.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2022-05-17 11:24:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-13 09:07:19.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-18 06:13:01.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2022-07-18 06:13:01.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2022-09-07 14:47:37.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"}],"messages":[{"id":"7565bad7721a0e872099f5651e1e85ff7537c535","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-05-17 09:46:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"43a9079b56b28e2a7df05680462a9d47d734ca1b","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-05-17 09:47:17.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"abecc0a719ebd51838537fd280780c49099df0ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-17 11:24:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/e281b446cfc34449b1e4dfc28d9eaa6f : SUCCESS in 49m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1a2b13ffb0c440cc92661887913cde16 : SUCCESS in 1h 07m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d30fbde989bd47e583f4174700079627 : SUCCESS in 45m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1406417efa534638a7c8f378ab59d860 : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a87f4fc650f34b57a56835184778e15c : FAILURE in 5m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/84d4a14588b744c5be36b682e6fb2990 : FAILURE in 5m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/631a938cd51e49ba9fcc883cb50f98ea : FAILURE in 5m 07s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/85cc6da78ccd4af28bf873d8914520d1 : FAILURE in 10m 58s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e7763852f05949779ad037f030d5fd56 : FAILURE in 9m 42s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/68c17ef1e1df46e1a01f0bb345d1ba25 : FAILURE in 8m 58s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/9a9102a84c6346f686a1c2fb89620d12 : SUCCESS in 4m 29s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/a589dad0fe664e34b8cd6b48bf6b8b35 : SUCCESS in 1h 10m 17s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7dd03fedbc07492eafd71e75051de0ab : SUCCESS in 1h 12m 04s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/a8e18c6e2e8e4d7e8de98f203904b754 : FAILURE in 29m 44s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f42c4e5585af4b4583937bca5703c7f4 : SUCCESS in 1h 31m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3a51afebaa9046eb9b159dc111525304 : SUCCESS in 51m 27s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/dbed633e8e79470cac451aeb4841c252 : FAILURE in 33m 19s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/be8eccba0de2415dbb7191f02846896e : SUCCESS in 39m 54s","accounts_in_message":[],"_revision_number":1},{"id":"e77266b8b5564a378617ff927227567a1638d32c","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-05-17 14:16:11.000000000","message":"Patch Set 1:\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"3e9dfc458f1f0911a31d78fa8c470334d53b27ef","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-05-17 14:38:32.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"fc3266ebdae30c5648770bfcf8261823c545795a","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-05-17 14:49:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cb7dd0eaff42e38e859fa349c59f2d0bcc3da02b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-05-19 07:29:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0e8aaada47c8cf4f4eafc6384e50512104919fbf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-19 09:04:26.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/0ad132a06ae445b397c29de8a4a2212a : SUCCESS in 49m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9062e83dbc854691b2c1f7d390cfc90e : SUCCESS in 1h 14m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3f51554a5ae04a65994ed66702fa3a76 : SUCCESS in 49m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/784e0d5a08274e7f8f9b74e99fcce796 : FAILURE in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/22072abbef7e4e579ffbf05d153bf934 : FAILURE in 5m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8cef34ea78364d1699255689bce78975 : FAILURE in 6m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f6d8eadc6cf4a94bec0139c57c7b673 : FAILURE in 5m 42s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/411f74bac09a4e8e8e54248aa9230f67 : FAILURE in 6m 36s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3403472061be46d3ace36f1f254e47d4 : FAILURE in 8m 57s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/95b597a28c0b4c7083f2d1793c65bb91 : FAILURE in 9m 40s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/a2de025294a64e3c98ab30904400e3af : SUCCESS in 5m 06s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/9af9b3a14d80475d975ab2bcb1dc8ef8 : SUCCESS in 1h 07m 26s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/199b66d25b0b40e29d563fd260ba929a : SUCCESS in 1h 28m 23s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/1232e58c4107437ab43f798177e4550f : FAILURE in 24m 55s\n- glance-multistore-cinder https://zuul.opendev.org/t/openstack/build/c6d66d5b2f7343a88e82b41e5286591f : SUCCESS in 1h 30m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7d86c5dd2fe34172aa325e20bf85ca26 : SUCCESS in 1h 01m 29s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/352f478c5fed49afac1a8aa00e6685f5 : FAILURE in 25m 26s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/9dfe2a9f29c542398dfa153e17925a09 : SUCCESS in 30m 45s","accounts_in_message":[],"_revision_number":2},{"id":"659d6bbef5db61556ec50aa3b58b3601d7ba9d06","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-06-10 17:14:11.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"bf950043860d573758cb146219720b94c1821ad9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-06-10 17:14:53.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e66763440d349e9f4e7cf6d1bd42f7c3cc91cc3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-10 18:02:32.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/2b7be43e7e7141698b2246fdff6a9f90 : FAILURE in 45m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/896909790fda427b84320e545c16cf10 : FAILURE in 22m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dc120a130b294fb8ad96cc4491335eef : POST_FAILURE in 25m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99b6e29299624765be704b6603c166fa : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d5bb60baf8024ad6868cbb37aa1b3a35 : FAILURE in 6m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f58c3aa9b21941a296b117e65a7ece57 : FAILURE in 7m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dcf67c60a55241c8836d406eb8da71fe : FAILURE in 7m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f4c5b73fac140f79ae16fd320ec01ba : FAILURE in 5m 56s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/999f42ba62494f6e9d73618ee1777edc : FAILURE in 8m 17s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/26357bcf9c784bc18ca91b28e6611aeb : FAILURE in 4m 47s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/64c337c45686499cbaaa41ae172f0af7 : FAILURE in 5m 07s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/f8d3ec6323e740119e76b47305ae9c79 : SUCCESS in 4m 26s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/3c975df30b1746158944b6039618f710 : FAILURE in 32m 44s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d19897279d2a4d72a48838da3d07560d : POST_FAILURE in 40m 57s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b00375fb2ae94696b33dfb3b610a6b8e : POST_FAILURE in 28m 53s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/88957078b0cd4ec79f7c402ef074279a : POST_FAILURE in 31m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2463e18274ee47fdb3b6009ec33e27bf : FAILURE in 21m 56s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/fe55e199a2e34c47810893af112b6527 : FAILURE in 37m 42s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/14de895615154eeeba398283b36c30e6 : FAILURE in 22m 22s","accounts_in_message":[],"_revision_number":3},{"id":"46a9de139655281c79ae85a3161f64609055f81e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-06-13 09:06:29.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"52a7b686cc377c1e0b7cb526e4381a26bb2b870d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-06-13 09:07:19.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/842070/842070/4/check-not-vote/ext-glance-zuul/998ebf6 : FAILURE in 18s","accounts_in_message":[],"_revision_number":4},{"id":"d15b153be28634742480bb8c5fb179941af47399","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-06-13 09:07:46.000000000","message":"Patch Set 4:\n\n(4 comments)","accounts_in_message":[],"_revision_number":4},{"id":"2dedff3d9f038afa90b597534ff00d667df136a6","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-06-13 09:08:06.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"720c30754aad1f73a8519b9c671aebd0164de8af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-06-13 09:09:07.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"2003509c2a27398a8dbb7b745c9b6561a7d0b930","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-06-13 09:10:01.000000000","message":"Patch Set 5:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/842070/842070/5/check-not-vote/ext-glance-zuul/4de58b4 : FAILURE in 24s","accounts_in_message":[],"_revision_number":5},{"id":"2577e3829e5eb74db33d245dd0722ff35b27ff9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-13 10:37:52.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/7eb68f8c85e648eb8fd1917bc61f3b8d : SUCCESS in 1h 02m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e825b7e8ee6b46bbb4111a991e21b19e : SUCCESS in 1h 23m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6aaec8e5a76b45cfb6f0bd3ca0bebdcc : SUCCESS in 49m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa9d309ea0314a86ab772cfa23210df9 : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d862a534ae4541a7872cb27be79b80db : FAILURE in 5m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3cb65ab2c4c745e88b458cb61ab6e033 : FAILURE in 5m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8d4184ed23294c18835e03105fe33c42 : FAILURE in 5m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20b55d8dba9642b0a0db4578890b9b63 : SUCCESS in 6m 02s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/3988465730cc413b945bd4b0a9caefe0 : SUCCESS in 11m 03s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/966f1de535f84a48934b972bde42fc34 : SUCCESS in 9m 23s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/d16b6ff8198946d498c15920212deafd : SUCCESS in 8m 22s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/42ceb0f99e02439e83b862fc6e4f1577 : SUCCESS in 5m 00s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/6ba5079819f440d8b03facc3d06c9f5b : SUCCESS in 1h 11m 33s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/db7f782e35084a0dab4702b82db54fba : SUCCESS in 1h 14m 37s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/df3f34c58a3b4bb3870e45dd5f15c3ba : SUCCESS in 1h 10m 18s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b7bc5c7273ce4e77a504d0cf3c2e2b52 : SUCCESS in 1h 22m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/664198410b544219a112549c3de3d0e0 : SUCCESS in 51m 45s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c318bf8dee394923bb544e3ef84947bd : SUCCESS in 1h 09m 00s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a1486d6d180340528e7d9a8de9f771a4 : SUCCESS in 38m 24s","accounts_in_message":[],"_revision_number":5},{"id":"1a452e37a1b2d83618fb15dd09df65a5be4e196a","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-06-13 14:14:07.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"730cbcc95b254f19a5da70ac3a1d56bf6e8691e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-06-13 17:48:55.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"d685dc4c550772ac5ea11ea4dccacd10493eaced","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-06-13 17:49:40.000000000","message":"Patch Set 6:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/842070/842070/6/check-not-vote/ext-glance-zuul/ae53bfd : FAILURE in 19s","accounts_in_message":[],"_revision_number":6},{"id":"0deadc948175c5cd9c314287c42d1e6b2b7ae87e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-13 19:30:01.000000000","message":"Patch Set 6: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/905830e733b14c38862767179c1c8b19 : SUCCESS in 50m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0ed6bf51241c485bbc782000c83c5105 : SUCCESS in 1h 25m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/22943180fa594901815621ff808e7ce6 : SUCCESS in 42m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6bf6c104b2d6476bb8c84896d330caba : FAILURE in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/787a7ae51ed442e4bc317a95ef23608a : SUCCESS in 5m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a50ac6821d704c4bb6d172cda5fdf6fb : SUCCESS in 6m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/243e4860602e40d7bf6b3f82637644ec : SUCCESS in 5m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c397389638dc499795000633046858c7 : SUCCESS in 5m 50s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/c8515141fd0e46899ccf5e292c72770a : SUCCESS in 10m 49s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/de4d51e761bb41dd992013c6fa9bcd08 : SUCCESS in 10m 40s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/5a27f272b3314e328fdb9a9f4fc4a3f2 : SUCCESS in 10m 14s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/4266bb324c19447d8b9b44593ae0baf3 : SUCCESS in 4m 25s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/af5ba791e265433ba962c5cb8e89628b : FAILURE in 1h 03m 32s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a6cfaaae045d49f0b593df18061f3288 : SUCCESS in 1h 02m 02s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/4d8f8c25bf0a42119234e21313c59984 : SUCCESS in 1h 30m 31s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/0f993db296204ba9883fc2722702cf6b : SUCCESS in 1h 35m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9762cf8b63664804afe41c202c210320 : SUCCESS in 59m 58s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3e3123048633493a8fc87138fc2f5912 : SUCCESS in 1h 12m 06s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/18ad515811c0475c81088c58cbc1f78e : SUCCESS in 38m 49s","accounts_in_message":[],"_revision_number":6},{"id":"1f4c731ee169432c1125673be70159c6d418e0e7","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-06-14 13:25:10.000000000","message":"Patch Set 6:\n\n(4 comments)","accounts_in_message":[],"_revision_number":6},{"id":"520b80fdd902b9d12fadaf254ecc435bb838168f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-06-15 08:24:04.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"3b761874ec3ad2384d6bd21ac7264e5935183e57","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-06-15 08:24:50.000000000","message":"Patch Set 7:\n\n(7 comments)","accounts_in_message":[],"_revision_number":7},{"id":"30455fd7b45db166924245a0c009a29dd5973c08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-15 10:14:02.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/bbe5241623e34afa8ea9b9bc0f5dc3b9 : SUCCESS in 54m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ff5a7a6fc87840579b102b244f28200f : SUCCESS in 1h 16m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4d480f8a0f2c4f9b9ae4fb2b4581c6d6 : SUCCESS in 45m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e397190ba194e65b850774cd4837861 : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bb9181c983d54335bd787d783008cead : SUCCESS in 6m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c904dc2735f14ff784f75a57c8a6d26a : SUCCESS in 6m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1dd9c8837e724b3db66d18317d8fd171 : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab48345b60a34df9b5d513d41e69713b : SUCCESS in 7m 27s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/f944d8685dfb46bcbe488e96acd268a5 : FAILURE in 12m 00s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/bd84cda1a93b4700b24318a49ba7253a : FAILURE in 8m 50s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/14402c5265b54704b2c79446c731f534 : FAILURE in 10m 09s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/57e94919ffa94b208a9c47a81e3aa6a0 : SUCCESS in 4m 45s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/82ed4a66d47b47ecb84469eff516e18a : SUCCESS in 1h 10m 23s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/87c5d4119ff3413385522762c7fb2d65 : SUCCESS in 1h 15m 11s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/1e79af740d994b39baaa7f456b173581 : SUCCESS in 1h 21m 23s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/98110f8990a94a93b0d149bba1b985a3 : SUCCESS in 1h 20m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8fc7392dbfcc4ec190efb06a512a85a6 : SUCCESS in 54m 45s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8031f2083da84367ba442197cd894573 : FAILURE in 1h 17m 19s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d11048110ee046c6a3ca809f293e6148 : SUCCESS in 34m 09s","accounts_in_message":[],"_revision_number":7},{"id":"cc1da356987950498e20dc10fa93f6beb0a24b50","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-06-16 05:04:15.000000000","message":"Patch Set 7:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/842070/842070/7/check-not-vote/ext-glance-zuul/0552696 : FAILURE in 21s","accounts_in_message":[],"_revision_number":7},{"id":"686542b2060ed1af67dd38d8d0cd78385010aa04","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-06-16 05:19:40.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"b505f77c24cd4c13c00212a574f9010d926a56be","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-06-16 05:22:03.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"cca63055aac34b591229eed832697e2e87aeaa34","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-06-16 13:30:45.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"1246f5766b57705d7d2d999cb59c306f2db02c24","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-06-16 14:00:16.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"f1b00e1e6a86b2e5b8eead89e5922a924da200d2","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-06-16 14:40:39.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"a342d1b5fb2e909e5434b5c95b13e9cab2d515fc","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-06-16 17:21:54.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"0a430a239e701567b85a7142a7063c775d83ee34","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-06-17 07:43:27.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"322ea7e2715ec5a7303c7ffea7271fe7e2fd5e4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-17 09:17:33.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/7adeb4ca00ab4371bedda7d32857f7d3 : SUCCESS in 55m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ef8d4cf49177410689651c358e8b5787 : SUCCESS in 1h 19m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/49cf1e12105c43a8bd44d90268437f88 : SUCCESS in 52m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40fb756f7433469d96bdd41c6fb3fdaf : SUCCESS in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/33baa8b220434b44a85db795739d7988 : SUCCESS in 5m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/71e1518ba31443f9a016f16fe7b3fe8f : SUCCESS in 6m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/456d77987c2245c1a85331e26f65380b : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1d23f5992394eae9ac6c7827d8acc7b : SUCCESS in 6m 24s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/d03160d4a9d9448ba2f6df2cfe40dd12 : SUCCESS in 12m 45s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/04344df942f346f9b72cf564cd24809e : SUCCESS in 10m 02s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/71c46ea9a15645c1b767bcd72bab5ea5 : SUCCESS in 8m 05s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/fe794a75d2854c8e924a41cab4942121 : SUCCESS in 5m 20s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/cd5e9806bf5e420ca02604d6e51ccb70 : SUCCESS in 1h 26m 27s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b7bb0577faf54691bf2748d9f50a8fbe : SUCCESS in 1h 23m 22s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/4b41b56f5cd644c680b2ca23ba4ef8a8 : SUCCESS in 1h 04m 33s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d3d0a563387d4921ae9d95f23d0b696b : SUCCESS in 1h 12m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fc4176e8e76441a597794c6e077ccd76 : SUCCESS in 47m 08s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/23b6a05b95b14ae69244c0c322ed1336 : SUCCESS in 1h 14m 45s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a9de245644324884853a41f4e3cfc4e9 : SUCCESS in 37m 07s","accounts_in_message":[],"_revision_number":8},{"id":"942242fa9feb7d6c87f2ad1bd4026e9ec992542c","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-06-20 06:51:41.000000000","message":"Patch Set 8:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/842070/842070/8/check-not-vote/ext-glance-zuul/4a8715b : FAILURE in 21s","accounts_in_message":[],"_revision_number":8},{"id":"829e09493ec4f57ed0d5375a692d6084309ff345","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-06-20 14:42:09.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"858762174a3c45dca8d0646b74ef783885fb9793","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-06-20 14:43:40.000000000","message":"Patch Set 9:\n\n(3 comments)","accounts_in_message":[],"_revision_number":9},{"id":"d83364cb32d964549da75d8737cc82009d324586","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 16:14:50.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/346f01c286da452a8fc3c7a2d9e3c8ce : SUCCESS in 56m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7a370bf38058433fbf56d2ff9a1f7f55 : SUCCESS in 1h 24m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/509d5ffb75d34941a1ab8129bde9c7cd : SUCCESS in 48m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3f3eb75353c4113be73d11eee4805d5 : SUCCESS in 5m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cc544e4dff374c99a70e87131c4513ac : SUCCESS in 5m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e2ba399ca63648bf9c173234635ee739 : SUCCESS in 6m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b6a5e9aa5ce14006836134550dd693a2 : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fa061a461fc4dd1bdb9745861f10a73 : SUCCESS in 6m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bf969b6fdb244492a31fd47f5977cce1 : SUCCESS in 5m 47s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/8a44a67f59ec48df9638e38e2e9ebd56 : RETRY_LIMIT in 6m 17s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/61591b62305643b397271f2d4e873726 : SUCCESS in 10m 23s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/dbdbf5f1e39b4146894db5e62c6e4ebd : SUCCESS in 8m 51s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/b98a47e09f3848cab0af19b63dd20fb1 : SUCCESS in 5m 19s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/7bed2017c41147b7a98448950d24d2d4 : SUCCESS in 1h 19m 43s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4f5427e2d90d44c9a877b0eaecbe3120 : SUCCESS in 1h 15m 54s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/4382843cc74c4db1afdec2ce5475f74f : SUCCESS in 1h 12m 18s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/5a73f837abb74def8f5d1fbd7d674dc9 : SUCCESS in 1h 25m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/73c428e54d504b20a47308cb89476293 : SUCCESS in 45m 40s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2ba82167cdad4458ad7b92c2e9d55349 : SUCCESS in 1h 12m 36s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0edd7a8466ac4ca193d76e5311dc4104 : SUCCESS in 36m 09s","accounts_in_message":[],"_revision_number":9},{"id":"104c67707cfcc1d47125eee90cb0448bd5fd64ee","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-06-21 05:57:03.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"852edece2711573b5244fe329fef93f4e0ebf40c","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-06-21 06:16:35.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"a9d3e091f589a923530ee40cf97a1fbdf96cbcc9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-21 07:51:03.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/fc285180f4634f879af0c6c271bcf18e : SUCCESS in 55m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6620503e6feb4780bcf5e92a7b1e2b83 : SUCCESS in 1h 14m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f2fd615a659249588abe0830cd80bcef : SUCCESS in 46m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/539d8734796e48629c194cd245668ae2 : SUCCESS in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4764c9b4a2cf4bca8b1fe32ec063fe95 : SUCCESS in 4m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3b10f8bbedf04b5ea19ca4d08ea7b14d : SUCCESS in 5m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3b3b05d6e967446caf7f6dd029544c46 : SUCCESS in 7m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce2ef6f745824e039bda27bc936d2fac : SUCCESS in 6m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2145d5de9dc04311be7106e40d7161bd : SUCCESS in 6m 10s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/0de7742a12504fff95a538da91ee2989 : RETRY_LIMIT in 5m 41s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a28744c9c09840938b82c6b606c8bac6 : SUCCESS in 8m 36s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/f3ca1b927d484532846a3d67f28d8f85 : SUCCESS in 8m 10s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/d0bb648cb4f94e11a45dfe5e2ec3380f : SUCCESS in 4m 57s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/ff81d9bc98d843bb84bc359a8d09fbe0 : FAILURE in 1h 48m 40s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4943f761aee14c00bd83c8bb07b37284 : SUCCESS in 1h 13m 17s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/42bc2305f7b54be7b070e49103ccc8da : SUCCESS in 1h 12m 42s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c0ccf18c528f417180caa8399f00283f : SUCCESS in 1h 17m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/21494e23d07d4a8ebeda713200e45198 : SUCCESS in 55m 24s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1b1539ef2e684f42aba48607ed4a47bc : SUCCESS in 1h 12m 00s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/281a80d3f56749ed946a6e57595dd0f7 : FAILURE in 20m 37s","accounts_in_message":[],"_revision_number":9},{"id":"ccb2f805c323665254dc082814f6bfdf48d9f399","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-06-21 15:24:21.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"6d2b96d46d9f94881fdcde8d2506249b6efc2a81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-21 16:58:00.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/b13a8f1b529245c8bdd495c5f25302e8 : SUCCESS in 47m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c4ce8535a7b147248ea495a9ceaf6d26 : SUCCESS in 57m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7ccd7e3262ae4788bcafd07c871275f0 : SUCCESS in 38m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24ef122aa6d549e790780c1f4536b9fd : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d04ec2fd478945c4a81a20a426935e74 : SUCCESS in 5m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/19bebb62f5144e43b808dc00623fba30 : SUCCESS in 5m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6b4f67210dc94ff7bfae09f9539b7765 : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27620b809e6e4ee59bbbdaea5b709c7c : SUCCESS in 8m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e823338eb1924f0bbdf12e44d9922e61 : SUCCESS in 5m 17s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/fc0c065fb55e49958f0157941342ddac : RETRY_LIMIT in 6m 54s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/bd255905c7c445b69052beb84be9a745 : SUCCESS in 9m 13s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/8dce7ee38e034835a3b1da87d548dc32 : SUCCESS in 8m 12s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/62b787471d254d4a9283e485f4f380a1 : SUCCESS in 4m 31s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c5e6223f34774c65bfdd2270cb7e3f27 : FAILURE in 1h 32m 04s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/8cdb1ebff6674e42bf39ac522707a410 : SUCCESS in 1h 08m 15s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/94d0aae66d9c49348c6e34a6d6c42574 : SUCCESS in 1h 12m 23s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f4f7668bd5e44677bba0adfad5ef5933 : SUCCESS in 1h 04m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0e2f880ae51b4697bc2cc28430104223 : SUCCESS in 53m 47s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9cba72c101714ef19a2329b5c3956b10 : SUCCESS in 1h 07m 43s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ae8fb709e14241f58580725790667cda : SUCCESS in 23m 55s","accounts_in_message":[],"_revision_number":9},{"id":"8cfb62a7544df72d7a723f5518ecfe22b64eb24f","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-06-24 05:34:45.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"b71dd11d5df28f0a32c4298da1eefe4aff33f476","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-24 06:58:57.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/5027064293df4d36b7068677bc2a6a89 : SUCCESS in 43m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3afbe73543aa4be98ae289d967d463f4 : SUCCESS in 1h 01m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7c3764d29bc74dbabc251fdc8b01978a : SUCCESS in 40m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a10a401429fa4a67b518c234c2ba0d00 : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a97894f40b954ad2aab36b872d1e4f9c : SUCCESS in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3c88fbc7974e44098d1827cb93a113b8 : SUCCESS in 5m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/99f81e69f8b34f91a2e42b465684445c : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/117225744cdc4e899fe044264ae96fc7 : SUCCESS in 5m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7fc0e5d123aa42e39ae0411ab941f54c : SUCCESS in 5m 41s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/df6df84273394f2aa25d2eaf3ed1ef6d : SUCCESS in 11m 47s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/81ac80728b694e6a8de254553ea1187f : SUCCESS in 8m 26s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/e6fba5df35e84c6db59d1bfe0c0c80e2 : SUCCESS in 7m 53s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/6c31bfac102a481c89536acf6c7d8fbf : SUCCESS in 4m 29s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/0048d0b17cb4456e93f6829ac8eb9e4d : SUCCESS in 1h 00m 46s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e7ab0d51fd71401b945bba86bf638283 : SUCCESS in 1h 16m 40s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/fed87cab3510452ea23170a83e357040 : SUCCESS in 1h 13m 29s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a2b9cbfcc17a40668fc295dbccb204f0 : SUCCESS in 1h 16m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/11c7ccc540fb406dacbd430e4affa093 : SUCCESS in 48m 58s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/073d9645982641bb913f89095c04a7aa : SUCCESS in 1h 06m 09s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7e2b3ac5b05448bb9450e46b49de7ca1 : SUCCESS in 32m 40s","accounts_in_message":[],"_revision_number":9},{"id":"10ebdc402756edf64d955c91556e82c955517737","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-07-05 14:14:17.000000000","message":"Patch Set 9: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":9},{"id":"4b49db24d72656000a79992851498b0b97207aeb","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-07-06 06:02:00.000000000","message":"Patch Set 9:\n\n(6 comments)","accounts_in_message":[],"_revision_number":9},{"id":"53e626481189bf2ca6bb2cb3aea320358766048f","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-07-11 18:22:02.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"4c8959be615332a76b1f5652dc6f697a8fd1986e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-07-12 06:40:00.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"489fd9d92c6927783c1e024803bfbe6e9c9a8e62","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-07-12 06:40:01.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"1167837fe76e663646896aa5018b1032d3c0bc1d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-07-12 06:40:43.000000000","message":"Patch Set 10:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/842070/842070/10/check-not-vote/ext-glance-zuul/def8051 : FAILURE in 16s","accounts_in_message":[],"_revision_number":10},{"id":"6c926a3ecc72ab486f9b3a509c16d9062d5c48ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 08:21:54.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5737b33884be4e3bbf918816a67b2719\n\n- grenade https://zuul.opendev.org/t/openstack/build/c1a5c890964c46698140dc12d42e5d4f : SUCCESS in 1h 04m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bc52d5da0f124508b73cd68f4b569b2e : SUCCESS in 1h 29m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/84cbdf2f07c649558ceef62c3349b720 : SUCCESS in 51m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a91d0a634955462d954dc96c683284e4 : SUCCESS in 7m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b30e71f53d624de2b76fe62f06dbe4e0 : SUCCESS in 5m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5a10b17d2f3c40c585d5284fcca47528 : SUCCESS in 7m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0473f3f000934b72a47e277673d3648d : SUCCESS in 7m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e16a780850ff4fbca3edf1b77c3d1da0 : SUCCESS in 7m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9d27f9aa26a9410caa1846afa7ddf7a1 : SUCCESS in 6m 09s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/bdd5af6f4ab3446898cbe3755e287e1f : SUCCESS in 12m 42s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7bc2f5a5d9784dbf8092bb30b36ea00e : SUCCESS in 9m 31s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/ee449c6caf444d8ca4cf2e0807927e2e : SUCCESS in 10m 57s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/11e97eb41170458ea4203e78fa4b5d3e : SUCCESS in 5m 38s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/fbc60c6d8955493f9636a4dfa4d4b58d : SUCCESS in 1h 18m 20s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/90b6039e25304011876e5848b2b5420c : SUCCESS in 1h 16m 40s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/733899dd0f3543d49ee51c310d29b82f : SUCCESS in 1h 14m 58s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/12e8225f29624415a4bc2647b1a4e23c : SUCCESS in 1h 36m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ab1e023b55fc4f89aee432b714ddf8d7 : SUCCESS in 41m 40s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4c15a418a2164cacba8c59b49a3458b6 : SUCCESS in 1h 23m 54s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1095d969a9254c8bac3eb0fdfd235380 : SUCCESS in 36m 39s","accounts_in_message":[],"_revision_number":10},{"id":"5a23cf043c71c051b043aa786c81ec9f793910ec","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-07-15 15:27:16.000000000","message":"Patch Set 10: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"9780ce991f245ba6f453fc2c7eb445b004879278","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-07-15 16:08:00.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"21272bfd2651b0ca6f2118f009e5f6eb5aa595ff","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-07-15 16:10:48.000000000","message":"Patch Set 11:\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"c39a966c33359f8919ae2c523be4e58349dc033a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 17:46:30.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27e929ae7d0a422da0e4d46a7872b4a8\n\n- grenade https://zuul.opendev.org/t/openstack/build/e9294a2ae2384193abc8e1d9d1be173e : SUCCESS in 1h 05m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/22f51ce3dd3742a69d79fa5ef1d3bbbe : SUCCESS in 1h 31m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2c5d8b68461e4ea2b7e3fe2edeef2db9 : SUCCESS in 56m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6daaa079c854919ab8fa0e79401a109 : SUCCESS in 5m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/334f99e18e5a4d2c8be62e59f9cac938 : SUCCESS in 6m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/48c2325ebaf24196a1cb509f9aa09e43 : SUCCESS in 5m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3e32dedc6aea4f40b3e19de7df58df2b : SUCCESS in 8m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4535a0e3f19d4d82b544be12b77c7b00 : SUCCESS in 9m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/870a4c4ded1b4bcdbbde907e8d0e6ae3 : SUCCESS in 7m 15s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/7db5fcad0f614154b16c97c9bb2798f3 : SUCCESS in 14m 57s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4309ffaa07b64e268d56b57d1a3b6b60 : SUCCESS in 11m 34s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/797d9f066d7449388f567480088ace80 : SUCCESS in 9m 40s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/5bae61f754e048118826abaa728942a6 : SUCCESS in 4m 26s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/61fc101a73894383ab2e5ca8bbac4673 : SUCCESS in 1h 31m 35s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/cb11a4bfb89f4cd0b5eb4b35acd13de8 : SUCCESS in 1h 12m 23s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/a25b13231d95473fb338f452b9a563d1 : SUCCESS in 1h 36m 07s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/952ac76e29ed418abf39a336ea8ad1ec : SUCCESS in 1h 23m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/276619507b8b403684174c7a0a63a41f : SUCCESS in 1h 06m 10s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/59e1c07c2adb428dbdc25da526146f91 : SUCCESS in 1h 05m 02s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2973f803b259451ba8c30bc714ad2fed : SUCCESS in 45m 37s","accounts_in_message":[],"_revision_number":11},{"id":"316896730a1a8c61b6e2918831c8ae5ac831a8ab","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-07-18 06:12:39.000000000","message":"Patch Set 11: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"d234c763f92ab705edd045a62accb5676a74afc6","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-07-18 14:57:48.000000000","message":"Patch Set 11:\n\n(4 comments)","accounts_in_message":[],"_revision_number":11},{"id":"91393c4742248969e485e1c758040362d4caddf3","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-07-18 16:27:41.000000000","message":"Patch Set 11:\n\n(3 comments)","accounts_in_message":[],"_revision_number":11},{"id":"e757d8d04ab003c84fa2e1d95da76177332c0e93","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-07-18 16:28:50.000000000","message":"Patch Set 11: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"11adeaf1c8e7f7b69acfa5535c16cd4ddbcf50e4","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-07-18 16:39:41.000000000","message":"Patch Set 11:\n\n(6 comments)","accounts_in_message":[],"_revision_number":11},{"id":"f9137fed4a55ef200d0627142ac043f2c64000bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-07-18 16:39:57.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"1d2a2cc60c40cc1f59ca05a4dd3bb683dd5d1f4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-18 18:21:40.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5269d4df344c4869991e6beaf3d228ac\n\n- grenade https://zuul.opendev.org/t/openstack/build/c152593af0634abc8898dd6bf17ffb7a : SUCCESS in 1h 03m 04s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2eda324f58564c6aa2c8b635664f8102 : SUCCESS in 1h 18m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9bec9ca712e6403fbfc3c3d4b3a9b86e : SUCCESS in 52m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd6065bd245a40d188b1ca0bb43d4896 : FAILURE in 6m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/44f51bf0ac174c56827c444e81545cb3 : SUCCESS in 6m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6b198f0dc1d741509966b1390d75750a : SUCCESS in 7m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/137a83e66d11437980e7e0c66f8e5924 : SUCCESS in 7m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8a9248f8599485498c1719657278ef4 : SUCCESS in 7m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/29649aef45df48abb6ea3878a937e051 : SUCCESS in 6m 26s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/c9070bb8b71a47878543a315fdacc52e : SUCCESS in 11m 52s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7b0e4dff501f42209b629cff87cd5672 : SUCCESS in 9m 55s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/351339f43c104b52ac3473b9fd60269e : SUCCESS in 10m 08s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/a686bc8d668349a187e00d1d3b826bf6 : SUCCESS in 5m 43s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/7999de83354744b4975ed58cccaeb825 : SUCCESS in 1h 19m 55s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/bb7ad97f1bda4179a6b8f93ce309987d : SUCCESS in 1h 24m 30s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/fbba27e05c11452981b6a747b2b4d7ce : SUCCESS in 1h 21m 58s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c596c9b5ac644bc9a666f3864298ce6f : SUCCESS in 1h 34m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9a87fddc49c040439b94964b443b372b : SUCCESS in 51m 59s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f952dbef34034cb2a95a8ed4c5804f22 : SUCCESS in 1h 16m 24s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e3a3219ba1bb4df6949b73ab610be311 : SUCCESS in 39m 50s","accounts_in_message":[],"_revision_number":12},{"id":"b25b64c5d2d9e8979d400da2304664ec73df94f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-07-19 05:14:39.000000000","message":"Patch Set 13: Published edit on patch set 12.","accounts_in_message":[],"_revision_number":13},{"id":"8fe6842afa9611d763bc6b39e9f5a872900167d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-19 07:07:45.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58fa13a3a0a648bd9c8d0cdaa3f3a5e7\n\n- grenade https://zuul.opendev.org/t/openstack/build/127f9434879841f3a4896383ef26f8d4 : SUCCESS in 1h 05m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a273c82013784dbb98bdeef17a8eda48 : SUCCESS in 1h 38m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2fdfa46b0a3a4b258029928e71203d56 : SUCCESS in 1h 02m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36f52ffa2e534aedb0d5330057b18c22 : SUCCESS in 7m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2af404a9b7fd4938b7f9bd2a75bc4429 : SUCCESS in 7m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/784f67959c99414289a96364baa799a0 : SUCCESS in 8m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bb1e03cef2ff43f5aafa19bac493d258 : SUCCESS in 6m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7076ceabcd5349b694d4d322f091a752 : SUCCESS in 9m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/73d77c5816654c30a50859b11e210ec9 : SUCCESS in 8m 06s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/88c42376229044209f03e0bec6b70f06 : SUCCESS in 14m 34s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2974f1e3ca3e42a89e8bdf5c92982c61 : SUCCESS in 11m 00s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/72c59cac808d44e0bf39c0c1e0ccf9c4 : SUCCESS in 11m 55s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/0f1d40fb4fae4713b982b3bf85212f5a : SUCCESS in 7m 21s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/776ad7a260664afdb212104e0e25f60f : SUCCESS in 1h 37m 24s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/25cccb64433c41e29ed29d031425817a : SUCCESS in 1h 47m 13s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/af4043cfc0f6440ea2566c9d8baf5c48 : SUCCESS in 1h 05m 50s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/4bc52e4c0a134405ac9d5b6e53ee5f9f : SUCCESS in 1h 45m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/054f7e554df94a3a95b656d93c0c7588 : SUCCESS in 56m 48s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/97170609567a4cdda33cabda295c7cb4 : SUCCESS in 1h 30m 54s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/41f19d4a8b2745fda1ced7909e0660b5 : SUCCESS in 29m 40s","accounts_in_message":[],"_revision_number":13},{"id":"c5becc1ca1f1f5583b0503a2501007528b22553f","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-07-19 12:07:43.000000000","message":"Patch Set 13: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"84e3ed853750c92e4e573fd6fc227af2ca7b9659","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-07-19 14:23:14.000000000","message":"Patch Set 13: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":13},{"id":"4b4657bc01d57b07f74e3099dcc3c5e6e07f4066","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-07-19 14:32:42.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"de55fafa7138a4d9ea8ebf194667f023ad90ff28","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-07-19 14:57:52.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"3786095cecff1745bd12282395386035b35703de","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-07-19 15:30:45.000000000","message":"Patch Set 13: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"b9936fe93949bc9d608fa79b1d80ed7a383034df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-19 15:31:41.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"7e9e0bf76e5e01a6ac59dd714625bfeae4b1df80","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-19 17:01:13.000000000","message":"Patch Set 13: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63372494705241dba83df32d560cf4ef\n\n- grenade https://zuul.opendev.org/t/openstack/build/11d7973441024bd69555d7af9b3e9046 : SUCCESS in 55m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/04e4fca11e0747ca86bb8a17be3ede0d : SUCCESS in 1h 21m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b82e1c854a1a4d6f9c147268b8b7b771 : SUCCESS in 52m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4825492d7e434e53a2966cc7a950fdc0 : SUCCESS in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5001b2483cd04c3187e164f1908ca088 : SUCCESS in 6m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0f7a665827c44e8caaf19d68ed7aaf66 : SUCCESS in 7m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd0ea2ea2ee04e2c817d0e75ec7c7c11 : SUCCESS in 8m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8bfb5320856c4664a6d55e965ad4d3ad : SUCCESS in 6m 38s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9e607bd4581c4800825408a7d1938a62 : SUCCESS in 9m 00s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2297c22072094f60a84d4ce3defb5d04 : SUCCESS in 12m 43s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/128530996b584e9daf2337493bd2ddb8 : SUCCESS in 1h 15m 02s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/de1d20b73bfe43a3b92c53804bada7eb : SUCCESS in 1h 28m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1bfc340b552e410f82f71619ef93c2b9 : SUCCESS in 1h 06m 10s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4df6d8a35af048aaad258671e519c755 : SUCCESS in 1h 05m 03s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3ab60483d856442d8a98e62bd1a6e8d3 : SUCCESS in 50m 19s","accounts_in_message":[],"_revision_number":13},{"id":"d7ca52e62f9dae3942d3fcb1ab944c848ba69a1d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-19 17:01:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":13},{"id":"1735735bea718a0d08fda95cfa7cd66b81f3e714","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-19 17:02:32.000000000","message":"Patch Set 13:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9446b925bfa4fd9bfc0b2158689dec5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e526fdbd576477296a11c051d188220 : SUCCESS in 56s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b9bfe1da0f264bf1aefca1c5fa0e14d4 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":13},{"id":"9eab077b7d85034d2414d63d7a067e0e463dc97d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-09-07 14:47:37.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"c1a2b9e2217498d86948a9be1b141483e5185453","revisions":{"9466ab9734d4f18cdf31af9fb5ec40f826de2cf6":{"kind":"REWORK","_number":1,"created":"2022-05-17 09:46:25.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/70/842070/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/70/842070/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/842070/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/842070/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/842070/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/70/842070/1"}}},"commit":{"parents":[{"commit":"35e6c57bc90365796f45ada9271e532068557b8a","subject":"Fix failing namespace list delete race","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/35e6c57bc90365796f45ada9271e532068557b8a"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2022-05-17 09:41:00.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2022-05-17 09:45:38.000000000","tz":0},"subject":"[WIP] Immediate caching of an image","message":"[WIP] Immediate caching of an image\n\nChange-Id: I9ab3f1b7595e22dbb03af95168314352a44eb930\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9466ab9734d4f18cdf31af9fb5ec40f826de2cf6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9466ab9734d4f18cdf31af9fb5ec40f826de2cf6"}]},"branch":"refs/heads/master"},"914b56a94729e18de8b35c099f9255c33d523e18":{"kind":"REWORK","_number":2,"created":"2022-05-19 07:29:07.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/70/842070/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/70/842070/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/842070/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/842070/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/842070/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/70/842070/2"}}},"commit":{"parents":[{"commit":"84c7ba3adfef7b486ae4ee098f67e59608fb6715","subject":"Merge \"Release notes for Zed Milestone 1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/84c7ba3adfef7b486ae4ee098f67e59608fb6715"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2022-05-17 09:41:00.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2022-05-19 07:27:47.000000000","tz":0},"subject":"[WIP] Immediate caching of an image","message":"[WIP] Immediate caching of an image\n\nChange-Id: I9ab3f1b7595e22dbb03af95168314352a44eb930\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/914b56a94729e18de8b35c099f9255c33d523e18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/914b56a94729e18de8b35c099f9255c33d523e18"}]},"branch":"refs/heads/master"},"7e43a706b27df6ebf9c55b1cbd05176b4e53508c":{"kind":"REWORK","_number":3,"created":"2022-06-10 17:14:53.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/70/842070/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/70/842070/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/842070/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/842070/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/842070/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/70/842070/3"}}},"commit":{"parents":[{"commit":"84c7ba3adfef7b486ae4ee098f67e59608fb6715","subject":"Merge \"Release notes for Zed Milestone 1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/84c7ba3adfef7b486ae4ee098f67e59608fb6715"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2022-05-17 09:41:00.000000000","tz":0},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2022-06-10 17:14:42.000000000","tz":-420},"subject":"[WIP] Immediate caching of an image","message":"[WIP] Immediate caching of an image\n\nChange-Id: I9ab3f1b7595e22dbb03af95168314352a44eb930\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7e43a706b27df6ebf9c55b1cbd05176b4e53508c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7e43a706b27df6ebf9c55b1cbd05176b4e53508c"}]},"branch":"refs/heads/master"},"39981c88b992250e66efd7a8ef3f60eddc1f0415":{"kind":"REWORK","_number":4,"created":"2022-06-13 09:06:29.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/70/842070/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/70/842070/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/842070/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/842070/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/842070/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/70/842070/4"}}},"commit":{"parents":[{"commit":"ebef97d776eb1982b948658eaf20566853dc70a9","subject":"Merge \"Modify wrong file path in README\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ebef97d776eb1982b948658eaf20566853dc70a9"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2022-05-17 09:41:00.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2022-06-13 09:03:40.000000000","tz":0},"subject":"[WIP] Immediate caching of an image","message":"[WIP] Immediate caching of an image\n\nChange-Id: I9ab3f1b7595e22dbb03af95168314352a44eb930\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/39981c88b992250e66efd7a8ef3f60eddc1f0415"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/39981c88b992250e66efd7a8ef3f60eddc1f0415"}]},"branch":"refs/heads/master"},"09f421ec74fdafba9394c39458ed097d7c9fb26d":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2022-06-13 09:09:07.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/70/842070/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/70/842070/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/842070/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/842070/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/842070/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/70/842070/5"}}},"commit":{"parents":[{"commit":"ebef97d776eb1982b948658eaf20566853dc70a9","subject":"Merge \"Modify wrong file path in README\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ebef97d776eb1982b948658eaf20566853dc70a9"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2022-05-17 09:41:00.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2022-06-13 09:09:07.000000000","tz":0},"subject":"[WIP] Immediate caching of an image","message":"[WIP] Immediate caching of an image\n\nCo-Author: Dan Smith \u003cdms@danplanet.com\u003e\nChange-Id: I9ab3f1b7595e22dbb03af95168314352a44eb930\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/09f421ec74fdafba9394c39458ed097d7c9fb26d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/09f421ec74fdafba9394c39458ed097d7c9fb26d"}]},"branch":"refs/heads/master","description":"Edit commit message"},"7b457f245b3c57977915c764dffc3a224da1fd7e":{"kind":"REWORK","_number":6,"created":"2022-06-13 17:48:55.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/70/842070/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/70/842070/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/842070/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/842070/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/842070/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/70/842070/6"}}},"commit":{"parents":[{"commit":"ebef97d776eb1982b948658eaf20566853dc70a9","subject":"Merge \"Modify wrong file path in README\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ebef97d776eb1982b948658eaf20566853dc70a9"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2022-05-17 09:41:00.000000000","tz":0},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2022-06-13 17:48:34.000000000","tz":-420},"subject":"[WIP] Immediate caching of an image","message":"[WIP] Immediate caching of an image\n\nCo-Author: Dan Smith \u003cdms@danplanet.com\u003e\nChange-Id: I9ab3f1b7595e22dbb03af95168314352a44eb930\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7b457f245b3c57977915c764dffc3a224da1fd7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7b457f245b3c57977915c764dffc3a224da1fd7e"}]},"branch":"refs/heads/master"},"337ecb34bca64dccd4b9b6b4340e19aac6737956":{"kind":"REWORK","_number":7,"created":"2022-06-15 08:24:04.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/70/842070/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/70/842070/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/842070/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/842070/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/842070/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/70/842070/7"}}},"commit":{"parents":[{"commit":"ebef97d776eb1982b948658eaf20566853dc70a9","subject":"Merge \"Modify wrong file path in README\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ebef97d776eb1982b948658eaf20566853dc70a9"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2022-05-17 09:41:00.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2022-06-15 08:16:58.000000000","tz":0},"subject":"Immediate caching of an image","message":"Immediate caching of an image\n\nTill now glance was dependent on periodic job to get image\ncached locally and also has dependency on ``cachemanage``\nmiddleware for the same.\n\nThis patch removes the periodic job and dependency of\n``cachemanage`` middleware and initiates immediate caching\nof an image via ``PUT /cache/image_id`` API call.\n\nCo-Author: Dan Smith \u003cdms@danplanet.com\u003e\n\nImplements: blueprint instant_caching\nChange-Id: I9ab3f1b7595e22dbb03af95168314352a44eb930\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/337ecb34bca64dccd4b9b6b4340e19aac6737956"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/337ecb34bca64dccd4b9b6b4340e19aac6737956"}]},"branch":"refs/heads/master"},"e90896509b89669f82129f78b8712ca9af7c255d":{"kind":"REWORK","_number":8,"created":"2022-06-17 07:43:27.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/70/842070/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/70/842070/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/842070/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/842070/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/842070/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/70/842070/8"}}},"commit":{"parents":[{"commit":"ebef97d776eb1982b948658eaf20566853dc70a9","subject":"Merge \"Modify wrong file path in README\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ebef97d776eb1982b948658eaf20566853dc70a9"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2022-05-17 09:41:00.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2022-06-17 07:40:04.000000000","tz":0},"subject":"Immediate caching of an image","message":"Immediate caching of an image\n\nTill now glance was dependent on periodic job to get image\ncached locally and also has dependency on ``cachemanage``\nmiddleware for the same.\n\nThis patch removes the periodic job and dependency of\n``cachemanage`` middleware and initiates immediate caching\nof an image via ``PUT /cache/image_id`` API call.\n\nCo-Author: Dan Smith \u003cdms@danplanet.com\u003e\n\nImplements: blueprint instant_caching\nChange-Id: I9ab3f1b7595e22dbb03af95168314352a44eb930\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e90896509b89669f82129f78b8712ca9af7c255d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e90896509b89669f82129f78b8712ca9af7c255d"}]},"branch":"refs/heads/master"},"bf226965af00bb989aec55ae72ef210c5b153cde":{"kind":"REWORK","_number":9,"created":"2022-06-20 14:42:09.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/70/842070/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/70/842070/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/842070/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/842070/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/842070/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/70/842070/9"}}},"commit":{"parents":[{"commit":"ebef97d776eb1982b948658eaf20566853dc70a9","subject":"Merge \"Modify wrong file path in README\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ebef97d776eb1982b948658eaf20566853dc70a9"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2022-05-17 09:41:00.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2022-06-20 14:38:39.000000000","tz":0},"subject":"Immediate caching of an image","message":"Immediate caching of an image\n\nTill now glance was dependent on periodic job to get image\ncached locally and also has dependency on ``cachemanage``\nmiddleware for the same.\n\nThis patch removes the periodic job and dependency of\n``cachemanage`` middleware and initiates immediate caching\nof an image via ``PUT /cache/image_id`` API call.\n\nCo-Author: Dan Smith \u003cdms@danplanet.com\u003e\n\nImplements: blueprint instant_caching\nChange-Id: I9ab3f1b7595e22dbb03af95168314352a44eb930\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bf226965af00bb989aec55ae72ef210c5b153cde"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bf226965af00bb989aec55ae72ef210c5b153cde"}]},"branch":"refs/heads/master"},"76570684caf6083e517f88c20d6485b3fac284bd":{"kind":"REWORK","_number":10,"created":"2022-07-12 06:40:00.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/70/842070/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/70/842070/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/842070/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/842070/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/842070/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/70/842070/10"}}},"commit":{"parents":[{"commit":"be9da314e724604723f2b62432595daf4a728e48","subject":"Merge \"doc: update the kvm kw_vif_model driver list\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/be9da314e724604723f2b62432595daf4a728e48"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2022-05-17 09:41:00.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2022-07-12 06:39:11.000000000","tz":0},"subject":"Immediate caching of an image","message":"Immediate caching of an image\n\nTill now glance was dependent on periodic job to get image\ncached locally and also has dependency on ``cachemanage``\nmiddleware for the same.\n\nThis patch removes the periodic job and dependency of\n``cachemanage`` middleware and initiates immediate caching\nof an image via ``PUT /cache/image_id`` API call.\n\nCo-Author: Dan Smith \u003cdms@danplanet.com\u003e\n\nImplements: blueprint instant_caching\nChange-Id: I9ab3f1b7595e22dbb03af95168314352a44eb930\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/76570684caf6083e517f88c20d6485b3fac284bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/76570684caf6083e517f88c20d6485b3fac284bd"}]},"branch":"refs/heads/master"},"1dbd0193601c773adbc4b3806aa4c002e698deea":{"kind":"REWORK","_number":11,"created":"2022-07-15 16:08:00.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/70/842070/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/70/842070/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/842070/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/842070/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/842070/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/70/842070/11"}}},"commit":{"parents":[{"commit":"be9da314e724604723f2b62432595daf4a728e48","subject":"Merge \"doc: update the kvm kw_vif_model driver list\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/be9da314e724604723f2b62432595daf4a728e48"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2022-05-17 09:41:00.000000000","tz":0},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2022-07-15 16:07:47.000000000","tz":-420},"subject":"Immediate caching of an image","message":"Immediate caching of an image\n\nTill now glance was dependent on periodic job to get image\ncached locally and also has dependency on ``cachemanage``\nmiddleware for the same.\n\nThis patch removes the periodic job and dependency of\n``cachemanage`` middleware and initiates immediate caching\nof an image via ``PUT /cache/image_id`` API call.\n\nCo-Author: Dan Smith \u003cdms@danplanet.com\u003e\n\nImplements: blueprint instant_caching\nChange-Id: I9ab3f1b7595e22dbb03af95168314352a44eb930\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1dbd0193601c773adbc4b3806aa4c002e698deea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1dbd0193601c773adbc4b3806aa4c002e698deea"}]},"branch":"refs/heads/master"},"b83e68cfe60b63a449e43aa6b9c95b67c517c724":{"kind":"REWORK","_number":12,"created":"2022-07-18 16:39:57.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/70/842070/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/70/842070/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/842070/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/842070/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/842070/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/70/842070/12"}}},"commit":{"parents":[{"commit":"95a02faf9c0f87fd553669aa71e0e6a6c204f801","subject":"Merge \"Include os-brick configuration options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/95a02faf9c0f87fd553669aa71e0e6a6c204f801"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2022-05-17 09:41:00.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2022-07-18 16:38:22.000000000","tz":0},"subject":"Immediate caching of an image","message":"Immediate caching of an image\n\nTill now glance was dependent on periodic job to get image\ncached locally and also has dependency on ``cachemanage``\nmiddleware for the same.\n\nThis patch removes the periodic job and dependency of\n``cachemanage`` middleware and initiates immediate caching\nof an image via ``PUT /cache/image_id`` API call.\n\nCo-Author: Dan Smith \u003cdms@danplanet.com\u003e\n\nImplements: blueprint instant-caching\nChange-Id: I9ab3f1b7595e22dbb03af95168314352a44eb930\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b83e68cfe60b63a449e43aa6b9c95b67c517c724"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b83e68cfe60b63a449e43aa6b9c95b67c517c724"}]},"branch":"refs/heads/master"},"c1a2b9e2217498d86948a9be1b141483e5185453":{"kind":"REWORK","_number":13,"created":"2022-07-19 05:14:39.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/70/842070/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/70/842070/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/842070/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/842070/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/842070/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/70/842070/13"}}},"commit":{"parents":[{"commit":"95a02faf9c0f87fd553669aa71e0e6a6c204f801","subject":"Merge \"Include os-brick configuration options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/95a02faf9c0f87fd553669aa71e0e6a6c204f801"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2022-05-17 09:41:00.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2022-07-19 05:14:38.000000000","tz":0},"subject":"Immediate caching of an image","message":"Immediate caching of an image\n\nTill now glance was dependent on periodic job to get image\ncached locally and also has dependency on ``cachemanage``\nmiddleware for the same.\n\nThis patch removes the periodic job and dependency of\n``cachemanage`` middleware and initiates immediate caching\nof an image via ``PUT /cache/image_id`` API call.\n\nCo-Author: Dan Smith \u003cdms@danplanet.com\u003e\n\nImplements: blueprint instant-caching\nChange-Id: I9ab3f1b7595e22dbb03af95168314352a44eb930\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c1a2b9e2217498d86948a9be1b141483e5185453"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c1a2b9e2217498d86948a9be1b141483e5185453"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
