)]}'
{"id":"openstack%2Fglance~842162","triplet_id":"openstack%2Fglance~stable%2Fxena~I5149df76943c1c19f3204b904c0e2d3ef846bdf7","project":"openstack/glance","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2022-06-02 14:51:10.000000000","reason":"Brian Rosmaita replied on the change"},"32755":{"account":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"last_update":"2022-06-02 18:43:23.000000000","reason":"Change was submitted"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2022-06-02 18:43:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5149df76943c1c19f3204b904c0e2d3ef846bdf7","subject":"Fix set_property_atomic() boolean type casting","status":"MERGED","created":"2022-05-17 18:04:10.000000000","updated":"2022-06-02 18:44:56.000000000","submitted":"2022-06-02 18:43:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"842162","meta_rev_id":"3b1db3900bab165f3cdd840891a79812d8f435d1","_number":842162,"virtual_id_number":842162,"owner":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-02 18:43:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2022-06-02 14:51:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2022-06-02 15:12:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2022-06-02 15:12:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}],"reviewers":{"CC":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-17 18:04:10.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"CC"},{"updated":"2022-05-17 18:06:34.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2022-05-17 18:06:34.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2022-05-17 19:23:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-18 12:14:52.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"CC"},{"updated":"2022-06-02 15:12:44.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"a8060206de6d66d92987d706bb0ac09c1329acab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2022-05-17 18:04:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"706192964205229f958700851da47302456c7287","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-17 19:23:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/f5c300e1622c41b7b65706181dacc87b : SUCCESS in 1h 00m 25s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/0e8bd8020d35421fa7ba8b275fb8566d : SUCCESS in 1h 01m 18s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bcd239f01fe14e889b682da44194c3a6 : SUCCESS in 1h 13m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2a5f06122abd46bc96329624cb06f08d : SUCCESS in 34m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e211d9729028486ca344748a799ea26d : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1398706615504019b55e9577056deadf : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1e790789e5594cefb69d53b3748f254a : SUCCESS in 6m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4e3e19ec928b41a69eef1359022d80fa : SUCCESS in 7m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85e4817997314809ad7c8dd9a5cb85a3 : SUCCESS in 5m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4df2d9695ef041ca8784c57ae0e051a9 : SUCCESS in 8m 39s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1f6405bcb8524784bea43e236eb2f4f4 : SUCCESS in 8m 34s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/3323b46692ac4578b1110ab952c2e4e9 : SUCCESS in 8m 52s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/bcb20804aa1c4bab817a02a26ee4a6e9 : SUCCESS in 4m 24s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/ec5337bb27164785b2052de88e00940e : SUCCESS in 1h 09m 00s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/2838f8dd6dcf423c847f7c2212438869 : SUCCESS in 1h 05m 45s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/3ea180f2df4f46db865b01786125b23d : SUCCESS in 1h 13m 10s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/9511f3389cb8492bbf2b30ac0002f72b : SUCCESS in 1h 08m 48s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/24652ceb319e476498e9a92b30a1bd14 : SUCCESS in 55m 42s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/818fc3597e6e4057aae980cbe6aec7ad : SUCCESS in 1h 07m 27s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/22d64eb6f4384d1f90c0747baa890454 : FAILURE in 39m 49s","accounts_in_message":[],"_revision_number":1},{"id":"42d6a5236e77537b8e77d99e5ed5240b1bdf5a4f","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2022-05-18 12:14:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"34c7dfd64817777c27b5a13dd1298cb374970027","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-18 13:31:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/c0eac11249c14a17b5f16c89057528be : SUCCESS in 54m 44s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/f6a9f5c1017341f1861954bd84a38bdc : SUCCESS in 56m 42s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4568171ad21d44258a644bde6613a981 : SUCCESS in 57m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4638d89eec8e451aa64914d064d654d5 : SUCCESS in 41m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/396280c3ca8342b399b79ecc14bf8ba6 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c16b7669cbd348cc8bbc57e964f01d60 : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cf7fb6f6fa1e4eb584b4e4d8b1a36a32 : SUCCESS in 4m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/45fc08c1ea20441ba692266f890d9058 : SUCCESS in 6m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48980c0de9e94d8aadfb534ae0f9a427 : SUCCESS in 5m 41s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f00b4c3f344a465da22d1a70bf3c1e01 : SUCCESS in 9m 12s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/58ee7b345d7e44efa2909c0cc7108936 : SUCCESS in 9m 56s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/04d50ec3b25d442f89075aaf44d07942 : SUCCESS in 8m 49s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/5871b5613c3f47c39d1a679c474290da : SUCCESS in 5m 20s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/5b62d75155864c88b4ff45af22f44ad5 : SUCCESS in 1h 04m 18s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/9d7c32cbe959498e9b150c1f699db724 : SUCCESS in 1h 08m 39s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/64346af247ab45f6b2dee40416b5125c : SUCCESS in 1h 02m 33s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/bec04e42aeeb40c59f36cb0f42fe888f : SUCCESS in 1h 13m 28s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f8c3c295705041ce888f09559b2404c9 : SUCCESS in 44m 51s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/17c14c66846c48f3add55f16d0e06d19 : SUCCESS in 1h 10m 38s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/bc7f88bf7090435082681036d081bc70 : SUCCESS in 31m 45s","accounts_in_message":[],"_revision_number":1},{"id":"208f2a761b92a54503455fa1be4c4eaf3e130428","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-06-02 14:51:10.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9cd857a11079d0f7aaf63218349fa7a243ce46f8","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-06-02 15:12:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"62896e25aeae49f0a7522313b4ea3ac7ac169085","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 15:13:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d3411dcb1b9cffe84b82a19f5a592764cfb4b56c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 18:43:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/83b1b38861d04e2a8b0e652a1dfd1ec6 : SUCCESS in 1h 09m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3f5c1d8514dd48218f3b873d46454d81 : SUCCESS in 1h 17m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8f91d31f544749eb97a6d40f385c172a : SUCCESS in 46m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8022079b4dd4ef1b4e11ef35ae918a9 : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/459f3cc208994b16a64fe25f419a6c3c : SUCCESS in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/be087282d3144b57a4e21efa9c4e1064 : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2458b8540904649bb11635bdfb268a6 : SUCCESS in 6m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4191293d49bc416c96b2ad7fbcff53d3 : SUCCESS in 10m 30s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d0962101ae254884900600c2d4063fc4 : SUCCESS in 9m 32s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c1f102576d0c46109fe3aaa26a5c86fb : SUCCESS in 1h 08m 08s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/840496e1426c4f7bbe9a10cd0a9a61fb : SUCCESS in 1h 09m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ed53e5d1644c4153a139f72aab95fb64 : SUCCESS in 47m 24s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9135371c7ecb40ba83ff1f9d1dcfe9b3 : SUCCESS in 1h 16m 29s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e54d470fa5a74fa49358bfe2f318ec99 : SUCCESS in 28m 23s","accounts_in_message":[],"_revision_number":1},{"id":"6e08ea7c5b60deef27080a9fcf35d2e097e3106f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 18:43:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3b1db3900bab165f3cdd840891a79812d8f435d1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 18:44:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8cfb2bf1580742b6a8510ec8e28d1b69 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"316ce9659acdec0bc45f8a33825ae3b076199b24","revisions":{"316ce9659acdec0bc45f8a33825ae3b076199b24":{"kind":"REWORK","_number":1,"created":"2022-05-17 18:04:10.000000000","uploader":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"ref":"refs/changes/62/842162/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/62/842162/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/842162/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/842162/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/842162/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/62/842162/1"}}},"commit":{"parents":[{"commit":"881b24a2ad882ba06e16acd8e5a54eeb9d30e1a5","subject":"Update qcow test to use qcow2 instead","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/881b24a2ad882ba06e16acd8e5a54eeb9d30e1a5"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-12-02 20:01:17.000000000","tz":-480},"committer":{"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","date":"2022-05-17 18:03:41.000000000","tz":120},"subject":"Fix set_property_atomic() boolean type casting","message":"Fix set_property_atomic() boolean type casting\n\nIn set_property_atomic() we pass an integer query filter for the\ndeleted field, since most projects use an integer for this column.\nHowever, in glance the column is a boolean, which trips up postgres\nsince the types are different (mysql and sqlite work fine). This\nminor change to use False instead of 0 should fix that for postgres\nusers.\n\nChange-Id: I5149df76943c1c19f3204b904c0e2d3ef846bdf7\nCloses-Bug: #1953063\n(cherry picked from commit 753c74c343ab4ef32879a3dd95c3a956af77f869)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/316ce9659acdec0bc45f8a33825ae3b076199b24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/316ce9659acdec0bc45f8a33825ae3b076199b24"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
