)]}'
{"id":"openstack%2Fglance~842164","triplet_id":"openstack%2Fglance~stable%2Fvictoria~I5149df76943c1c19f3204b904c0e2d3ef846bdf7","project":"openstack/glance","branch":"stable/victoria","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2022-06-02 14:56:15.000000000","reason":"Change was abandoned"},"32755":{"account":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"last_update":"2022-06-02 14:56:15.000000000","reason":"Change was abandoned"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2022-06-02 14:37:10.000000000","reason":"Abhishek Kekane replied on the change"}},"hashtags":[],"change_id":"I5149df76943c1c19f3204b904c0e2d3ef846bdf7","subject":"Fix set_property_atomic() boolean type casting","status":"ABANDONED","created":"2022-05-17 18:05:52.000000000","updated":"2022-06-02 14:56:15.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7f98f73433da0e50a156bb60e6dcf6a0a511504f","_number":842164,"virtual_id_number":842164,"owner":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-05-17 19:41:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-17 18:05:52.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"CC"},{"updated":"2022-05-17 18:06:51.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2022-05-17 18:06:51.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2022-05-17 19:41:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5e0d56e7ae1cc7402c4715af157aa9d383219617","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2022-05-17 18:05:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8d70646456ab1d5785a4e7cc948b525826d214fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-17 19:41:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/6c4b67f126ac47978fda09af5f504020 : SUCCESS in 1h 18m 21s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/2ed1de52e8b842359120a3162b40574b : FAILURE in 56m 23s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7ada42d4f03e4231a5b4a0bdfd21382b : SUCCESS in 1h 27m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0f3c8a27b336457d82888c37c48eca59 : SUCCESS in 49m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7ec43a557e54b809280c2212286a984 : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dcb8503335974b3ba7345bc839948fc4 : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6bd8f51468ce41b5973d160e87029aa8 : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6bfe5d229504c69a01d2af2ba2a0f08 : SUCCESS in 5m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/45980a7537e243d5afc2e5449fa5705d : TIMED_OUT in 31m 12s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9239914d02344f35abcf9d14dae4616e : SUCCESS in 8m 16s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/307f1f69b20549f593493f3dcdc1b1c1 : SUCCESS in 4m 36s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/fea60f8a5d084f73bf9a6edefeea8b39 : SUCCESS in 1h 16m 26s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/25ec795031f04f979cdcb006a80019a2 : SUCCESS in 1h 17m 42s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/88e68fad6eee4fd59154de28b0089188 : SUCCESS in 1h 12m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e4cef3efc900400e80662739642449e2 : SUCCESS in 56m 20s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ed7644b6fdc0461ab8ac9919483bdfa8 : SUCCESS in 1h 19m 24s","accounts_in_message":[],"_revision_number":1},{"id":"05af0093bafe39a641002e59ed9b027599d63121","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2022-05-23 12:17:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0aa3525498c0d683ca97545310fd09ae07d1ca0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 13:40:04.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/cbb229518723497aa01d62cd56fe5312 : FAILURE in 24m 04s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/7ac10fedf3ab46459fad9d4c5d14325d : FAILURE in 18m 10s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b12d0a935b4b4c72a645621830d8bacb : SUCCESS in 1h 10m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/298a6ae61fbd47f4a9b9dca13fe977e6 : SUCCESS in 48m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03083ef0d3c44a99814f98dcc9cbf761 : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f2b0d757378a4ec289545c94e31b9888 : SUCCESS in 7m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d7342e63382541fe800d485e75056690 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72d943a1b5604e979e8654911f96ccbb : SUCCESS in 8m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/492c9f525a754c7088ffab034a7aa991 : SUCCESS in 7m 12s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/597682ccdce54b679e2ecaf6b1c9d5dd : SUCCESS in 8m 34s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/9ff5460ad7c1495ea09811da5ef28ba4 : SUCCESS in 5m 10s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/0418e0620d6946a88eb28af869a6454e : SUCCESS in 1h 12m 32s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/e86991f99f8e472dae8af1fdbfd1254f : SUCCESS in 1h 09m 42s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/b3d7752d6c5f4cb9b9f1838a00e6bd50 : SUCCESS in 1h 12m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6a548def9f48434781d811a835abd068 : SUCCESS in 51m 41s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/beb64772e5d14ae5bbc2db01883bd12a : SUCCESS in 1h 11m 45s","accounts_in_message":[],"_revision_number":1},{"id":"98be96de250b80c4beb35f6ea92b56a92e33d18c","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-06-02 14:37:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7f98f73433da0e50a156bb60e6dcf6a0a511504f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-06-02 14:56:15.000000000","message":"Abandoned\n\nVictoria is too old, we won\u0027t be merging this","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a698cc322d6d02daeefe451bec315dc9fccda1fb","revisions":{"a698cc322d6d02daeefe451bec315dc9fccda1fb":{"kind":"REWORK","_number":1,"created":"2022-05-17 18:05:52.000000000","uploader":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"ref":"refs/changes/64/842164/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/64/842164/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/64/842164/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/64/842164/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/64/842164/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/64/842164/1"}}},"commit":{"parents":[{"commit":"78d63ec150b764f15a48ae9a32890e1c06abc163","subject":"[stable-only] Specify openstacksdk stable branch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/78d63ec150b764f15a48ae9a32890e1c06abc163"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-12-02 20:01:17.000000000","tz":-480},"committer":{"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","date":"2022-05-17 18:05:38.000000000","tz":120},"subject":"Fix set_property_atomic() boolean type casting","message":"Fix set_property_atomic() boolean type casting\n\nIn set_property_atomic() we pass an integer query filter for the\ndeleted field, since most projects use an integer for this column.\nHowever, in glance the column is a boolean, which trips up postgres\nsince the types are different (mysql and sqlite work fine). This\nminor change to use False instead of 0 should fix that for postgres\nusers.\n\nChange-Id: I5149df76943c1c19f3204b904c0e2d3ef846bdf7\nCloses-Bug: #1953063\n(cherry picked from commit 753c74c343ab4ef32879a3dd95c3a956af77f869)\n(cherry picked from commit 316ce9659acdec0bc45f8a33825ae3b076199b24)\n(cherry picked from commit 677c89c23631e9083261a1a18ed438d8966e0de2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a698cc322d6d02daeefe451bec315dc9fccda1fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a698cc322d6d02daeefe451bec315dc9fccda1fb"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
