)]}'
{"id":"openstack%2Fglance~842336","triplet_id":"openstack%2Fglance~master~If174762b20ae0ddb468757a81dbb38e01f4061e4","project":"openstack/glance","branch":"master","topic":"zed-m1-config-refresh","attention_set":{},"removed_from_attention_set":{"19138":{"account":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"last_update":"2022-05-18 16:33:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If174762b20ae0ddb468757a81dbb38e01f4061e4","subject":"Refresh Glance example configs for Zed milestone 1","status":"MERGED","created":"2022-05-18 11:41:29.000000000","updated":"2022-05-18 16:36:24.000000000","submitted":"2022-05-18 16:33:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"842336-zed-m1-config-refresh","meta_rev_id":"04cb4832d75129c62c30fd3169c4034a67aa7a8e","_number":842336,"virtual_id_number":842336,"owner":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-18 16:33:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":2,"date":"2022-05-18 12:10:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":2,"date":"2022-05-18 14:16:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":1,"date":"2022-05-18 14:16:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-18 12:10:26.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2022-05-18 13:22:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-18 14:16:34.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"a1adf3254b6d9125f3aae27000b2f99387f65dce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2022-05-18 11:41:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5ff0ab40abba4eb6d8e437d83762ea514ea64d85","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2022-05-18 12:10:26.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bb589265d09200a607eefe468164797d751c23ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-18 13:22:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/639636310a3f4200a4178f054fae4b8e : SUCCESS in 56m 43s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/86f42cccaaba4b5c9424ada61d60ef71 : SUCCESS in 1h 17m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/83ef9987b8564bcf81e405d5ae87ac96 : SUCCESS in 1h 03m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb1f1d52e2c64e5caf9b9457a68f19b1 : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/72275a2532cf4516bbbb913bf3161e08 : SUCCESS in 5m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aafd8d5ab9c647d58b571f0c2a780322 : SUCCESS in 6m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4901d669bd64f27ae9be3692cf8b8ae : SUCCESS in 7m 19s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/ae75ce7c9c724c9ca10b210fb463b78c : SUCCESS in 10m 36s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/937d2317949f4030931bcc9ee320fe22 : SUCCESS in 9m 19s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/e6388ec8e84c46d09f7ebb5e9ca3a7cc : SUCCESS in 9m 19s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/df8fa23ad4c14e34ae54d8ed7de795cc : SUCCESS in 1h 27m 14s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/948788260e294609ba363c47341e2340 : SUCCESS in 1h 32m 38s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/3c2e29f9c7ba442883b6a5ed5da5f615 : SUCCESS in 1h 30m 43s\n- glance-multistore-cinder https://zuul.opendev.org/t/openstack/build/68bf7c20904d4f268ad06f229a08adfc : SUCCESS in 1h 32m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/278cbd6af8634f008f1ed580c26cdce8 : SUCCESS in 59m 46s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/88d782215e6a4b1295cac478789a929c : SUCCESS in 1h 12m 47s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/4a61446cc8cd45928cd5b2d30e14c615 : SUCCESS in 38m 58s","accounts_in_message":[],"_revision_number":1},{"id":"fb255a1dae4eed8acdb2ec0cac6473d1449c0987","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-05-18 14:16:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fefb8489065bbfcde52e68b2bf2d91fdef4113d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-18 14:17:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6962def9d47c9aa5672165fa056cf90b1ba5687b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-18 16:33:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/2f5f279beb544a98819ccccabf180f07 : SUCCESS in 48m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/647f536b41df4fafb5c2956b36d55050 : SUCCESS in 1h 15m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e090eb5d97ad427b828d8b3aad071002 : SUCCESS in 44m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7792fe08c1844d1a220d821baa8a001 : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d566b0f2169441ef9fd1c4091120ba7a : SUCCESS in 5m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bc35db2ea11541e9bbc631042b9c006a : SUCCESS in 5m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f076c74a80254a1e8c0aadcf86154e9c : SUCCESS in 6m 24s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c68c0ec913b74bf4bebaa782f01981f0 : SUCCESS in 8m 17s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f4b29e5b7197410f8fab538fb9eedfa5 : SUCCESS in 9m 16s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7a272c70f9df4cea97bb22ade9cec574 : SUCCESS in 1h 04m 19s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/bc53d85b0a3648538f70a07e9980a450 : SUCCESS in 1h 09m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fb6fd4f28537494bb30316fe15f2e767 : SUCCESS in 43m 37s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/42afc0330d8d422e86e1b59b38634cce : SUCCESS in 1h 10m 23s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/467df7c0e2ce4a47bf9fc685f07b8e55 : SUCCESS in 29m 53s","accounts_in_message":[],"_revision_number":1},{"id":"a2a3b5d63cd86c146293d4c0b8bb0a20dfd5f4f8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-18 16:33:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"04cb4832d75129c62c30fd3169c4034a67aa7a8e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-18 16:36:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76c33b3437cf43f0af71fb94888f61d1 : SUCCESS in 2m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"416a56c492b73f1ce2f7952995741347fb707ef4","revisions":{"416a56c492b73f1ce2f7952995741347fb707ef4":{"kind":"REWORK","_number":1,"created":"2022-05-18 11:41:29.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/36/842336/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/36/842336/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/842336/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/842336/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/842336/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/36/842336/1"}}},"commit":{"parents":[{"commit":"f01f31a792c7d0befde9a8ab486ffdc57f8ca483","subject":"Merge \"doc: add aarch64 to list of architecture image properties\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f01f31a792c7d0befde9a8ab486ffdc57f8ca483"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2022-05-18 11:21:51.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2022-05-18 11:21:51.000000000","tz":0},"subject":"Refresh Glance example configs for Zed milestone 1","message":"Refresh Glance example configs for Zed milestone 1\n\nChange-Id: If174762b20ae0ddb468757a81dbb38e01f4061e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/416a56c492b73f1ce2f7952995741347fb707ef4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/416a56c492b73f1ce2f7952995741347fb707ef4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
