)]}'
{"id":"openstack%2Fglance~842400","triplet_id":"openstack%2Fglance~master~Ie72a2531172208e32626ad2060737257663eba46","project":"openstack/glance","branch":"master","topic":"bug/1973136","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2022-05-20 02:57:00.000000000","reason":"Brian Rosmaita replied on the change"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2022-05-23 06:00:53.000000000","reason":"Abhishek Kekane replied on the change"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2022-05-23 07:15:32.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie72a2531172208e32626ad2060737257663eba46","subject":"Revert \"Override GLANCE_USE_IMPORT_WORKFLOW in cinder jobs\"","status":"MERGED","created":"2022-05-19 08:26:47.000000000","updated":"2022-05-23 07:18:06.000000000","submitted":"2022-05-23 07:15:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"revert_of":841837,"submission_id":"842400-bug/1973136","meta_rev_id":"b38af9c49117309f6d165111088de40711528859","_number":842400,"virtual_id_number":842400,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-23 07:15:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2022-05-23 06:00:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2022-05-23 06:00:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-19 08:26:47.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-19 08:26:47.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2022-05-19 08:26:47.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2022-05-19 08:26:47.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9e04f2e6efb5b3c369c0905ba71896b9c15159f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-05-19 08:26:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f7bfa810801eef02eb58a899827bead6d6ca71fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-19 10:03:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d1ed42b7d83040d29f179bdfadea9636 : SUCCESS in 47m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2943392bab684df9b11cffbff433ada2 : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/430d720ed17b4e7293b5b22effd1ac99 : FAILURE in 3m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b2751b1f5f3e401783ae573270c0e55a : SUCCESS in 6m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34f8f6bac8074252bebe02594a03cc06 : SUCCESS in 7m 02s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/cc757f8defbd4973b3e0e78afaf93daa : FAILURE in 6m 17s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c9c538ad05bb49278153a36f33727d15 : SUCCESS in 9m 31s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/0f4d6ad395b34f5e93985895fe7a564e : SUCCESS in 9m 52s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/80d8185d744e453598d626c15789796e : SUCCESS in 1h 29m 46s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8a5a522e8efe47f5805a2112b22c3caf : SUCCESS in 35m 32s","accounts_in_message":[],"_revision_number":1},{"id":"baaaeae73c42a3dd18d607a4e6121033f54e0759","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-05-19 10:45:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9d5c5a0928bb9598cdaa84f35eb0463aabc24acd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-19 12:57:45.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e36c68b9d9d84f5680e360fca24a3046 : SUCCESS in 39m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c6dcb202f1b409390be49792a7cb47e : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f933d5a01d394dc292fea838cf0bc99c : FAILURE in 3m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/01f0178446bc48f9a3f0d153e0a5c136 : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/242fd601558444ee84ac10fc61e69bc5 : SUCCESS in 6m 18s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/3936c7242258425fae5c6b66100d0ceb : FAILURE in 6m 22s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c0a058c2526c41929f3c9d8809e687dc : SUCCESS in 9m 20s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/ca3ec7eb9a9148b6896877348b57c38e : SUCCESS in 9m 06s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/26cd76c3cd6541dab8824cdc53778b6d : TIMED_OUT in 2h 05m 41s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f932e700937149319bed2d942cd9b8f0 : SUCCESS in 40m 46s","accounts_in_message":[],"_revision_number":1},{"id":"3ea4400c8258d16ae944e1f5890c3d786005c447","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-05-20 02:57:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"777130846e7826f121a2e5f96a9e6906ffd0d573","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-20 04:25:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ccf08ab6ebd84db1885c693041637705 : SUCCESS in 48m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50b497c31bf54b818ab93b8973934de3 : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/41c62db96a7b4a2abc05b925518c81fd : SUCCESS in 5m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9f99c8df3bf64ae3b374f2a47a512cae : SUCCESS in 6m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6e4fbec686e1403285eb9fce8c53ed9e : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0014312cbfbc4bb4bc6af65c451a1321 : SUCCESS in 6m 53s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7942a85ebed2432093d434d7980232df : SUCCESS in 10m 43s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/fb1e3bef17ff4befbdd6a9b57d05cdee : SUCCESS in 10m 26s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/5f10c2286f1c4bcfb09254f04b9912ab : SUCCESS in 1h 24m 51s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2209efbcac064a9e9d044fe672e088f7 : SUCCESS in 43m 42s","accounts_in_message":[],"_revision_number":1},{"id":"c941681d7755d72c6c7014e6ca5aa1976c55ac5a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-05-23 06:00:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ba0270cde2f7d351bf5b714c9c8f3599ecf280fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 06:01:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"234e3d6a511ac9e83fe81a2ddefcd63b9a2ee6f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 07:15:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/16e0670abf1c48e2a529036475c0b421 : SUCCESS in 40m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56acc5d501544f93bcc739bf8f27a54f : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6916b21d38384b55bc8c656465415d45 : SUCCESS in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f5a09f6e65144cc8ac76af81eddb40c0 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc46f27999574740b8423d224b6f84b4 : SUCCESS in 5m 26s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/bafda9b7dd794f9a9f73455e4515c1d1 : SUCCESS in 7m 24s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f8e864479a294905a8ff15fa406f69f1 : SUCCESS in 9m 33s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/419f3bc65a76438ea136fba2eb3a7feb : SUCCESS in 26m 42s","accounts_in_message":[],"_revision_number":1},{"id":"4af9ed27257a38f9edec229c5a432de4ec5e3dc3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 07:15:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b38af9c49117309f6d165111088de40711528859","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 07:18:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b54981b6d1284e9eba7dfde7162bb96c : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f9fad19249cef90f53545068cb842e299ca0ad61","revisions":{"f9fad19249cef90f53545068cb842e299ca0ad61":{"kind":"REWORK","_number":1,"created":"2022-05-19 08:26:47.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/00/842400/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/00/842400/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/00/842400/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/00/842400/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/00/842400/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/00/842400/1"}}},"commit":{"parents":[{"commit":"025ba83c470fd58425273ad10975a1eb3d095452","subject":"Override GLANCE_USE_IMPORT_WORKFLOW in cinder jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/025ba83c470fd58425273ad10975a1eb3d095452"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2022-05-19 08:26:47.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2022-05-19 08:26:47.000000000","tz":0},"subject":"Revert \"Override GLANCE_USE_IMPORT_WORKFLOW in cinder jobs\"","message":"Revert \"Override GLANCE_USE_IMPORT_WORKFLOW in cinder jobs\"\n\nThis reverts commit 025ba83c470fd58425273ad10975a1eb3d095452.\n\nReason for revert: This is no longer needed as the devstack patch is merged to fix this issue https://review.opendev.org/c/openstack/devstack/+/841804\n\nChange-Id: Ie72a2531172208e32626ad2060737257663eba46\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f9fad19249cef90f53545068cb842e299ca0ad61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f9fad19249cef90f53545068cb842e299ca0ad61"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
